)]}'
{"id":"openstack%2Frequirements~692156","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I0bc671ba2644fc3f74e1f95c53be55e6bb8f1a88","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I0bc671ba2644fc3f74e1f95c53be55e6bb8f1a88","subject":"update constraint for keystoneauth1 to new release 3.4.1","status":"MERGED","created":"2019-10-30 14:37:32.000000000","updated":"2019-10-31 07:11:58.000000000","submitted":"2019-10-31 07:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692156-1572505918333-cd6aae0a","meta_rev_id":"855d043d953aa73be702455b2d633a14335c3863","_number":692156,"virtual_id_number":692156,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 07:11:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-30 14:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-31 05:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-31 05:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 14:39:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-31 05:23:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-31 07:11:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eaa5bed4faa152a8200317511d655013be4ea3d7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-30 14:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fc2534a0e49fd760117a4b306c137a08ecbd872","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 14:39:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5c98f1e995c3065b8f678988e4bfe197ced2d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 22:31:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/91ad04f878954ba2b234b8d0b1ca75ab : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d1888bdeba4cd399733bbef5c9c986 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/858c2534b71d4c90947130a2a2db8ace : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0de1de415d304fc9902b13d072f0d8bf : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0227bd2c7d5b4df89326f72bbab51152 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72736523028049928c52f675211c6b12 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6b9daedd544a4385b6e9ee2274a53903 : SUCCESS in 6m 01s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/2596a9021dde4f51986ff0a1ae9d347f : SUCCESS in 13m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e580403610b34b5da17f8fea6399bb06 : SUCCESS in 10m 38s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c268dfb6a59c40baad5655cf5feb5eb8 : SUCCESS in 9m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4fd693e90e914944a6ceb087940c6b79 : SUCCESS in 5m 31s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/822dd41f4d084d2d8bc051e846fdacee : SUCCESS in 7m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0a69f19577fe4bafaeef74d7bf78afc3 : SUCCESS in 9m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/3935ee4b6b524242b941fd924a869848 : SUCCESS in 8m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fd351e7945ea4cae949725d3efe1f799 : SUCCESS in 34m 28s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5eb1b50a79174017a98c6d4e28557dbc : SUCCESS in 18m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/575885f7a1ae4ac990cadd7246dc09ed : SUCCESS in 22m 51s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/52503e1c6d524a7da361a58ef7a9b9bc : SUCCESS in 20m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a56258508f8c420e9f24157014d149bb : SUCCESS in 16m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/78bece339ace4f99b7b23c34b847a712 : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/328c23c1f2e84a9fbefa464c1b8307b2 : SUCCESS in 18m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f310c38830f4832977f32e939dcd845 : SUCCESS in 6m 52s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4bd475eba8ff4f14abe157241b449617 : SUCCESS in 4m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5260aa659c584388b18f516f42ab3432 : SUCCESS in 50m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/858ec726c786468699bdeee5d25ebaab : FAILURE in 22m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/188e768eb7544d0b9c4beab04443db01 : SUCCESS in 1h 35m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91229e540f3d43d4a2a3ff2159dbc331 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c590072275ecb576472c8b4cc57e3346aa36b8ca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-30 22:33:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"440075a5df27e192436353f076f1548db7a03208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:22:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4f4a6f761b704730a8b46a3d5ad67536 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6df461825f0426abbc10371636a42db : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ec5a03847804c38bb93538a5b02c521 : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2a173b16f3d34df0b74ca1d89e3b2bfc : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d087253792f94866b9a4901e2eee4fed : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d350fdbb575418e8829996503add9aa : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1f028e718a194893b404a3f3e027f381 : SUCCESS in 8m 37s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/18521eee2df94b6db5250fb2be6a5159 : SUCCESS in 8m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c35ada1aa6584567b28b74ccfa3324ab : SUCCESS in 9m 55s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9740b20700e8420584fe994f45266fda : SUCCESS in 8m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6b52fce5194c4440ad62d05482be34c2 : SUCCESS in 6m 09s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/268b4b4e1ba545c184143f240f8452c4 : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3446a5969fef43dfb9805c9d0e05c99e : SUCCESS in 10m 25s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8ac3b5b1287b4735861654e25be5824c : SUCCESS in 8m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f960358abd7b4820858bde7f5e575f98 : SUCCESS in 17m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/eafe269b443346219d135c61620a7fc8 : SUCCESS in 15m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6d8438ce47da4bb89da58a92954c47e1 : SUCCESS in 24m 37s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cfcdc4c00c724a0597d437cc43f7beaf : SUCCESS in 22m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7899d3e5a62746878abdb26e4a7b0078 : SUCCESS in 14m 00s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e82eda059ad943e2901a9f83ebcbaafc : SUCCESS in 15m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0f39ad4ae454d95b9ba34c1dcb826ad : SUCCESS in 21m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d5a38eb146bf45f1a6973503a80cea65 : SUCCESS in 6m 55s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/37bf3982c0c7482a987a254921ca2081 : SUCCESS in 3m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1697a375f08a4cbfa2894507cef0cf37 : SUCCESS in 48m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16aca423e4134fac9613145b55631733 : SUCCESS in 54m 40s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e4ccd63c236341cf8db933a158859f23 : SUCCESS in 1h 18m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d872bebf4dd4965b74fa3d91dcc0596 : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":1},{"id":"83d04ba8143e0730d4699d6b5cf0c01acfb3de5d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 05:23:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d77329b7ef77ec0600fe353602fb6ee4e38a14f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:23:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba7e35eced04e2a43ab88ee7a0a3f9891aa3dc7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 07:11:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44433bd732f4ad0b321f4888532403f : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/393a48f8630f437d90b0eefbc92b138a : SUCCESS in 3m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1ac3242741584bac83ab7977de5f2ea5 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84126abb66cc4cc690044ef7a47739d1 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56aedd0eb7e448efb400228b0df95f35 : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e239eae9843a451784952df43d59cdff : SUCCESS in 6m 40s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/ad26d2b7bba545de9b42e92785f87864 : SUCCESS in 7m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3f25c18a765242cca102ae1471df743c : SUCCESS in 10m 36s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/efff1cf13945430c812d90e5e851f20e : SUCCESS in 8m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/241486958c7842e3b62e7ecd55e57df2 : SUCCESS in 5m 04s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/90b7946d34e448e295d5a3b64809747f : SUCCESS in 5m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a2668b56194f46e0a98956e64d0cf9d4 : SUCCESS in 9m 19s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1d2778181fd745f9a9ee64a2df427e84 : SUCCESS in 10m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f82e3a2d4b8b4d238e8c83767008e7be : SUCCESS in 16m 21s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/516fd039680149d7b53f8cbb23e856ce : SUCCESS in 22m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/db6e6b1998bf4072bfe2c2b4d13fe4fb : SUCCESS in 28m 34s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/91e086a7e7a049889cfb8f9fb1c03859 : SUCCESS in 20m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/45fe3b86d1ee4c1d937fbaf0a6491901 : SUCCESS in 15m 55s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0945b00ad7924325918d6a65cf61b1bf : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6ceeb9e454943c99f594b271933730d : SUCCESS in 23m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32892ed41c974bd497cecbd59b7efc13 : SUCCESS in 6m 33s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e48a49bec87042ebb2cd9a4c7a80efd6 : SUCCESS in 3m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/44685d8bd7654d56887850c0dc0ed300 : SUCCESS in 1h 01m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2996330d0a7c494490a923d0cfb398e1 : SUCCESS in 59m 21s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/744c3a8007b248ba8e1b3aa0d778ef35 : SUCCESS in 1h 35m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e537d57a610b43c7b5a481f5546e1ac8 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"26e49aab1f293d5ec8ec72322a998479ed04d0d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 07:11:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b22b150739ac79f701c68ab08f0d5f303f0a501","revisions":{"1b22b150739ac79f701c68ab08f0d5f303f0a501":{"kind":"REWORK","_number":1,"created":"2019-10-30 14:37:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/692156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/692156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/692156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/692156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/692156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/692156/1"}}},"commit":{"parents":[{"commit":"9344102614a3aded9a44ba6de06e2e501ab95059","subject":"Merge \"update constraint for python-swiftclient to new release 3.5.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9344102614a3aded9a44ba6de06e2e501ab95059"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:37:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:37:29.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 3.4.1","message":"update constraint for keystoneauth1 to new release 3.4.1\n\nChange-Id: I0bc671ba2644fc3f74e1f95c53be55e6bb8f1a88\nmeta:version: 3.4.1\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b22b150739ac79f701c68ab08f0d5f303f0a501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b22b150739ac79f701c68ab08f0d5f303f0a501"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
