)]}'
{"id":"openstack%2Frequirements~692157","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I402c72866d0f2336224221b022aeda619467e8ed","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I402c72866d0f2336224221b022aeda619467e8ed","subject":"update constraint for ldappool to new release 2.2.1","status":"MERGED","created":"2019-10-30 14:38:31.000000000","updated":"2019-11-01 11:47:04.000000000","submitted":"2019-11-01 11:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692157-1572608824524-e7a1f455","meta_rev_id":"1e6e39d29821a13489061cf9170542cbdb31e95a","_number":692157,"virtual_id_number":692157,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 11:47:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-30 14:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-31 15:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-31 15:00:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 14:39:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-31 15:00:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-01 11:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7d569bbd5254a02bd98a5ed5380020ab4c43918","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-30 14:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffec07a1aece00edb10eaaa9e5f8cbb22cc3bff2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 14:39:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1cc17f570dc0a9de4a20d177b4bc3dd675772aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 22:38:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3079871810d6450aad12a36aeed4a361 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b99ecdfb8e541bcae95d8a1968f9814 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e01dfb7e8283434d8d195c0c7912e435 : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4952f1e6bef4caa85a50a43a0eb0ce7 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13410ba0fe12457d8050e720606387a1 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e1faba994fc422ab666d618c6185a78 : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2216f497573d4cde977c10b5501c9b47 : SUCCESS in 9m 53s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/bcc98ad7d8ef480d8035d42a65d89810 : SUCCESS in 10m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f02d5dda35e0426b9ea8fca421c69e7f : SUCCESS in 12m 58s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9f2c797674054f868e260a9661a565cb : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c2d7f56bfba641ad89c55eb3b9176676 : SUCCESS in 5m 52s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c11d4217675241778772928e3eaad693 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a9246b5779c84843912603d567678246 : SUCCESS in 8m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/519f89ce246d4435a089d37aec1eaa28 : SUCCESS in 10m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f47832b995b04b4eb78a3d4247909fb2 : SUCCESS in 14m 50s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4afe79e3ef0b4af6b3275a64242c7fcc : TIMED_OUT in 40m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f94771937a6a403f9d17c94f4fc7ec40 : SUCCESS in 21m 21s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/76c21d7b68264b13ab48b56c5c84ad1b : SUCCESS in 24m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/57c62d01fcd848ed839f06cd3dbc6ab2 : SUCCESS in 16m 25s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/eec8c4d180a14439af1767dcbdafc48a : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0489e7a253f4148b0490a80108a1c76 : SUCCESS in 21m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/050af41e5a93420799ed47d4f0d58a6b : SUCCESS in 6m 55s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/7fe7eb77d71e448fa1e272614f8725af : SUCCESS in 3m 46s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6ba51242a64144e88d88f792df8b7119 : SUCCESS in 1h 08m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a16acfce14904ef8a512ea2b0e6bcdb1 : SUCCESS in 59m 18s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f09d46050b444255a050b46edeb69caa : SUCCESS in 1h 38m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/10ff26c2b53f4eef9c5c46a0605a6724 : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8fe8f034de08eb4a03e34c21eb277e4c0f75d9af","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-30 22:45:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03168faa87ba1961593ae87a252eab87a4e77a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:32:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ef8bd26b7fe94e50919009e95c6a0fbd : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ac0abf19284419b3bfd11604424a32 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42ef66ed9bbc43afbca2879d05ca843a : SUCCESS in 4m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e317f9be425a4c6e8ddd292e8cec7823 : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1818a769eb954b4ca41f7f1b059bacf8 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc6fa8333aef4110869546db630de138 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4317ca431c0b4ac2a41efdc29cd02f4e : SUCCESS in 7m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5d1c4be2d75e4124b9d3e91ce7816227 : SUCCESS in 6m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/35fb1c54272d4f8b87a10b5661c418eb : SUCCESS in 9m 22s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9b5c3e1161ac4b9aae8e577c158f4968 : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/117b01e05c9d4fe9b29e03f3d9953da3 : SUCCESS in 5m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d0754c6d11164820b9de39284e2180d2 : SUCCESS in 5m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d1f21013dff6491ba5581bca7906a04e : SUCCESS in 9m 42s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a60cd26e06e646ca8e521f8ecdb46148 : SUCCESS in 9m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/354b81bcebe7412eb0e03063c4e5ccd2 : TIMED_OUT in 40m 50s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/295ab47ad1b04793827c60bfd07fa165 : SUCCESS in 16m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/be5e440b582445d78bfff1841292946a : SUCCESS in 29m 28s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/12929fee5ede44288247c353a39db692 : SUCCESS in 23m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f2c9c11449a14c0093e3763a8cf54526 : SUCCESS in 12m 12s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/beb33299bc514f00ade84479c50c5b34 : SUCCESS in 15m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d97475ee4e6b483ea876b1198bcdfa13 : SUCCESS in 17m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d266dbdf395e4d6da860cc5f93e7ad69 : SUCCESS in 7m 05s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/43b96a3a12bd49b39e6d5b3a055b1f04 : SUCCESS in 3m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7c459a58741f47219a701a4a16ac9ecf : SUCCESS in 51m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36bdc293953e4682856b777aa3a69122 : SUCCESS in 1h 03m 09s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/5ece247a38d54b789cb6588c3ec0c402 : SUCCESS in 1h 32m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b1cfca3105f9403b944109cadf5486de : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e843110edb172e887e1254efdb3128371a829f16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 15:00:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98c584fd6687b59038103bb6853fcf6bde34e01f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 02:37:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b3a39996192346d18407aaf46b97dc5c : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6c79d5ae3f484cb403db99532ed9f3 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/434307d9f49c48338731e3e5e9151621 : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de85bc4d030c4b6daf0479eb0d1f02b4 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8603350a37e43f8b18fbd0e10868776 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6c1d6165c144c37b1eec4d326b39434 : SUCCESS in 4m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c0afdad4081c4f2991bdce16324674d8 : SUCCESS in 8m 07s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/ae3a7c6003124b08b388d9f06bca44fc : SUCCESS in 10m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2dfd78c8a6e44609a953e6a22eab3bdd : SUCCESS in 8m 45s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1899964dcc59456396e7689254048549 : SUCCESS in 10m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/81eb50e969bd4e4e9ebf1ac35e6e650a : SUCCESS in 6m 54s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3646c7d27d954b7d9aed02e16be1b2fc : SUCCESS in 6m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/97d60df412b0428aa5dfb41dfaf73b19 : SUCCESS in 11m 03s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1bc0888a511c4c23a840640b92bb607a : SUCCESS in 8m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ef25c034938b4c9eb4653fb29e5e6f4c : SUCCESS in 16m 52s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a4875303674e4fea8b518018abdfc825 : SUCCESS in 17m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d094d8d5b48c4f87896aebd3e3e1d040 : SUCCESS in 24m 59s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5fdd16caa8ab443c8ee74acabffbb1fa : SUCCESS in 24m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b3a7639739174d43988cede4492488e3 : SUCCESS in 13m 37s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8c8de6e0947449d8b56e2f02326b7905 : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca7a16805fe548168005c439e3613158 : SUCCESS in 20m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90a41f8f365341a5a21c9836741c43d4 : SUCCESS in 6m 16s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/270c9e6ea53d4af781e1510fe2974612 : SUCCESS in 2m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e688e2f933c34ab48489008d89460831 : SUCCESS in 1h 05m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2835eb941604f4ea75c03ec102168e7 : SUCCESS in 59m 53s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0885ff4c16bd4acc9e29468d52ccf51c : FAILURE in 1h 06m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b7f7c939b7c4eccb8c1ee4310df9133 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d7ee5e59bc7b28bcc443c443c6a60cf3ec980787","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-01 03:00:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36cc38f554f0e4607e16caf032750e18b5a00fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:23:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9fc1cf2f45514999ab419d408e820750 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c733f8a614d411f88054f2ad83b9621 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3d4b0488f6d4f57951ab87387ae2deb : SUCCESS in 10m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d105938b06164dc1a7dfce645dd8a191 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c4e8042cabd4deba7e399e68bdf2cdc : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55e6fb7ae1e84f718d41af37877b8b0c : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a10599c8bd58467d898ef14c1717140c : SUCCESS in 6m 57s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/62a21e3aae5b41ba9dc6e24814810145 : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c0c0ee60928b456bb224269627b1bd95 : FAILURE in 15m 58s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bd06bc2b4ae045879bae73bad16f21ff : SUCCESS in 8m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1e2072f2cedc4e74bf55642382befdfd : SUCCESS in 4m 07s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e0c6feed2c0340039246c73a5a9fd68b : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3a7bd56bfb274aa081a3b7ec0bda87f9 : SUCCESS in 11m 12s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f7f60fbcc06c4f738348bbf8043e6bc9 : SUCCESS in 8m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ec757eed4dd944349304d56f1340dc29 : SUCCESS in 18m 11s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a53d03a92dbd455995bdf951933f5281 : SUCCESS in 16m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/972d674f6ca0424eb5ad236e42d161a5 : SUCCESS in 27m 45s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/6f478be5b9fd4d4685cb9e509a194e8b : SUCCESS in 22m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/56915d4a114d43f79619a9f7c9455466 : SUCCESS in 12m 19s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6ec89a032dfb4674a54ab72884da7a5b : SUCCESS in 15m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffb4eef9a22c42ab814c5b209187991b : SUCCESS in 18m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c3ba7d67395545da80a8424647cd64af : SUCCESS in 6m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2c40051af42145d5a1ab88c3e83b8346 : SUCCESS in 4m 00s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/44eda9cebd8340e095d2da192a07c23d : SUCCESS in 54m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9770b764b18429d8105068a998aee2d : SUCCESS in 1h 02m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/346dd756866046abaac9d39bd0b2283d : SUCCESS in 1h 21m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/33ffeec0686e47cd821bcb500d5331da : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6971e547b9df4454d71df4727f22401888ba3acf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-01 06:38:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51db1f1c7e964f4e812b7a203d693a7af5cfe3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 10:25:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0073d2025e8a4b0f87ab98a68c762eb9 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c870061b2047e3a48a7d79c92c8ff2 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74925c4913594e60878bb8d8ff712f0f : SUCCESS in 6m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c663154ef06f48dea7eb80b7e027a37d : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c51d2a814304e61992bb7c2568ffec1 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73b276897ca94a059407ed7fc5787955 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/63b9ff5b9ad24a6f934a976cebccfc05 : SUCCESS in 7m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/008e5d9969074d72a5496bbe3e13a9f0 : SUCCESS in 6m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/24ffa244184448dbb84438cf44f67c66 : SUCCESS in 10m 20s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/83e5ac1ab67b44828abb6fbca03348dc : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3b612b3184684214a28ad082f6aec684 : SUCCESS in 5m 27s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/017ff99b7904412bacc6e2959a4f2a97 : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d5acbfa560804f37bc0fece3cd5cdc3e : SUCCESS in 12m 46s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7bc495a0a7224644afbfde14d42b2b9e : SUCCESS in 8m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/771efaf8bbc74c5794975df9e8ec1968 : SUCCESS in 24m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d6b71a129b3d428886bb5211e354cfcb : SUCCESS in 23m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b4c595cfcbf44386a72c595e7eeb1204 : SUCCESS in 20m 41s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/dd9bcaee5d284d90afd94ae2cc7521ad : SUCCESS in 22m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dd8937cead244c9abf14e30ee99739b0 : SUCCESS in 13m 35s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9cd7353218b34b26ac4a975afeb1e44f : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3cb840d79e84478ebb2ed04287eb4e9d : SUCCESS in 19m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6fa64eab360742e98ceefd358111dfab : SUCCESS in 6m 04s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2e9fefe061504cf3897c89aace56bb61 : SUCCESS in 5m 24s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8504720b25674affaaca4ca645cb74ac : SUCCESS in 48m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a35d915949994a51928b85d6c1c1f303 : SUCCESS in 59m 34s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6c4b109a4bf1476db12706c3705a256a : SUCCESS in 1h 16m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/600dc8fcf1424b0481fc76609ee5b9cf : SUCCESS in 49m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b180988c3930f26c526c375505b2962d5436c9a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 10:25:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e783af4664e9e17ac7f02cbc4fb9178551469361","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 11:47:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79d884deb36881548598c8bec9eb632493ce6a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 11:47:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8396d54e903142fbae805c2a4b836180 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cbb09edbd3c40e99bc714782642edd7 : SUCCESS in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cb0b452a49c047cbaf526a60a765bf0c : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4320bacb5a854e50bec822907792b501 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8e1bae8d0b34b4c98eae905caa157fc : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f96d117ecde8452aa036e5b75cfe5ef5 : SUCCESS in 7m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/396e682aebee4ba9962d0415d3c8b6be : SUCCESS in 6m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/60651cc37ecb408a9c845dda1b0bc0dc : SUCCESS in 11m 13s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6c86c03c8308416caa7f6b99b6e0f3cb : SUCCESS in 8m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/616e3715d9bc4efd94891ee902bd7df6 : SUCCESS in 5m 39s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2ed2231e23084720809c81c6708e348c : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/73412746035747568b6ea7f2f3eea72c : SUCCESS in 10m 32s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/06771ab342624bad961b60b13789ebc5 : SUCCESS in 8m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/12e423a51f63430f95fc65e0a7fe269b : SUCCESS in 12m 36s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b1c44af9e4394aeb8342fdb4b05261e2 : SUCCESS in 20m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0068a805be154e76a88c021b5f058dc3 : SUCCESS in 21m 33s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/2bce55990eaf4e499eaaafda7c26b84a : SUCCESS in 26m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/83cf62574a07427eb07a6273e5a41825 : SUCCESS in 16m 47s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/113f3edcad264dd4869fb971d7472e75 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cabece1010e14f6a91e375040ec67129 : SUCCESS in 18m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4a38ce3aa0cb49c98b5552fab0637e58 : SUCCESS in 5m 34s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/07653e94ac9b45b78557f0d130e3d52e : SUCCESS in 3m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5786a6e7ae0845c5a99ffba1394d85e5 : SUCCESS in 52m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9e9eb312d5e4433b0f953f21d63f89c : SUCCESS in 57m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d3be8b6be1db4d83964961c97fdcc9c1 : SUCCESS in 1h 19m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e995ef0be84a4c9d94cb93015b92c7c5 : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c904427dcd45edd9b1b4e29177a78c2a071cf4b0","revisions":{"c904427dcd45edd9b1b4e29177a78c2a071cf4b0":{"kind":"REWORK","_number":1,"created":"2019-10-30 14:38:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/692157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/692157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/692157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/692157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/692157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/692157/1"}}},"commit":{"parents":[{"commit":"9344102614a3aded9a44ba6de06e2e501ab95059","subject":"Merge \"update constraint for python-swiftclient to new release 3.5.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9344102614a3aded9a44ba6de06e2e501ab95059"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:38:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:38:28.000000000","tz":0},"subject":"update constraint for ldappool to new release 2.2.1","message":"update constraint for ldappool to new release 2.2.1\n\nChange-Id: I402c72866d0f2336224221b022aeda619467e8ed\nmeta:version: 2.2.1\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c904427dcd45edd9b1b4e29177a78c2a071cf4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c904427dcd45edd9b1b4e29177a78c2a071cf4b0"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
