)]}'
{"id":"openstack%2Frequirements~692159","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I39eb7110808ece9defc225d465a3af57720c261f","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I39eb7110808ece9defc225d465a3af57720c261f","subject":"update constraint for python-keystoneclient to new release 3.15.1","status":"MERGED","created":"2019-10-30 14:41:28.000000000","updated":"2019-10-31 06:38:46.000000000","submitted":"2019-10-31 06:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692159-1572503926672-fa70c53b","meta_rev_id":"575a72896e4bd9f71d2f7d6d3dd16831b7f799c3","_number":692159,"virtual_id_number":692159,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 06:38:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-30 14:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-31 05:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-31 05:15:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 14:45:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-31 05:15:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-31 06:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b073dea033378f0fc63b89e15c8bb7fb7767e99","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-30 14:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd597e8be39f1413ad045dd4c92dd91d2314e769","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 14:45:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"399aa90ccae22674b5bfd4ce101b96154a60873f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 22:40:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cb6bc9fe26684f65887afafc87c19a3e : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d81d9da77f54ce18ee45b91b51d16f2 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/073cbb730a384a6683dfd47b5cf1571a : SUCCESS in 6m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/884760ef24e64d879fb069c0ac2401b7 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/799044aa8c8344d5939ef2bc3b1d6bb0 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d69c07a33304532be918935e514b44f : SUCCESS in 4m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cf438ad0f8374220ba8fc6d95f1f00b2 : SUCCESS in 5m 57s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d2bb66d5daba402b9536070c672b3cff : SUCCESS in 11m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/93eb5be44b0743fb8dfdcf02c27a4b5e : FAILURE in 12m 39s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/951745bb48e84926800b4c7a76da4656 : SUCCESS in 8m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ece7dea93eec46f6acd50229b2a9548e : SUCCESS in 5m 02s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/5ce0ff4d6ce74d9587a513d96d8e06fb : SUCCESS in 5m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4f20b21635724971aaae8c9daa81091f : SUCCESS in 11m 01s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1ce3c536a2214730ba4dab5c7f5d7ce9 : SUCCESS in 9m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0aa34ce273af4a57889798fc548e6fca : SUCCESS in 22m 04s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c043d4aa6b3d40feb7592f1b1744ed66 : SUCCESS in 24m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/542423fce9724e4aa9335a92a5176e9c : SUCCESS in 21m 11s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e28fa0b7ca094325824bd72f01cc2907 : SUCCESS in 24m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d26f73cc22cf4d6e8dd0305b70771713 : SUCCESS in 17m 01s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d24e14d15585410db1a2d68a0bc396b2 : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db1bd712f8e149f887cfe38c82d8555b : SUCCESS in 21m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f4a7282a470a4916973590029d81acf5 : SUCCESS in 5m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/60e451ccd8d44231b79e5312709026bc : SUCCESS in 4m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ef201c6f751b4a5e8f57beaf34ee0b06 : SUCCESS in 50m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed480a1adb6a44e3ab47961a66d1e9b9 : SUCCESS in 1h 10m 27s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8c79828a041941a1b2fea044a0358095 : SUCCESS in 1h 28m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/550d9b98cf7a4f06840f32a751314a2f : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"adc2ddc1809037ccaa1a82cc7bfb8cb1606854d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-30 22:45:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1eee4008d3f2c1a06512f9988af83b84a8afa701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:12:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/22f06918098a4f1e8839ba670ac602cc : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea0757a75894737b4289fd7928566be : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1214a4b8974429ca1a3a770f817b5ef : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/509ee89491234cfdbc39b15f0b767ff2 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0068dc864f7e4bd2b7abe8eeaa0f45fc : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/887e6f1b0e954ba4a12bc6f658e3d376 : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dcb1a3231d9c4e6cbe24a807c0cd44b3 : SUCCESS in 7m 14s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/17cab84e29b94cd595978b45630aa7e2 : SUCCESS in 9m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f47ab23256da42d2b77368ea42b5c77d : SUCCESS in 9m 37s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/4588e099f90f4133acbe1e52d1b951a7 : SUCCESS in 9m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/338daa0a190c4641aa9d7b11103ad132 : SUCCESS in 6m 08s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3cee1ef864414cf1a1ce5add00724522 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2d938a0282114639a0d8b491f8f7e1a7 : SUCCESS in 8m 58s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c66829c59fb24cab88a530929e201252 : SUCCESS in 9m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/daef3e4049b14efb9c9400975fe5b2de : SUCCESS in 22m 05s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/75fd372febe94bd2a3174520d07d3aa5 : SUCCESS in 18m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e3009f78652d40a68ba3b0146a3fab3b : SUCCESS in 20m 56s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4abda5a8c7a24cf0b83eb22eaee5cf34 : SUCCESS in 24m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e76f06393e1d4236b9d3bb145c342520 : SUCCESS in 11m 10s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4759027ccb85458cb66d11bacfede655 : SUCCESS in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/912c30f31332469c90c92d3bc543a4ef : SUCCESS in 20m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/80505433c0b94bd4b8d5ba20c298d5cd : SUCCESS in 6m 19s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4190b2f1e069405dbb81c1358ec4537c : SUCCESS in 3m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bdc31be936e94715a46b93a9bc31bb93 : SUCCESS in 55m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35fa42f840fa4820b481aa4ed9f278c7 : SUCCESS in 1h 07m 26s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b520e3ea5e66480f969b9df06b83a59a : SUCCESS in 1h 18m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a6f1110cd21440cdb84bc5e0132947ef : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":1},{"id":"97b6a611707e19c149309b4bb7558ff37990d870","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 05:15:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d8b87b96d712f203493c37fd34cad34672d0cf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:15:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c14c08b634172dfcdf8b4c6370243e19d48e25f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 06:38:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7511e6fbdea3432d88ab00fbbe29a55d : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2c4a485f419466b9dc8f688297ff228 : SUCCESS in 3m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/da12cee45dbc4ef882db5920f5cb52aa : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5a737c39dc9478aa9f39d474eba9d7a : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a85f12a4c5d34ec199b7ae757165e796 : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/300b2b17a70a4ee6add2c81fd4c53aac : SUCCESS in 6m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e95640ee9ffc47c59438636370182391 : SUCCESS in 6m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8085cd242c814c148c473814c4daa541 : SUCCESS in 8m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9d2a7766ee1f436e9fa3ecbedcd0f186 : SUCCESS in 9m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/998cc0119bf140dba82adfc5fa5fd85e : SUCCESS in 4m 51s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/784e35586b204d66824ba7012c0cbef0 : SUCCESS in 6m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ce3da06003af4c5c88fc6d5d0c76a742 : SUCCESS in 9m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ffb95474eba244ddb5a595df0b00cbd8 : SUCCESS in 8m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a94553f739504cb4b65254424af81690 : SUCCESS in 15m 14s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/96213c0c328347a9bfd574518f8a61cb : SUCCESS in 19m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9d7d411eb4664e2cb14d29a482e5c2aa : SUCCESS in 21m 40s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f814bb5db9bd446f9a2fed6b7b7661dc : SUCCESS in 20m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f060162266ba47d6be520c65e17c1e0e : SUCCESS in 15m 27s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/77c6f7fb46404c828931a0418c3b5fa4 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef236dd2bb5543518c9a0b1725daffdb : SUCCESS in 21m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bd99f5522f644619b740a1e64a7f9d29 : SUCCESS in 6m 15s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ebf10fb6a8a14bed91d5f344f60bd49c : SUCCESS in 3m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6ec9e3d1bfd04bedafd4e1d270dae8de : SUCCESS in 49m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/074a422a70534152999139fe031652e9 : SUCCESS in 1h 00m 27s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0db71fe789e64d52aed1472626546ebe : SUCCESS in 1h 19m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c2d230c2057e4e669b8c9ef46dc6d6db : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3af9d55e60565f17e43e2aa8985b621b36685cd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 06:38:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45353cdcb985613556cf0a7dcbdd09ed7fe7f8ae","revisions":{"45353cdcb985613556cf0a7dcbdd09ed7fe7f8ae":{"kind":"REWORK","_number":1,"created":"2019-10-30 14:41:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/692159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/692159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/692159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/692159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/692159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/692159/1"}}},"commit":{"parents":[{"commit":"9344102614a3aded9a44ba6de06e2e501ab95059","subject":"Merge \"update constraint for python-swiftclient to new release 3.5.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9344102614a3aded9a44ba6de06e2e501ab95059"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:41:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:41:26.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 3.15.1","message":"update constraint for python-keystoneclient to new release 3.15.1\n\nChange-Id: I39eb7110808ece9defc225d465a3af57720c261f\nmeta:version: 3.15.1\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45353cdcb985613556cf0a7dcbdd09ed7fe7f8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45353cdcb985613556cf0a7dcbdd09ed7fe7f8ae"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
