)]}'
{"id":"openstack%2Frequirements~692274","triplet_id":"openstack%2Frequirements~master~I0f730de5df622800bdcff03e4bec1fe2daddabb7","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I0f730de5df622800bdcff03e4bec1fe2daddabb7","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-10-31 06:36:24.000000000","updated":"2019-11-01 08:42:36.000000000","submitted":"2019-11-01 08:40:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692274-1572597640651-c91a571c","meta_rev_id":"43b4d013148cae44c7a47c6af908c1b7aa2071ff","_number":692274,"virtual_id_number":692274,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 08:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-01 04:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-01 04:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 04:39:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-01 08:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c004aaf844b19303b42b02fe3de0f9dbcc85b130","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-31 06:36:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f389290c7ac090599292c36eb9d6c07f3455330c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:12:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70cba86cfc444bad90fc25be569439f4 : SUCCESS in 4m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bcd1041b045d4af49da957dfe106755b : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7f1cb46f78484197ee5fc5f26e2c1e : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b070a5dba4b0407b880a3c4863d7f8b0 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7cdb341d3584ae1b8c5b0e89a01be62 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/328b2f0d3ea946e581c017b631879259 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bca65d255e5e478ebfccc0ff3e92dac2 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0f8d5a37486461d827fb502a94aed88 : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9d424d47045a4df6a70cc870d145f986 : FAILURE in 9m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d078f27cc9d48aab6dbf9153bda815a : FAILURE in 7m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7e272cb626e042349876f093a0f635ea : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3baa1120e3346dcacb9837b280f26c6 : FAILURE in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9bdf500ebed941c1acbf34fe7135f224 : SUCCESS in 4m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db2e700df5bb43ec9846809671d901f1 : FAILURE in 6m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f95db47e69e04f29b297614026d76355 : SUCCESS in 11m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee84645de3d2467b85ccb138fd57d40f : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/301a056aa5304ca38482cf78b851a36d : SUCCESS in 7m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a0fee490bf224bbeb3c9a5455c7ee5db : SUCCESS in 19m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/144fcd085ee548f28d3522ffa9aec1b6 : SUCCESS in 19m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cfddc9f67fdb45fb9feb45bb44f5ba80 : SUCCESS in 21m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/50617f90ae31451d9ce56d31a385159c : SUCCESS in 18m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/431cb02f177543c387a91fa0f08c1d92 : FAILURE in 6m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5db6a45e52754ae58b7ccff3bf49dd4c : FAILURE in 7m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/03e5f7f6e1034ca1bbebeb265bbbe08b : FAILURE in 9m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/862d2d07be974cccab2604eba2093f5a : FAILURE in 6m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3df787db016461d8c89bc2364eb9fd9 : FAILURE in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5b4eb2b258584c60adf1717fd92b63b5 : FAILURE in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/978e51e1962341fdb66b28a068dfb968 : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/664a370e72844611b65ee8515d725010 : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a2a40c4d7d6947f28cd37dab087e2256 : SUCCESS in 10m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fad0d2e666e34625a3891be6c8cca208 : SUCCESS in 1h 16m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f6f486b560b47e3a45575093544a36a : FAILURE in 1h 04m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc63d08026274542903f3ff8ec2f5548 : SUCCESS in 1h 39m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4a8d1f588c7a488c9d9fd3299b39614b : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bff66e203b301b927a41711e12ebf24613fad930","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 15:03:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"165f0609e5feca0a69b48de3175fbb5f3ab6a085","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 15:04:07.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"81aaf8563b5f8a46338a9e326f36ef718036011d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 04:02:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df336ae6e0c44497b6a380fe78fa766a : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5951e195872140c8bce7a27a38adf5b1 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fc91a514b644dea3828e6345bf51a3 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f91308da3e3c48e291e80eceea837782 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfa078a015a94b3896a82dec6eea36d0 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64785a9c72f245318b422be19b7a0088 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8e95f32153146dd95d6bbb56036ac8b : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2624162042124380878f242a34e33d8f : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/076f2c32d1f04bfda81006174753bc37 : SUCCESS in 7m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2004020759ab474c9fc96025d157c394 : SUCCESS in 6m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/adf865b966a549b29287a89a0daa0496 : SUCCESS in 9m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c54f360d81b14dac953e091e4faae709 : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/abf1794a3b3a45ef954751edb4a6f2c1 : SUCCESS in 3m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49f16700c9b446e4b09990c39761adb1 : SUCCESS in 6m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/54036d22fea64047bbf429fffe2ae446 : SUCCESS in 12m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa6ea6e3cda3485d84b0de7c32391e8d : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6dc10ac2c3e4557a3025906e012aa57 : SUCCESS in 6m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/88413bbad24b48499bf39c0bb46707c1 : SUCCESS in 15m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c6a0cc200ca4e7596be873095c048ee : SUCCESS in 24m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1b14891f20d94be7ac30eee39ab15691 : SUCCESS in 24m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e41a525cccc54f8295f6ddb3335f2323 : SUCCESS in 19m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a44d4c3bdcf64bc498c2fa130ff20ae6 : SUCCESS in 12m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b89f678bfa5d4f26b2ee43eba7dbc9d1 : SUCCESS in 11m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f493488439d24dc885fe6765ae18278d : SUCCESS in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6001dd4d07b452598be31e19c66d41c : SUCCESS in 18m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5c39afe7f0044c8395b19fd8352b12b9 : SUCCESS in 4m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/439ae7512be9484a8731c3ae43ab6957 : SUCCESS in 12m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a81daedd950e4b83870cb1c6f5b0a610 : SUCCESS in 8m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/29a8b47b18034140b9cc25f28c2aaa31 : SUCCESS in 9m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8dfc94018acb4764bc636333302effcf : SUCCESS in 10m 54s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/90d365774dda477eb3db471c6be77d94 : SUCCESS in 1h 08m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4a6ba82e47048b4a2af0d0984024d1b : SUCCESS in 1h 04m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ab72bd543cb544ffb669ffa29f690d83 : TIMED_OUT in 2h 26m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/031adfc0e60f4edbaadf4b3d851e9578 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1073f28f7f4b3dada3562aada13d65e5a7de5d6c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-01 04:39:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"11396b3d6a9b64907b14cec63c00428f133d51d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:38:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1decb2093a45aeb25e70d438cf78a9 : SUCCESS in 3m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/860cfd2cff204cb7b10bb07b793300d0 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c780e3b76c4d4eb228080059c6aef8 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acd02b2cac954d7795baf8d53684127b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e987034ff69a4774a180e451d186fdb1 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/575be696c15d4c8a8fa2ac99004f2f21 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93f566c2140a4becb2250d6ecfbc1458 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/512b5fdc52274a4d90188b6e41a34ca8 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/363f5e9becd443938450b23618f06353 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a3dc8cabb6de41888499bf8b9fee83de : SUCCESS in 8m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/abf16c72b53f4c30b196246d4c1c4813 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/463af0ea4ad94b6081ec0aebf8df80ad : SUCCESS in 8m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0e8a37b38530497dadfde5cb9fa345ab : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb33d670ad0d43c79d4f2fc276a8fc21 : SUCCESS in 4m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/aad1d55f31c649ea87b4fbec0c689d9e : SUCCESS in 8m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8f16fa1294342ab832bca9179f1a101 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09870b40d4664871987015034fec4980 : SUCCESS in 6m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a50f1203c7e349aa977b5a0df9e038aa : SUCCESS in 21m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/347b9bea2be24a958d00c6ad012b6e55 : SUCCESS in 19m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f2fc1f34bd4245fd8a495c3987799985 : SUCCESS in 18m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/157edbf9e5db43e78e0d92cc17432fb9 : SUCCESS in 21m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c9bab33fe0ca45778f30d29d0dcb7f3f : SUCCESS in 14m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1524810346d451ebf1b90879b9852bf : SUCCESS in 14m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f337a2d2fcca425c9e68282917af226c : SUCCESS in 14m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6b47cfc2a6c42fc878f0e0904d9d42a : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8300b3a2028a4d5392b6be07c89c51d0 : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/883d57c7776140ee903c3f12bd930f8c : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c56cb4f03e8f43929d1170fbeb00952b : SUCCESS in 5m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f3c07627235b4276a1ef65d7d0b858d0 : SUCCESS in 9m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c989fe30c99d4026ae9854ccce9c9ce1 : SUCCESS in 10m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3ee9b7f5e8fc44db91c40f020a5ec86b : SUCCESS in 55m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0a34d74012a456aa68967ee9f8c530c : SUCCESS in 1h 03m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cbcee993f44e4a9fb1ed5bdbe502848b : SUCCESS in 1h 34m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb87bbe2bb8245afba2feb1c95a8abef : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f6552fbea97d39780fcbe866b5983327329cc400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:38:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"88e7901e0132a53d6a8907099c3b6d1813b3e2b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 08:40:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7e18c477db6f8e0ae233e8614ca9bc2468bbd38c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 08:40:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f700c445bff24069abe1fc7f4aeb1657 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f096fef9754f4ea84019088a30c3e7 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3834d79d8484c76bf5fcb674293c6ee : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33d8e56751194ba295dacb260236f7ae : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd29f667dd9f4545a03a4ef4d065316d : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d699bb5773f4e12b341739aa1a43873 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c28a49fc4ccc472195930405d4adb1b3 : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9c744284f7324f36881133fee72cb97c : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/885c9f074b034b3688cee679c4ec6500 : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/42e9b6c04f2d43bcacc08592e5101257 : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f1fa4bf839641c0b99dcf9636c4ee20 : SUCCESS in 8m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/76a94db22b6947e2b72a22de11cdd450 : SUCCESS in 5m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2cee323f3457406280a664db65297e3b : SUCCESS in 4m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c817bd2953574ae295164b47ed1b1f87 : SUCCESS in 9m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e269df9effe1493195a434809481b61d : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed0ef69504094a18be94296f5b915a2d : SUCCESS in 7m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5fcfa84c92b9412fa4cd98db0e9afbe8 : SUCCESS in 16m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8020cc1a13494179893018d9537eb6dc : SUCCESS in 18m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/25a2d454268a41099f767071d7bbc9ca : SUCCESS in 19m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52bda8788d56435a84f514d692bd9a05 : SUCCESS in 23m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5ca4b2e018514dc4b58bf5fde9d6abd1 : SUCCESS in 12m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf69ea65341847918563f4e9516aff06 : SUCCESS in 15m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/477bb2e04f77498eb37aac1c67a52797 : SUCCESS in 15m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f67343e444646d085f9de5314348d08 : SUCCESS in 16m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f68c1971abf249a194acc3e1593ff6af : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0c39f7646dc54e8c8002c4bf934d7e07 : SUCCESS in 7m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb74ab5770294ccca2575abe421be407 : SUCCESS in 7m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b2b91eb9e9364ea4962b734d7e6db096 : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/20fdd7466c76426c8e518ad5a65a02e0 : SUCCESS in 9m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6481e53c3767428c86b59ebaa927a534 : SUCCESS in 1h 16m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7342daaf313f4d10815c921213f93621 : SUCCESS in 1h 01m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67476b9b5cd1416fbac1283b5a7ccbcb : SUCCESS in 1h 50m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/72f92d40694c49d5a04083d4f6c04fb8 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":2},{"id":"43b4d013148cae44c7a47c6af908c1b7aa2071ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 08:42:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a77a69f965c4d06a8fe89ca4c332865 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ab1d033d17f6f979baf6a84e5b70614527dcafc","revisions":{"6eef2100a3de39baccdcb19cdb9da6c91d83e357":{"kind":"REWORK","_number":1,"created":"2019-10-31 06:36:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/692274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/692274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/692274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/692274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/692274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/692274/1"}}},"commit":{"parents":[{"commit":"4e7d3ce46bed5a7c9f267b49c05663f82400bb95","subject":"Merge \"block versions of salt that cap pyyaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e7d3ce46bed5a7c9f267b49c05663f82400bb95"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-31 06:36:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-31 06:36:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0f730de5df622800bdcff03e4bec1fe2daddabb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eef2100a3de39baccdcb19cdb9da6c91d83e357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eef2100a3de39baccdcb19cdb9da6c91d83e357"}]},"branch":"refs/heads/master"},"3ab1d033d17f6f979baf6a84e5b70614527dcafc":{"kind":"REWORK","_number":2,"created":"2019-10-31 15:03:54.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/74/692274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/692274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/692274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/692274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/692274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/692274/2"}}},"commit":{"parents":[{"commit":"4e7d3ce46bed5a7c9f267b49c05663f82400bb95","subject":"Merge \"block versions of salt that cap pyyaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e7d3ce46bed5a7c9f267b49c05663f82400bb95"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-31 06:36:19.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-31 15:03:44.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0f730de5df622800bdcff03e4bec1fe2daddabb7\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ab1d033d17f6f979baf6a84e5b70614527dcafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ab1d033d17f6f979baf6a84e5b70614527dcafc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
