)]}'
{"id":"openstack%2Frequirements~692637","triplet_id":"openstack%2Frequirements~master~I0329775744b90c0aa46cdfe91b2db4954221e0a8","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I0329775744b90c0aa46cdfe91b2db4954221e0a8","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-02 06:15:48.000000000","updated":"2019-11-03 10:27:31.000000000","submitted":"2019-11-03 10:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692637-1572776789121-752aeac9","meta_rev_id":"06c51564a828ae317ff10f114724a4a25b8f19f6","_number":692637,"virtual_id_number":692637,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-03 10:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-03 06:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-03 06:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-03 06:53:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-03 10:26:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"600e235e898f4f63f5d522e944ce7e4d7488fa10","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-02 06:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be1500ea0db1c60ce6be625554c41843411c72aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 08:01:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd84c66819f4b88b2d410ab9234475e : SUCCESS in 4m 47s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/05552b79c926424a912334a4b36058a5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93de713b707d48849659b33113654b89 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe8e6917e3f040678b41bc67b7a9369d : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ccec638aeef4a5b900354a5bb1e362e : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1d30586adc846e39a57c4a9aed48895 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dfc0b354df946688fa9ae4fbf4f4dfc : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c04877c081e04ae2a787de65aafc0339 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/06a30a14194e4ada8c9c8d7f9e7073d5 : FAILURE in 9m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d02c3da49100490c84e86adf8fe3012b : FAILURE in 9m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c66f1540e87b4ded94d784408324f5df : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6ac0cac167824aefac895b4dc98e773e : FAILURE in 10m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/82fccb634e38481fbe776e849b0f8954 : SUCCESS in 6m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d870e42070254293a5e2e6f5d263f644 : FAILURE in 5m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0468fc086c1649e7b07b114597b43b1c : SUCCESS in 12m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/95c9cf5a786e46f8a5f0d387e512a0de : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6fc3b167791419398db9c8ee0c9a860 : SUCCESS in 8m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/88a938ec9c304065a23a5fef74513c73 : SUCCESS in 23m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/769c9fb4941b4e678ef510d41d2b91ae : SUCCESS in 24m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9f7729c3308f45acba3d9fbb4fe27c2b : FAILURE in 23m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dcf66dfb8c4c469cb8149e817116b02f : SUCCESS in 23m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/88778bce22be49ca9667edab392df8be : FAILURE in 6m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b5e231f1cce8465fa16bd4a5a199bd45 : FAILURE in 5m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/01a3eb6d73044c82bbc5dcb2977ea60c : FAILURE in 6m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4efb704146754e6ebc44e96d7af93801 : FAILURE in 6m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/192b81dec27f4bbda771b01e0b1abd60 : FAILURE in 6m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/833147418f37423eaa83985ed28e6a9a : FAILURE in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c2284356f244f93a599155da6109cda : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c043f65df85414cade4f3016a0c2db3 : SUCCESS in 10m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f2cefd5dda5a48c0b02de8e918ed24a3 : SUCCESS in 9m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6048e5033d62442f869969253029ce80 : SUCCESS in 1h 05m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc96f8bf209e4eb3884b16f911d52c36 : FAILURE in 1h 33m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67b6c94017dd437b911a443587f6d118 : SUCCESS in 1h 40m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2375b362c2324143b6c2b6323e6a8975 : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d1b16f5a01e341489c888b03c7c76875dae82922","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-03 06:19:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e546d244fe1c9dad3518afb5df15060fb6aea3c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-03 06:26:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fc0fad79712955dd9dfe3bf66497253cb1069e12","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-03 06:26:08.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"8a02854204733edb79c067596cb934d1d090f288","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-03 06:53:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fd8ec2950d5ab71425e85ea9534333362e5a77b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 08:44:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8a418b9b3242469a32cc11adf41acf : SUCCESS in 3m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/54b50990f183403787b6a06a2a555808 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ce5c1ef4c9487694bb77ab83e86cf6 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d23b54ef695420589dd3f827f86eb81 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d5c5f63fe6f4e8fb5211fcbeeb10d10 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f00acf8674954e8b8eadab92040405ec : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf27eb2cbd354a728f3d4e86473d2dc7 : SUCCESS in 6m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d4dea6e1bc54ea78cda5677385cc35c : SUCCESS in 5m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/87de02f89d804b85bb0b4cbef41b457a : SUCCESS in 22m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93b123bc64684d94945bff3111beaa12 : SUCCESS in 22m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5c4b5987bb134a78848e47ee40d77ce4 : SUCCESS in 14m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d339e002bd994c74aef18852fa909515 : SUCCESS in 9m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/45c1a18dba0c4069b30fbd136e72cfc4 : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c18c24454c84373aa8589849bba0383 : SUCCESS in 5m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3f0c4639b860471ca65391f1af5e25f4 : SUCCESS in 8m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5e7a736b4e834bd1923ff9bab95e7a54 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb9407f9881f4f4f83ffb299e267d430 : SUCCESS in 7m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6a5b2a142beb4e55a24ad851f34a6045 : SUCCESS in 13m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/72b580462fe84540b1ee223ff9096dc4 : SUCCESS in 16m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/48b81f3c03834f53a9ddea489b224b64 : SUCCESS in 22m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/45c6bd424ab54aeb9a373acd1231992f : SUCCESS in 20m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ef040204f38b40589dc31fdcb42124e8 : SUCCESS in 14m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea1410c960304a7598eb979fc44f9e11 : SUCCESS in 13m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d6c2ce5dee1d4ea59eb27f277717662b : SUCCESS in 17m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e0aebfc901c4c519fe8e009d502cc4b : SUCCESS in 28m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/125773d04c2f468aa089354aba4ef5a3 : SUCCESS in 7m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b347744c1d3346fca74139640c4c1ff8 : SUCCESS in 10m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3d0da91fd75468ea2ab0f19c8dffabb : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/94a4d307c1af476c97b0bd9ab3e8e7f6 : SUCCESS in 10m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/56768bc3fe8049429fb78d942649007f : SUCCESS in 10m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9383b0de250b42a09414dbc49a6ac096 : SUCCESS in 44m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f597b89e313d407b86413e01574a63b8 : SUCCESS in 1h 02m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b14c262cb5ad46e38b8646efd5ff57c4 : SUCCESS in 1h 41m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/52962004e6334dee957799e224f66674 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b0ce707b86d7e6c40f213d5e935f3ef85d1b2db7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 08:44:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9cbf15c6cafb162bba7590dd9f624970d6817308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 10:26:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61250b372be4e288af4160d6ad5c177 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e2fa1475fb462e9868b7ed89ae3e45 : SUCCESS in 2m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2201131b8e849a48ccaf30194decd59 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a8daf0fd42f4311aacb4244b4f5d249 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9119dd0aa4444d89e71f8a3cf258873 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f13062ae969e4f7c89d4ba14b609c94d : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2869f77c22b74644979717bf8d1380f2 : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a6b9d7ba67f44094bc0785ddc1ac2083 : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/340c51e976614038a5be078918be72e8 : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f46d7e52cd924937ac8fc1ffee39715e : SUCCESS in 8m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/da607cdf62964c6ea26c93b546c87450 : SUCCESS in 9m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0ac453daebb7492aa0cf8e746a77e638 : SUCCESS in 3m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6abb0575dea44af79db57e6378fdeef0 : SUCCESS in 4m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/df3aec55c3ef472d9c14b33bf2ff5ccf : SUCCESS in 9m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5cdaff3938c347eda60f7e8176488752 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c46d35dbfed742cc8691585c49bf5de4 : SUCCESS in 6m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7b88fa9e2a204efabfc8583d85e1210d : SUCCESS in 19m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06fed657f736434cbc093f084ea6acbb : SUCCESS in 12m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e041e5fe895f491fa2a3f9150fe8f7c7 : SUCCESS in 23m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/880f16475e0e4f4888e1254e905dd213 : SUCCESS in 20m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2942794bc01349d3a354aaeae127e167 : SUCCESS in 13m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/025b975f2dfc4c07b55061a2cc62735d : SUCCESS in 11m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1085d925c95d43febb9cb2495d973208 : SUCCESS in 13m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20206c4e00da4201aad3b68bc3195642 : SUCCESS in 17m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b209c6b7a793474490cee1c6521a68c9 : SUCCESS in 3m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7c0ae45576ab48aeb5488251335b28bf : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb80ccc3e46d41faa9425ce954bc7a4e : SUCCESS in 5m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bb541b92660c43a6b420dddd69e24c34 : SUCCESS in 9m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a58ed8f74224f21b57dd3cf9153ecd2 : SUCCESS in 9m 19s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c733f71226b74ad9bc6c120a06141197 : SUCCESS in 55m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aaed5a13d16349289c9052886a001d18 : SUCCESS in 1h 02m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/45d8ddc63ad94f5baa2add13ed2dad50 : SUCCESS in 1h 40m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/19ea9edb436f4c0a860cfba6edc38bb1 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"b4af810f320bbfc0a548173fa0c393397d2c131e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 10:26:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"06c51564a828ae317ff10f114724a4a25b8f19f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 10:27:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477a6b4ef72943dd824dfa6418f442e7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bdf703deed16c0acc42e5fa0d6ee3fc167ab6803","revisions":{"b80b6c9874e8ce947e2cdd3285b2d57892159433":{"kind":"REWORK","_number":1,"created":"2019-11-02 06:15:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/692637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/692637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/692637/1"}}},"commit":{"parents":[{"commit":"0a5c0fcbf6fca86e58d173b194112a9ab5fdef34","subject":"update constraint for nosehtmloutput to new release 0.0.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a5c0fcbf6fca86e58d173b194112a9ab5fdef34"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-02 06:15:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-02 06:15:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0329775744b90c0aa46cdfe91b2db4954221e0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b80b6c9874e8ce947e2cdd3285b2d57892159433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b80b6c9874e8ce947e2cdd3285b2d57892159433"}]},"branch":"refs/heads/master"},"b9a2e3242ce67a3f098564b0238318720c0d43ee":{"kind":"REWORK","_number":2,"created":"2019-11-03 06:19:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/692637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/692637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/692637/2"}}},"commit":{"parents":[{"commit":"0a5c0fcbf6fca86e58d173b194112a9ab5fdef34","subject":"update constraint for nosehtmloutput to new release 0.0.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a5c0fcbf6fca86e58d173b194112a9ab5fdef34"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-03 06:19:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-03 06:19:28.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0329775744b90c0aa46cdfe91b2db4954221e0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9a2e3242ce67a3f098564b0238318720c0d43ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9a2e3242ce67a3f098564b0238318720c0d43ee"}]},"branch":"refs/heads/master"},"bdf703deed16c0acc42e5fa0d6ee3fc167ab6803":{"kind":"REWORK","_number":3,"created":"2019-11-03 06:26:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/37/692637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/692637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/692637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/692637/3"}}},"commit":{"parents":[{"commit":"0a5c0fcbf6fca86e58d173b194112a9ab5fdef34","subject":"update constraint for nosehtmloutput to new release 0.0.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a5c0fcbf6fca86e58d173b194112a9ab5fdef34"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-03 06:19:28.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-03 06:25:52.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0329775744b90c0aa46cdfe91b2db4954221e0a8\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdf703deed16c0acc42e5fa0d6ee3fc167ab6803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdf703deed16c0acc42e5fa0d6ee3fc167ab6803"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
