)]}'
{"id":"openstack%2Frequirements~692819","triplet_id":"openstack%2Frequirements~master~Id0a8de7917bb91608b16953ed63722ac2d061ac3","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Id0a8de7917bb91608b16953ed63722ac2d061ac3","subject":"Add new dependency for akamai API library","status":"MERGED","created":"2019-11-04 13:52:12.000000000","updated":"2019-11-28 00:17:15.000000000","submitted":"2019-11-28 00:16:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692819-1574900174107-ded4c5e8","meta_rev_id":"0e2cbaceb9d000b6a7a6e1d2398a94c0c4ec19f3","_number":692819,"virtual_id_number":692819,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-11-12 17:51:36.000000000","_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 00:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-12 21:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-27 21:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2019-11-17 11:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":1,"date":"2019-11-13 08:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-13 00:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-27 21:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 21:01:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-13 00:40:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-11-13 08:44:11.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-17 11:32:53.000000000","updated_by":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2019-11-27 21:56:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-28 00:16:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6d64039f7e78ca5dce766d65bd3cb00660ccba7","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-04 13:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8b3cd19bfff02dd2a88a04e3d61dbe9de6b1733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:52:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9454cfaa8e644910a3b50c767845371d : SUCCESS in 3m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/25611cfecb3b4baca3506432d4087f8b : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1714106ffd48d783661ad2d34e71f3 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13223b3443644d418674fdeea896a043 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad6175760baf4b32afe7c72d1d14a314 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2c7bd15992b46a094874c051b64ffc9 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4865618a69684e67b08f21ca14d3f015 : FAILURE in 3m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a507a2d704fc477dad3d5637d27ace7d : SUCCESS in 55m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60fbaa3600a34203b0a4c2ccdf744a67 : SUCCESS in 59m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9ed9d96ac6934705b1265b062c7855b1 : SUCCESS in 1h 54m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/70ee0cd77b2740748fe793ed6eb10b56 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1b09aa0b8b9f275419c815940b047e51ed46f69f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-04 20:26:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41b2e645874119c7fad6b467f13dab50b987bd72","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-05 20:34:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"507c47849f17831ea33dc6af70012fc4a29c8dd7","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-05 20:34:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6b82b05c085a86b5159dd91ea3dd500d7e8bc9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 22:24:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83023ddb68349c3bedd9157d5df3f51 : SUCCESS in 3m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b5051bbe08894826b059342488caf688 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d732d791a64980b51ca6a76bc1aafb : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8df0d53b75c243249ea401b7298ad202 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/355ee424f09940a7abf55f97f30c7be3 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f62ff8602ff42dba06b92128169e397 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b3f5ba806d9495c82a9758d7689b1a4 : FAILURE in 3m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/513f56a26acb452aa4b7dabdbe9f51fe : SUCCESS in 56m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f31623c2895146adaad020e2bdfbb9a2 : SUCCESS in 1h 12m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0998b445915b461ba42b389f6f8ed7b9 : SUCCESS in 1h 45m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/15b222a505e7408d988f2e04b3fbe579 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"55eb3bbf9f5cfff888500921d7c30292268f7a4d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-10 16:54:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf0344f7d23bcb23f5243ad713c6e4e79b062105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 18:30:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48cbf88aa8ae482eaddb26d4f147dcf2 : SUCCESS in 3m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/217b632ba5e945c4bb815ea7c468b034 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6f5cab50d5431dade5658079c09121 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7181b92ab7a0490bae6c1c1591fe8ec3 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c067cfc198e4cce9e9c2e719691d4df : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/164a2d851dda45f7ad9d4078f4a7911c : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac36a99d00be4e189abbd11d487e9a66 : FAILURE in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c66bac2b43714b3faca4cd954a475051 : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/16e2c3eb21ef4a02bf37e0b31124599a : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/07eb7eb76ac0439a907b923a97d767ba : SUCCESS in 7m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/afdc299e4f024aad933c90c523574836 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3675ea44acb3403791d0ea87b9d3eccd : SUCCESS in 7m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c3921785116e4080b99f27dab2c90e2d : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/806803aa634e489b878c8b6fe695c397 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e2512bdf350f4167b787a991f9aa7013 : SUCCESS in 9m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6d2d094a43434b75a352b779ecedb1ab : SUCCESS in 7m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d643bbd880ce4d309ddb4d6b09b570aa : SUCCESS in 7m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a65ad39152984cd8bc3a85eb851385d0 : SUCCESS in 18m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1cfb43911be4de19dbdaf767a708d34 : SUCCESS in 17m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cf4167ccc2b641c1a686c38b6e5c1796 : SUCCESS in 18m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1a1289ba1fb445a813c07953bdaa926 : SUCCESS in 18m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/718078c5edb94b30be0aaebe7eb2104e : SUCCESS in 11m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/87fe6147d1c649ed97641aa59a7298ef : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9c9a302958a84fd6a1b6cdea478f1cb5 : SUCCESS in 13m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/296c3bf43c234fb2b285ec18ea9bf85e : SUCCESS in 15m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/35972276065340778e1feed1526b97ef : SUCCESS in 3m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c2c2e8f3203046ddbac83da1c01eb645 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d758930871d431494f60b0e226ed13a : SUCCESS in 6m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5703cb343c78499d81ab433357931017 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f87706ac116c4cd4b7cdef70cd12e6ed : SUCCESS in 10m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cd5fe5da60cb49cda73a4563d39bb803 : SUCCESS in 48m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9265c9c305a5434b8f7078aa268f5bd8 : SUCCESS in 1h 09m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2bfe36b09f9d4adaabece66930cade8c : SUCCESS in 1h 33m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3936db2496894d97a0723c3b067e03c1 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":3},{"id":"9147ac30edc02c8bc3fad8ce29af579710e262d4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-11 14:43:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0ea840e164c9d81a13c2d8dfbe3049eef4091022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:13:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4fa6f39a454b6fb35b796ef557799f : SUCCESS in 3m 48s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/25055611d397444c92ec5d7c9007a16c : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc3ac303fb641da9497d2d107b36d8a : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abe9ac88a5de442cbf2aad0c48154a0e : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c07debf9e4264f84883b4689f12ad003 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e7bf462395944d5a7a0291de7462ef1 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a09d811cebdd48dda474146c84cc4dec : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7d75fb04f0347878b0231ee4f85b551 : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1ee9ed4a2a244964bca1a1569392e65b : SUCCESS in 7m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88444e870b4a4a789bfbbf54850092b2 : SUCCESS in 6m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1773205830b948ef9f02e99ce7520e3f : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a6492870bdb47cbb9293001c1b26096 : SUCCESS in 9m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/029ae4c8a182466aa8b67cce7c730f91 : SUCCESS in 4m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/072633896f89420d9d5bd4066a8e77f5 : SUCCESS in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/75245753de8d48bc98b0872781431bad : SUCCESS in 11m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f9e27ae418a4e0caf32898f04dd904b : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80257890ecd04e3ea7eedb7079ca2f2f : SUCCESS in 6m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/233a6dfc203d4292b5bf7d8289b3e62e : SUCCESS in 14m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ca01451bfc374328bbda49b69016671e : SUCCESS in 12m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/508f857ba868444a802a282a560d09e9 : SUCCESS in 24m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1df856575b1d4b40b799dba4dc5ff28c : SUCCESS in 26m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0cb403f90608415b97d81e7ff41c3604 : SUCCESS in 15m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5e59dc0a3b924e4e9886dea55e8f2734 : SUCCESS in 14m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4acde82fecde424ea59b82754ceb3656 : SUCCESS in 14m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4cf266ea1974d578217152077301e1a : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/186eed9372094dd3a9e70632168ab4e8 : SUCCESS in 4m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1c4609f96aff422fb76b011cdc8c083a : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e58d3853a20c4eb685ffeb3305d26e3e : SUCCESS in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6e1b09b1fcc04c4aaf226cb8cb68c60b : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cf3c92313d184dd983871e576cbaa16f : SUCCESS in 10m 21s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b24e01aae1fc49349caa38acc98cc889 : FAILURE in 34m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22fff8a6800d4f10b5b59418f290bd72 : SUCCESS in 1h 25m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67c13259dca24d4c9c425e2f6d3efe40 : SUCCESS in 1h 25m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c8a7dae88d41472b876807480cfca2b3 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":4},{"id":"b266d93143b84d9d348e0bd2021a1f75a3c541aa","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-12 17:51:36.000000000","message":"Patch Set 4:\n\nreverify","accounts_in_message":[],"_revision_number":4},{"id":"fdcfe2e1705c0e52cc21258cb91e761c807f5a6b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 19:16:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c864459e592277e8c76e874ee492ae08df6016df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 20:58:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d011d65fa743a3a8a8e072735fa99f : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/46262a08aefb4c9097406d3454e5d18c : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7795601e2a74ee39ffa773b4cf10793 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a569d222dc89468f9ef27d04032b349a : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc53983d241c42ea84616dce483ebdce : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d374ff7884a4b479075cb36ee63cbf8 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adfa77469fbe43eb8d0d15287e2b383e : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f53772388f8b4e5ab8f4162497fe9b17 : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e55396ad68134c5ea6ad757fab1e98be : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8262ba4e1d04d6ebf8338eb08bfa0db : SUCCESS in 12m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8386e1864c594483ac65384964286b2e : SUCCESS in 16m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/afdad1d0f53f4026b0c39eb4e8acf223 : SUCCESS in 9m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/afdb236bf4b2404a8bb29fc0cd7fa334 : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3703e38769924a65a6326188fa2f209f : SUCCESS in 6m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7bdc3401fc8a41aaa70ee522be21b423 : SUCCESS in 13m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fddf36a0b9994c40bfaaca76a2f51b61 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/047a44cdc5ea4468ac3fade6fb948066 : SUCCESS in 7m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1c01646774294aa19118c5fe1a1fcc9a : SUCCESS in 32m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f8b93cb4ae945cba9922c91691fa075 : SUCCESS in 15m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8fb503b0bdd447558c2e8ea478502d66 : SUCCESS in 28m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8fc105fe4e346a7adb152da7dabcfc1 : SUCCESS in 30m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c681d5b72e0b4f848dc0f5c040b79c0d : SUCCESS in 14m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1cb499ca9e6a4fa884eb9c59494ea2a7 : SUCCESS in 21m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0e205dfbe0e44dc4a80aca065af6c99c : SUCCESS in 22m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e6f2edca335407cbc4cd34e26c66dbc : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f436d43a2fd8434b8f1568e44fcbff02 : SUCCESS in 4m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2b9740a5adc84bc688e89f0e08a3ece7 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f8a2047566a480b9a444c809c92d70b : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d6f3274fbe64acda64848b845878a53 : SUCCESS in 16m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/32a45e2d301b467190c27406c2723b4b : SUCCESS in 15m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8c9cbb3b52664cc09a4335dbf8bbcd2e : SUCCESS in 57m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/303a37d318384fa18e2b38db3d7c97b6 : SUCCESS in 1h 08m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/981faa4b809f438a9479d4f2a92f04cc : SUCCESS in 1h 39m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8c966381e2754d738fc913fa7726f7ac : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":4},{"id":"abb04e451b06bdd6fa4ec38cd45e0877ec1b7e83","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 21:01:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e69f0bcfb3f52102a7e84bc679b41eb83bc5815e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-13 00:40:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c79cfd336e7dd323240bf3b910289a20d37d4fbf","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 08:44:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1dcfb521e4daf7677f5dc3a4584a55a6797e1606","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2019-11-17 11:32:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8c326bfcc421c862796bbce647dc86d37cb7de6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 21:56:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7981bd3a13a66e7d5773b61502f7d18b8b4b0cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 21:56:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"49eb8db162111eebc2b4f165b504bda9d10fc54d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:16:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3dbf32655a4aba8b632081ea37fa2b : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e994603d3ac6478bb7ff1b51d92f27e9 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e8ca49a2f6345fa9b52fa8a286aa421 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d9ccdbb62dc4c4c8f32451d1718c9d6 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fceba7229e884921a7e808f368bdb99f : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99c4413517e0457db987e85d7b989f17 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9b58d7fca864001b941fc51a4aca7ea : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/52bf396987374863851cf4f5aece9fcc : SUCCESS in 7m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a6aee663a59e4cd2b727d7c726b90b61 : SUCCESS in 7m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3ce0389148d94aaa855cc45f14c2cd10 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ed63237956aa4ee29bb0499496f34134 : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8ce06d6cb1784813a037e796ac1fe8f8 : SUCCESS in 4m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cae0eb3fefd041468e301b15f36c7045 : SUCCESS in 4m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/52fec80356c84f74b6ea02cda9a32e89 : SUCCESS in 9m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9d089427fe2e4190a44307f46a5af789 : SUCCESS in 7m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcf636f6667044b1925a01de20a04f9d : SUCCESS in 7m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1386a97f1f3e49cebe737d0e16c0140e : SUCCESS in 17m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a539ffbd282f41d6a19a7df788b6023b : SUCCESS in 18m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9423c6e6665a425c8d5be80b84cfe455 : SUCCESS in 19m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b254f24095034996b6a40049d7a82e85 : SUCCESS in 19m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8abe2a12037e4fd49a9c4e89b91a11c7 : SUCCESS in 12m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46cd749e3c724a30bdd59da917919c33 : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4030e960b66b4cea8a915881c19f64d6 : SUCCESS in 13m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b2473f2bfad480e9ccb10e408b37f73 : SUCCESS in 14m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7eead8c89fa34caaae5eabd94208ba58 : SUCCESS in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b1b07b77b5924233a903752331722915 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1807b83cee964ade9bb329fb64f8954b : SUCCESS in 5m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6623af8bf15840b695b2f7cc9c60ef28 : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab1cea2ef553462391934850a48b8fe2 : SUCCESS in 10m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fcfd9d4690204a299734cbd8c66bc068 : SUCCESS in 1h 13m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b7d3fd3616514851ad0fcaa48b40d100 : SUCCESS in 1h 22m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/783a376d05c74fa8946677f8c29aa8c1 : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":4},{"id":"095e7b6b53c7e14c75aa1326b6370acf9055a5f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:16:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0e2cbaceb9d000b6a7a6e1d2398a94c0c4ec19f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:17:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6839c339c1ef4b1bb7bd17dc19464dae : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"847f59c69a1416212f215d2a3c13563b12508abc","revisions":{"05a8416628029893f7ad05bdfa4b8fc439ca09de":{"kind":"REWORK","_number":1,"created":"2019-11-04 13:52:12.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/19/692819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/692819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/692819/1"}}},"commit":{"parents":[{"commit":"bdf703deed16c0acc42e5fa0d6ee3fc167ab6803","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdf703deed16c0acc42e5fa0d6ee3fc167ab6803"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-03 17:18:40.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-04 13:52:10.000000000","tz":240},"subject":"Add new dependency for akamai API library","message":"Add new dependency for akamai API library\n\n* Is the library actively maintained?\n\nThis library has not a lot of commits per day, because implements\nAuthentication handler for requests that provides the Akamai {OPEN}\nEdgegrid Authentication scheme to work with Akamai API.\nSo it\u0027s stable enough and last merged pull request was in this year\n(28 Feb 2019).\n\n* Is the library good code?\n\nYes. It\u0027s simple enough and has a batch of tests for the existing code.\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/tree/master/akamai/edgegrid/test\n\n* Is the library python 3 compatible?\n\nYes. According the history it has necessary support:\nit\u0027s support pytho3 since 1.0.4 version\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/blob/master/HISTORY.rst#104-2014-11-05\n\n* Is the library license compatible?\n\nYes, library is under an Apache 2.0 license.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nIt has not package for Ubuntu, but there is a package for Fedora.\nhttps://koji.fedoraproject.org/koji/packageinfo?packageID\u003d28118\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nThere is no alternative libraries can be used to authenticate with Akamai Edge.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nIt\u0027s used for implementation one of backends for Designate project.\nMoreover akamai fast dns api v1 is deprecated, therefore a new back-end\ndriver is needed in Openstack Designate to support akamai fast dns v2.\n\n* If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nN/A\n\nChange-Id: Id0a8de7917bb91608b16953ed63722ac2d061ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05a8416628029893f7ad05bdfa4b8fc439ca09de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05a8416628029893f7ad05bdfa4b8fc439ca09de"}]},"branch":"refs/heads/master"},"1fe8c560e14612d4176f69e1cf552f3d13a4a3b6":{"kind":"REWORK","_number":2,"created":"2019-11-05 20:34:27.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/19/692819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/692819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/692819/2"}}},"commit":{"parents":[{"commit":"bdf703deed16c0acc42e5fa0d6ee3fc167ab6803","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdf703deed16c0acc42e5fa0d6ee3fc167ab6803"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-03 17:18:40.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-05 20:34:09.000000000","tz":240},"subject":"Add new dependency for akamai API library","message":"Add new dependency for akamai API library\n\n* Is the library actively maintained?\n\nThis library has not a lot of commits per day, because implements\nAuthentication handler for requests that provides the Akamai {OPEN}\nEdgegrid Authentication scheme to work with Akamai API.\nSo it\u0027s stable enough and last merged pull request was in this year\n(28 Feb 2019).\n\n* Is the library good code?\n\nYes. It\u0027s simple enough and has a batch of tests for the existing code.\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/tree/master/akamai/edgegrid/test\n\n* Is the library python 3 compatible?\n\nYes. According the history it has necessary support:\nit\u0027s support pytho3 since 1.0.4 version\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/blob/master/HISTORY.rst#104-2014-11-05\n\n* Is the library license compatible?\n\nYes, library is under an Apache 2.0 license.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nIt has not package for Ubuntu, but there is a package for Fedora.\nhttps://koji.fedoraproject.org/koji/packageinfo?packageID\u003d28118\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nThere is no alternative libraries can be used to authenticate with Akamai Edge.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nIt\u0027s used for implementation one of backends for Designate project.\nMoreover akamai fast dns api v1 is deprecated, therefore a new back-end\ndriver is needed in Openstack Designate to support akamai fast dns v2.\n\n* If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nN/A\n\nChange-Id: Id0a8de7917bb91608b16953ed63722ac2d061ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fe8c560e14612d4176f69e1cf552f3d13a4a3b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fe8c560e14612d4176f69e1cf552f3d13a4a3b6"}]},"branch":"refs/heads/master"},"ceacdf653b714a02fcfe2b9edbfdbe23065e43e3":{"kind":"REWORK","_number":3,"created":"2019-11-10 16:54:39.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/19/692819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/692819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/692819/3"}}},"commit":{"parents":[{"commit":"bdf703deed16c0acc42e5fa0d6ee3fc167ab6803","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdf703deed16c0acc42e5fa0d6ee3fc167ab6803"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-03 17:18:40.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-10 16:54:24.000000000","tz":240},"subject":"Add new dependency for akamai API library","message":"Add new dependency for akamai API library\n\n* Is the library actively maintained?\n\nThis library has not a lot of commits per day, because implements\nAuthentication handler for requests that provides the Akamai {OPEN}\nEdgegrid Authentication scheme to work with Akamai API.\nSo it\u0027s stable enough and last merged pull request was in this year\n(28 Feb 2019).\n\n* Is the library good code?\n\nYes. It\u0027s simple enough and has a batch of tests for the existing code.\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/tree/master/akamai/edgegrid/test\n\n* Is the library python 3 compatible?\n\nYes. According the history it has necessary support:\nit\u0027s support pytho3 since 1.0.4 version\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/blob/master/HISTORY.rst#104-2014-11-05\n\n* Is the library license compatible?\n\nYes, library is under an Apache 2.0 license.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nIt has not package for Ubuntu, but there is a package for Fedora.\nhttps://koji.fedoraproject.org/koji/packageinfo?packageID\u003d28118\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nThere is no alternative libraries can be used to authenticate with Akamai Edge.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nIt\u0027s used for implementation one of backends for Designate project.\nMoreover akamai fast dns api v1 is deprecated, therefore a new back-end\ndriver is needed in Openstack Designate to support akamai fast dns v2.\n\n* If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nN/A\n\nChange-Id: Id0a8de7917bb91608b16953ed63722ac2d061ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceacdf653b714a02fcfe2b9edbfdbe23065e43e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceacdf653b714a02fcfe2b9edbfdbe23065e43e3"}]},"branch":"refs/heads/master"},"847f59c69a1416212f215d2a3c13563b12508abc":{"kind":"REWORK","_number":4,"created":"2019-11-11 14:43:48.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/19/692819/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/692819/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/692819/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/692819/4"}}},"commit":{"parents":[{"commit":"bdf703deed16c0acc42e5fa0d6ee3fc167ab6803","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdf703deed16c0acc42e5fa0d6ee3fc167ab6803"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-03 17:18:40.000000000","tz":240},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-11 14:43:45.000000000","tz":0},"subject":"Add new dependency for akamai API library","message":"Add new dependency for akamai API library\n\n* Is the library actively maintained?\n\nThis library has not a lot of commits per day, because implements\nAuthentication handler for requests that provides the Akamai {OPEN}\nEdgegrid Authentication scheme to work with Akamai API.\nSo it\u0027s stable enough and last merged pull request was in this year\n(28 Feb 2019).\n\n* Is the library good code?\n\nYes. It\u0027s simple enough and has a batch of tests for the existing code.\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/tree/master/akamai/edgegrid/test\n\n* Is the library python 3 compatible?\n\nYes. According the history it has necessary support:\nit\u0027s support pytho3 since 1.0.4 version\nhttps://github.com/akamai/AkamaiOPEN-edgegrid-python/blob/master/HISTORY.rst#104-2014-11-05\n\n* Is the library license compatible?\n\nYes, library is under an Apache 2.0 license.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nIt has not package for Ubuntu, but there is a package for Fedora.\nhttps://koji.fedoraproject.org/koji/packageinfo?packageID\u003d28118\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nThere is no alternative libraries can be used to authenticate with Akamai Edge.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nIt\u0027s used for implementation one of backends for Designate project.\nMoreover akamai fast dns api v1 is deprecated, therefore a new back-end\ndriver is needed in Openstack Designate to support akamai fast dns v2.\n\n* If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nN/A\n\nChange-Id: Id0a8de7917bb91608b16953ed63722ac2d061ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/847f59c69a1416212f215d2a3c13563b12508abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/847f59c69a1416212f215d2a3c13563b12508abc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
