)]}'
{"id":"openstack%2Frequirements~693161","triplet_id":"openstack%2Frequirements~master~I2091519f2527bdacb60cff39acf3092072d0f419","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2091519f2527bdacb60cff39acf3092072d0f419","subject":"Revert \"update constraint for python-novaclient to new release 16.0.0\"","status":"ABANDONED","created":"2019-11-06 11:39:52.000000000","updated":"2019-11-07 10:55:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56d1ec2e8e71d0623e673e89a51e3abff43cbcca","_number":693161,"virtual_id_number":693161,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-06 13:26:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"date":"2019-11-06 11:39:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-11-06 11:39:52.000000000","_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":-1,"date":"2019-11-06 12:29:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"date":"2019-11-06 11:39:52.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"date":"2019-11-06 11:39:52.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"date":"2019-11-06 11:39:52.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"date":"2019-11-06 11:39:52.000000000","_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-06 11:39:52.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-11-06 11:39:52.000000000","updated_by":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"reviewer":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"state":"REVIEWER"},{"updated":"2019-11-06 11:39:52.000000000","updated_by":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-06 11:39:52.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-06 11:39:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-06 11:39:52.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-11-06 13:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6967eba2865160504ef8ead26ba680f52e1c0c40","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-06 11:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc8d6641547c0663299e7754af4fe288233c13c8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-06 12:29:33.000000000","message":"Patch Set 1: Workflow-1\n\nA fix for heat has been proposed at https://review.opendev.org/693165 , so likely we can wait for that.","accounts_in_message":[],"_revision_number":1},{"id":"3cb343d7647c1eef69189fdd74ecd1f63abfe2c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 13:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2d2a2844e349508fc40929c8e9e526 : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e90d0861d2ed44dab62e563fad043ff7 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35e860b56a74dd094c5adbc24bbbca8 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/adb1e6a586c34b8ca26d64fe37c77dfd : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/015f8ecf5e4c415da8afbcc6af6b4363 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acd8ca9c49fa4143acaf1a5ba6847904 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24f9ead52fc649878006fa019316975f : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28708bb3b7c34e6fb999145ae03e6135 : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c2e7f4e10b0d49b3b814b9105a4dff3b : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4195ea0cee941e29a708679f86a4a85 : SUCCESS in 7m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d08a4f30244943b8ae7cfab83f30f154 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b39ae5fd2f654fa4a9f2512e159d1cd5 : SUCCESS in 9m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9dffda948af94ce99a62a19e1c2d9072 : SUCCESS in 4m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fabb468d9c8646d388f2b9cc82e7643f : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f39650a49076409ea80ee39fa3d0621d : SUCCESS in 9m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ecb7018c4e5d4744a40d8a17dad4de7b : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9469fa2d74794d49b098ef605d3b498a : SUCCESS in 7m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eef7a06e3dec4a2bb056afce84daeb7e : SUCCESS in 12m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3f5a478225264389adcc1f975d740f49 : SUCCESS in 19m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/021b49fe669843928aaa42085eee5827 : SUCCESS in 25m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b0d870cc26384f0bb708a1ac274afab1 : SUCCESS in 24m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2e64131df9824186b1af7c6c812abba8 : SUCCESS in 13m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/411674a036ac4e22916962d561def156 : SUCCESS in 12m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5334a3144bc245e6b0feea87da5a1173 : SUCCESS in 11m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d60916b55af440e981c0d5c2b75cbc0 : SUCCESS in 19m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d371272e8bd8466ea6e2d2a1d24f7823 : SUCCESS in 4m 05s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3b13d92ecc7d4a41ab506abea8262fcf : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aefcca0e4ca414aa92074afea420d8f : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/78429cead0314570936dbde036d224b9 : SUCCESS in 8m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/152ede5068ad40299f9e746c1a54843a : SUCCESS in 9m 12s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4bbc0389346f4856b6de99745b37718c : SUCCESS in 51m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b59a57ac5646429aa372c7f37170fe9c : SUCCESS in 1h 10m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ad24f198d9a0402097af1d92a80d75c9 : SUCCESS in 1h 40m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/960346b3bbe74577b543d21811f943b1 : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"42d598f748c69dca9fe35d4f8b1160b01cfcb283","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-07 10:55:07.000000000","message":"Abandoned\n\nheat got fixed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab2528cd6718219fc830261dbf9fd505dbee49fb","revisions":{"ab2528cd6718219fc830261dbf9fd505dbee49fb":{"kind":"REWORK","_number":1,"created":"2019-11-06 11:39:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/61/693161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/693161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/693161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/693161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/693161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/693161/1"}}},"commit":{"parents":[{"commit":"7ff8d5d71481bd964c12f8639b5d49695af6252c","subject":"update constraint for python-novaclient to new release 16.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ff8d5d71481bd964c12f8639b5d49695af6252c"}]}],"author":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2019-11-06 11:39:52.000000000","tz":0},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2019-11-06 11:39:52.000000000","tz":0},"subject":"Revert \"update constraint for python-novaclient to new release 16.0.0\"","message":"Revert \"update constraint for python-novaclient to new release 16.0.0\"\n\nThis breaks heat breaking openstacksdk tests, see e.g. https://zuul.opendev.org/t/openstack/build/2c83d0168fda4384837b676d336694ed/log/controller/logs/screen-h-eng.txt.gz#1379-1410\n\nThis reverts commit 7ff8d5d71481bd964c12f8639b5d49695af6252c.\n\nChange-Id: I2091519f2527bdacb60cff39acf3092072d0f419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab2528cd6718219fc830261dbf9fd505dbee49fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab2528cd6718219fc830261dbf9fd505dbee49fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
