)]}'
{"id":"openstack%2Frequirements~693197","triplet_id":"openstack%2Frequirements~master~I2a5968b9a09529742af4b7057498a4fb7323080d","project":"openstack/requirements","branch":"master","topic":"add-sdk-job","hashtags":[],"change_id":"I2a5968b9a09529742af4b7057498a4fb7323080d","subject":"Add openstacksdk-functional-devstack job","status":"MERGED","created":"2019-11-06 15:21:40.000000000","updated":"2019-12-20 06:36:35.000000000","submitted":"2019-12-20 06:33:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693197-1576823605456-bbef5f2b","meta_rev_id":"90393f103b0b338e068c94d205ea6e4d7eaf0992","_number":693197,"virtual_id_number":693197,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 06:33:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-20 03:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-20 03:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 21:50:34.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-12 00:45:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-12-20 03:13:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-20 06:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe0048612c8e07df3ab2542523fb938e2a4410ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-06 15:21:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"391a7a755ad38154ddb4cd726c05c03deb181928","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-06 15:55:44.000000000","message":"Patch Set 1: Code-Review+1\n\nlike","accounts_in_message":[],"_revision_number":1},{"id":"1ee07cc576dafbe978a5c7d9cf84fc2f87b1ee4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 16:59:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c75c114678c460596451c47216b1c06 : SUCCESS in 3m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/11c6db20c3b64fc79aec73f09c02d237 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8406f7ee70094422a252dd2cbb286a97 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2cf84182bd44ee5a761fa6c662d8c1a : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fcad915537b42a6abb3cbd3dccbdf51 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8116b755907f4586a6cb08fcc921eae6 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69e62df8c1494d4da6d02f6d80956699 : SUCCESS in 2m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c83b061d7b84c68a248c6c905f200ed : FAILURE in 52m 12s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/096192cb0ff3450dae20a417e640f8ef : SUCCESS in 50m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/505c653de382426fadfc13ecb07ac1f5 : SUCCESS in 1h 10m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b052bd4631349d3b48e8a7d9dd26311 : SUCCESS in 1h 32m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/db9852ad28c248dabe7cd5d78d953c5a : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"839ccf9329476b3a29b3aca2f90391a2fd0b7595","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-06 17:09:34.000000000","message":"Patch Set 1:\n\nThis failed exactly the way we wanted [1].\n\nRecheck once [2] merges (and is released and required by sdk?)\n\n[1] https://zuul.opendev.org/t/openstack/build/7c83b061d7b84c68a248c6c905f200ed/log/controller/logs/screen-h-eng.txt.gz?severity\u003d4\n[2] https://review.opendev.org/#/c/693165/","accounts_in_message":[],"_revision_number":1},{"id":"569e831455bc44e92989d0af2e314f1862acb172","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-07 10:54:38.000000000","message":"Patch Set 1:\n\nrecheck heat fix has merged","accounts_in_message":[],"_revision_number":1},{"id":"b1890df7e59c13d41159d27214a512128a72fcb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 12:42:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3b980dd95240f0a06b5d78b34d211b : SUCCESS in 3m 27s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5b1c8b12cad94050b72390041ca03828 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa9d225a9524b44bdae3eeb36fa9dfb : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a818b9df84a43daad4288f9ba5079eb : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac11f0dfda0a466cb3727b8826f7a04c : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74b7925cd7e54edb8eaddd029c8a765b : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e73481d13ee94167bc9c1363800a3b53 : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b287256487f4b8d9e305427b165fa28 : SUCCESS in 1h 08m 21s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f672960cd1e2419380ba37d06272ae1b : SUCCESS in 1h 18m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d6025a598dc483396f85c0a4430ae2d : SUCCESS in 1h 11m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/737c0c0c57ce4e2ebafbcecce7cc4781 : SUCCESS in 1h 41m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ac8ff9d1d5b3456c9f597e12a03c18ac : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":1},{"id":"43557e1cca62b685726807a457c8378975e92641","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-07 13:34:33.000000000","message":"Patch Set 1: -Code-Review\n\nOkay, good that this passed since the heat change merged, but I want to think through what that means from the perspective of the requirements repo.\n\nThat is, shouldn\u0027t we need to be proposing a bump to upper-constraints that pulls in the new release of heat in order for this to work?\n\nI suspect (but haven\u0027t confirmed) that this passes because the job is set up to pull in master tips. So we may want to override it to have an empty required_projects...","accounts_in_message":[],"_revision_number":1},{"id":"5cc807af98f71543f4620853aade2c43318c6487","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-07 13:36:50.000000000","message":"Patch Set 1: Code-Review-1\n\nMy point being that sdk would still fail in the wild until heat is released and required by the consuming project.\n\nSo yeah, this is an improvement, but it doesn\u0027t go far enough.","accounts_in_message":[],"_revision_number":1},{"id":"3de321d975ae7e12bac90b7ec7116fbaaedc1790","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-08 09:28:16.000000000","message":"Patch Set 1:\n\nIt was the heat service that was failing because it was using functions that were removed from python-novaclient. The sdk job just exercises some calls to heat and thus fails if heat is broken.\n\nI don\u0027t think that the other jobs test against published versions of keystone, cinder, nova etc., so why would you want to change that for this one?\n\nOne could double check which version of openstacksdk is actually being tested though.","accounts_in_message":[],"_revision_number":1},{"id":"23620a43948ac4652ec287db53a2e432497f3e6f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-08 16:58:41.000000000","message":"Patch Set 1: -Code-Review\n\nMy concern is this:\n\nWe run CI against $random_change. It fails in heat because $random_change exposed a heat bug. We fix heat in master. CI passes. We let $random_change go through. But $random_project now starts failing *its* CI because the *released* heat -- against which $random_project is running, at least for some jobs, because heat isn\u0027t considered a \"required project\" -- is still buggy.\n\nFrom the perspective of openstack/requirements, this seems like exactly the reason to make sure the job is pinned to real released-and-upper-constrained (and even lower-constrained??) versions of the projects it touches.\n\nI could be way off base here, so I\u0027ll at least remove my -1, but (obviously) we need requirements cores to weigh in anyway.","accounts_in_message":[],"_revision_number":1},{"id":"f6c79e3ac092d018777b3d9fd849f1f6eb7df1e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-08 17:52:17.000000000","message":"Patch Set 1:\n\nThe problem there is that we don\u0027t have releases tracked in reqs (on purpose) for top level projects.\n\nProjects don\u0027t do releases often enough.\nIt encourages using projects as libraries (This is me staring at neutron...)\n\nWhat is generally recommended is that projects actually follow deprication warnings and remove depending on features that will be pulled.  If projects don\u0027t do that then this will happen even if we do pull in tags.","accounts_in_message":[],"_revision_number":1},{"id":"0122ffed5db693299b4b3394b77daa6f06559d17","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-08 21:50:34.000000000","message":"Patch Set 1: Code-Review+1\n\nThat makes sense, thanks prometheanfire.\n\nThis is certainly an improvement, so I guess we should go forward with it.","accounts_in_message":[],"_revision_number":1},{"id":"7df429f074e0a5d44c1e79a471b6f725a3e7057d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-08 22:44:47.000000000","message":"Patch Set 1:\n\nThinking on this more, I seem to remember that devstack-func didn\u0027t actually use the updated constraints file but I could be wrong","accounts_in_message":[],"_revision_number":1},{"id":"06ae1daaf12cb883f13f955bd0e5bd427e70bd7a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-12 00:45:41.000000000","message":"Patch Set 1: Code-Review+2\n\nSure.\n\nI feel a little strange about this but I can\u0027t quite put my finger on what my gut is telling me is wrong.  It\u0027s a vague felling that if we\u0027re broken then we\u0027re broken and this isn\u0027t the right place to find it we\u0027re missing the actual testing we need.\n\nHaving said that I don\u0027t really have the stack-space to work out what exactly we\u0027re missing","accounts_in_message":[],"_revision_number":1},{"id":"777c9c6fe547f68ecbb45ada3a653a131b66eed6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-18 16:41:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d7b9a1965fd12448557e72dd6f6fcb65ec2e858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 21:01:48.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"969561d28333f21474cbed75d3045d7a3dd273d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 20:42:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d1b0ed946f0153af777b7dede2a114c9ee037d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 20:42:11.000000000","message":"Patch Set 1:\n\noh, merge fail","accounts_in_message":[],"_revision_number":1},{"id":"5e9477c7d2b1a9f5656888ccf63054b487846ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 20:42:13.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"bc03975bf2810fa775223dadb930807b72224448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 20:42:24.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5c6716609991ea1e20d509b1bcab2d0656e13077","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 01:01:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e73ab6a8c24fe935cf0f75754026026eac73c44e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:02:11.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"071a68a58ac74c90578b41484e89fdc483c2a455","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 03:13:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"563d0c8cb3d14f1a684cdd3cd585a2f8bdd13570","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 03:13:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6242f1368ebb0c1af63e340700a616e87bf1368e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 04:51:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740e107a522446a88041de8c1d6eacfb : SUCCESS in 5m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5d88af612e8249aeb5e71f2d8a88895c : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24292b6176d049bd92a07f52a81c2e98 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ef3ff39fc1242fe8acbbacb23b03d45 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/606d1d95c17d4bfc81f07bb5076a0507 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/787d7b799bd447f98b8b684d7bfb7810 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50e8af79728648579cab0ec00158a033 : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13a20a968fc0493baacd92b86b5e6061 : SUCCESS in 59m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1208c28090ad456783950b4fab5e7ed8 : SUCCESS in 1h 21m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/10a55fd2d89d4f3c8d239b49100200e3 : SUCCESS in 1h 32m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6c921f43dc6c48319b04a560b8f24f54 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":2},{"id":"cf4b1c6449de5aa766d4594d9952893f2128d121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 04:51:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f877934d3aeaa609923e7c43ac2f342e0f5c09b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 06:33:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cdf0567e0734b08a07f56c15873a8e7 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0a252e792d41f9ac13fbcfe8753cd9 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d9003b126884e32ae6c764c50a049c8 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5498fea93f64452b363fe245b0d539b : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6743cfc451864e75b37486aa0e40891f : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/682cb58363094777aeec7cbeb8962224 : SUCCESS in 3m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61bc0002896544239fb5628153e949be : SUCCESS in 57m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73800f74b41e466596141b923ca3f2a4 : SUCCESS in 1h 19m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/34420ae59dec44e9be8f74ce532076b4 : SUCCESS in 1h 36m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/37dc3e3347cb4c619904ea6323de235c : SUCCESS in 1h 21m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0141b1cf0909d289521d84e9426fffc53c275291","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 06:33:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d84cfa9a7f88c860873dd0dc2d19c73f06371bf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 06:36:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18a3065ef6542fea4dc2371082baaca : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d237107f8949e0e3ba146b9c4d8db3c74f44a7ed","revisions":{"47bdbcc4ec82f5867b6be1f9cec9c813b1086584":{"kind":"REWORK","_number":1,"created":"2019-11-06 15:21:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/97/693197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/693197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/693197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/693197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/693197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/693197/1"}}},"commit":{"parents":[{"commit":"1b0fbf8f6798013d49dc94bdfae462ffc2f7b50b","subject":"Merge \"update constraint for python-novaclient to new release 16.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b0fbf8f6798013d49dc94bdfae462ffc2f7b50b"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-11-06 15:18:14.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-11-06 15:18:14.000000000","tz":0},"subject":"Add openstacksdk-functional-devstack job","message":"Add openstacksdk-functional-devstack job\n\nThe openstacksdk project\u0027s functional job is also running for devstack\npatches, so a breakage there is affecting several important projects.\n\nThe main component that seems not to be covered by existing tests is the\nfunctional testing of the heat service, so adding this jobs closes that\nhole.\n\nChange-Id: I2a5968b9a09529742af4b7057498a4fb7323080d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47bdbcc4ec82f5867b6be1f9cec9c813b1086584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47bdbcc4ec82f5867b6be1f9cec9c813b1086584"}]},"branch":"refs/heads/master"},"d237107f8949e0e3ba146b9c4d8db3c74f44a7ed":{"kind":"REWORK","_number":2,"created":"2019-12-20 03:13:21.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/97/693197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/693197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/693197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/693197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/693197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/693197/2"}}},"commit":{"parents":[{"commit":"12aab4eec9bbf3d1548c369772c12cef69845a95","subject":"Merge \"trivial: Variable rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12aab4eec9bbf3d1548c369772c12cef69845a95"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-11-06 15:18:14.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-20 03:13:13.000000000","tz":-360},"subject":"Add openstacksdk-functional-devstack job","message":"Add openstacksdk-functional-devstack job\n\nThe openstacksdk project\u0027s functional job is also running for devstack\npatches, so a breakage there is affecting several important projects.\n\nThe main component that seems not to be covered by existing tests is the\nfunctional testing of the heat service, so adding this jobs closes that\nhole.\n\nChange-Id: I2a5968b9a09529742af4b7057498a4fb7323080d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d237107f8949e0e3ba146b9c4d8db3c74f44a7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d237107f8949e0e3ba146b9c4d8db3c74f44a7ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
