)]}'
{"id":"openstack%2Frequirements~693606","triplet_id":"openstack%2Frequirements~master~I7960b30e43153704197edcbd2f5dcd60eccc8abd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I7960b30e43153704197edcbd2f5dcd60eccc8abd","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-09 06:29:46.000000000","updated":"2019-11-10 10:06:23.000000000","submitted":"2019-11-10 10:05:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693606-1573380320056-2bb4bfd8","meta_rev_id":"49b61e21d20e18f1cc2cd1b934da56aa3b771ab9","_number":693606,"virtual_id_number":693606,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-10 10:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-10 06:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-10 06:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-10 06:42:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-10 10:05:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e434c40e32236afd4118ea299872b36fc1b9d7ff","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-09 06:29:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c91f958a730fe6638a1f329895b695f107562374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 07:59:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d96f69771934855a8ca80f7a2f223a6 : SUCCESS in 6m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d021754972ee44a6a81259e57ce0ef46 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c9a95223bf4ca78ddc432af2a5617d : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9cf48be506b45018f7cf35c0cb3cf9a : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e9813917eb498e81eff9b81357f349 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e4b3d2548fe428ca85fdf79d8282741 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f928a582164f4f3d86658c1bd657aefb : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bfeee79050f4a49878fed0341770260 : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/108d2aa1bf764f17af713a8af47cb673 : FAILURE in 15m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b844ae8a6f6e40d4b4af47f9f4eedc8a : FAILURE in 13m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a6edc91ff4534484a540c66ce169c7ef : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb12d800a1c842a5aa78a172e22505a8 : FAILURE in 8m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2e4d4ee80ba54de1b8bb3e0085f2a1b5 : SUCCESS in 5m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7fcbc7d777d743c8886aa8b4bdb60f22 : FAILURE in 4m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/485c7125ab4246bd82eda42122ef8177 : SUCCESS in 9m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/609b4bf715674252b6f6d926d050f1d7 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c7708c4be3e4a0c9764561ec16b157d : SUCCESS in 8m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/09e24af8633f4c71a9a633417ae67ddd : SUCCESS in 13m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e21d33ec25d4c679323906fa2731f40 : SUCCESS in 12m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/28c25b68d88c4fc1b28caef35e66d9a5 : SUCCESS in 21m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8df984e7698b42b4bc6e9f9cb561ca35 : SUCCESS in 20m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0e1b9e817cfc4ac29e9776084fa890b1 : FAILURE in 4m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/92ede94559fd4b5fb1d4d23077a84def : FAILURE in 5m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b348f355c4a44008b8887ce3860a5cf3 : FAILURE in 6m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e58a99994bf648f6b2ec3d6504e0cf02 : FAILURE in 6m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/260cb405b9714b8994dd7373eb6fb234 : FAILURE in 4m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b0820d5bb84349d98c788bd5936d5ce5 : FAILURE in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c7d3e7247ad45169ec496b9e76e123e : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/639afffbcc3843858499cb70445cff6b : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/466bdf9108534a159f604e03dd548124 : SUCCESS in 9m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e493c93c762a4a998afa29b11e171ae0 : FAILURE in 42m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abf7d610c2614ea489f44aae845178ff : FAILURE in 1h 22m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a0665d7d30b24d24ad1b872c78c5ec65 : FAILURE in 27m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d06ea8e038e34f2b95149a4994612946 : FAILURE in 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"74ae0a61ed803b33d01b1c7f3e6fa5cdde38c3bc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-10 06:31:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5cea59f122eb6081203464089553987245ae2b7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-10 06:41:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"242edb9b582a0023a75f712618ef49f54e377bfb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-10 06:41:51.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"586762bc0d806653d24b045bebeeea92226441df","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-10 06:42:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"640c2f5f2c28f96a7a991d1cd5c0993a9f9b7e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 08:36:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9064be2b3b047aebcf0e710f59ec157 : SUCCESS in 4m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d1e311a1156f4e398492656109dc6dc0 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644b3e3923d444f28c07b83fe78e91d4 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fad93be2ec894ad59647b4d64bcf99f1 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a0836c1ff2549a7846da5fe1cef2e21 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0d02eaa7c90401f95fa164f2268fa09 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7c16969582d4e43832d7a6cd32b714d : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/451747106a114852a3be3025e3a5694d : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2b52dda8a8a24ee59d1ab0bfb257daa9 : SUCCESS in 7m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/663691cb7933499b986b2b5d88f5385d : SUCCESS in 8m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c3eb61d0f1a34640bdc4831ba6a6cc75 : SUCCESS in 13m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81c2959a764846339c1886ed2b7ef951 : SUCCESS in 17m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9dc2c2d0f9404cd9916290fb5015c9e4 : SUCCESS in 4m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c221738091d3451ea2f691f476a62752 : SUCCESS in 7m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e55404a14d1a445fb3cd0c09257cbbb1 : SUCCESS in 9m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/31c922b9a984441c9b3abbf476e18ebc : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cff54ace352e41bea1cc03a274e9f0f6 : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f557cef5e6d7485ea54b6ce4a2ed6cd0 : SUCCESS in 17m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c80a6e87a2824e69a780e2af588a3e8a : SUCCESS in 12m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7c1b03b997a744bf99752accdf4bc4aa : SUCCESS in 18m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/345a4f514c144e8da208b9ad1cf113e8 : SUCCESS in 22m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ea152a496e774960a110ebfeeff4887e : SUCCESS in 11m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c23277c433942db8f09ce3b950c474b : SUCCESS in 24m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e918504a7462426ca1339399826563fa : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c562b54b70b470592d7e62a4c00bd56 : SUCCESS in 22m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6c5d6b4bd77649c8881acf9a6a4a2e8e : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/abba8bc6848a4ac9bf6f651b7cb0c8dd : SUCCESS in 8m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49338285059c43b5ab2a3502b3619497 : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/988ef1b4457a4a16bf0d850708fd9073 : SUCCESS in 11m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2196bd29f23443559f006ea2f0021de3 : SUCCESS in 11m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cb5f6e92b4a24e10a1fa4d17ffae981c : SUCCESS in 55m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87863100a6724e77bd82d404f388debf : SUCCESS in 1h 09m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/23c00b0dd9aa4644a3cb2c52099aa3f7 : SUCCESS in 1h 49m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af0043fbdbce4a4a8c5c7e0170b6edb3 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a71606c64f2a6a1b296c4c7efe3a7be36bc96578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 08:36:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4317230fa3560ef28f1d3732c57ed9255899647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 10:05:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34608ed07d0240178561db406bfa004e : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241c20e97c904e3b834a379f2cd38a37 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f35c6784ae1474da9ed7159868df2f6 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44597d02937c4d06abe036319d7fd290 : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae263c1795314c0fa2e31f490282ff3a : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe2e308e5cc7400db055ecc98635ed91 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a1b7e3c72d144a38b866a5c1d82465f : SUCCESS in 2m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aedb5c6f71ed455a971919732799cf78 : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/402f1608ab22469db62a2393a272014a : SUCCESS in 6m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3dbb8dffa5c84aaf87b7f1ceb7b1922a : SUCCESS in 11m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc0ff7e371824069aa658dc1d24538b8 : SUCCESS in 8m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a51434b98eba406ca2477ab65c09e1d0 : SUCCESS in 5m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad09f2aa0bdb4c46b07479abe6579982 : SUCCESS in 4m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f1436f17a46145d2af03ef99e6ad03f7 : SUCCESS in 9m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/20c441b2d1cb42d5b832ea07a211bae5 : SUCCESS in 8m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/deec23a155bf426dac36699fa80b747f : SUCCESS in 7m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cf16ab9083384c33b38ff0a8b2e339b5 : SUCCESS in 46m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ee19189f93f4b299796b58f7b8220ba : SUCCESS in 13m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f077146264e14657ac5480dc327331d2 : SUCCESS in 19m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3e908ffe7474d47920cf75d8b96e916 : SUCCESS in 20m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b9d18f8774b9499c9be93a5a91139923 : SUCCESS in 13m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7af1bd23fe3842d99722d11541758886 : SUCCESS in 21m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/20c53a4cfa544bf3877455e91e37ec33 : SUCCESS in 11m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83a2cc39493b400cb99ee0c7a6581237 : SUCCESS in 16m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6a2d2eb92ce4222a45a23ce99ccf654 : SUCCESS in 3m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e221a256a6e40b49b2f66aa955f5649 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07ba5c0263474c7ab91835d7f3321e05 : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/14e5b6e31cc9418babc681d17a8836da : SUCCESS in 8m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03bd1e04c3ca4a469e79d2f326a713c2 : SUCCESS in 8m 21s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d86f0d18bfd64a9aa94b9b289b133855 : SUCCESS in 51m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c5e274961ed4e21a6e3cf558f3c59d2 : SUCCESS in 1h 10m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd377d4faf4e4133890418a820699861 : SUCCESS in 1h 24m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/201a05bf0a06496da466023e5f7b5e01 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e8514c4da1d716d2ce3c738115390c93b604815f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 10:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"49b61e21d20e18f1cc2cd1b934da56aa3b771ab9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 10:06:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece65b7527b54dc0a50500dec190a32c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d5d92e6ef8f5b02dda50f694642c0dec2acaadc","revisions":{"b318f35065ccc14df00d58327f5e0598e2655d39":{"kind":"REWORK","_number":1,"created":"2019-11-09 06:29:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/693606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/693606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/693606/1"}}},"commit":{"parents":[{"commit":"ecd5f77bb01785a10aa08b4e2a9bff741f5c6953","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd5f77bb01785a10aa08b4e2a9bff741f5c6953"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-09 06:29:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-09 06:29:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7960b30e43153704197edcbd2f5dcd60eccc8abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b318f35065ccc14df00d58327f5e0598e2655d39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b318f35065ccc14df00d58327f5e0598e2655d39"}]},"branch":"refs/heads/master"},"63e7ea669da79db86faf39cef00ce3966f422e49":{"kind":"REWORK","_number":2,"created":"2019-11-10 06:31:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/693606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/693606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/693606/2"}}},"commit":{"parents":[{"commit":"ecd5f77bb01785a10aa08b4e2a9bff741f5c6953","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd5f77bb01785a10aa08b4e2a9bff741f5c6953"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-10 06:31:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-10 06:31:42.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7960b30e43153704197edcbd2f5dcd60eccc8abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63e7ea669da79db86faf39cef00ce3966f422e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63e7ea669da79db86faf39cef00ce3966f422e49"}]},"branch":"refs/heads/master"},"9d5d92e6ef8f5b02dda50f694642c0dec2acaadc":{"kind":"REWORK","_number":3,"created":"2019-11-10 06:41:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/06/693606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/693606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/693606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/693606/3"}}},"commit":{"parents":[{"commit":"ecd5f77bb01785a10aa08b4e2a9bff741f5c6953","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd5f77bb01785a10aa08b4e2a9bff741f5c6953"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-10 06:31:42.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-10 06:41:37.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7960b30e43153704197edcbd2f5dcd60eccc8abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d5d92e6ef8f5b02dda50f694642c0dec2acaadc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d5d92e6ef8f5b02dda50f694642c0dec2acaadc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
