)]}'
{"id":"openstack%2Frequirements~693615","triplet_id":"openstack%2Frequirements~master~I0b77859eea4d2841ba148623b5bca7d1d165c89b","project":"openstack/requirements","branch":"master","topic":"py3venv","hashtags":[],"change_id":"I0b77859eea4d2841ba148623b5bca7d1d165c89b","subject":"Handle relative paths for local check-requirements","status":"MERGED","created":"2019-11-09 20:57:56.000000000","updated":"2019-11-13 07:40:14.000000000","submitted":"2019-11-13 07:38:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693615-1573630718610-752aeac9","meta_rev_id":"2b1a178533c73e4fa670583ef312b77fa6559035","_number":693615,"virtual_id_number":693615,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-11-12 11:55:04.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 07:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-10 17:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-12 00:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-12 12:27:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-12 00:52:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-10 17:52:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-12 00:52:33.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-11-12 12:27:15.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-13 07:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b0b93cda498df2f1352bde93e6e51d70016b9cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-09 20:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"069d4e00bdcf8635ed18f64353500438201bbd71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 22:38:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef218143760844d58bac3cba86ca2923 : SUCCESS in 6m 13s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/df2c8de0b4034650911c2c9d941d7018 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc957d7e010a481fa4482f7a7bd93f3b : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9a3cbbfe86e4ac89afa8373f86a4169 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97dad20660264064851d16f72308863f : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3360930f67ce434b92c481ac41ffa39d : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82b2b67b001f477589f68128d34d6d7f : SUCCESS in 4m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/93d7ec007bab4fa2b7030bdaa09bac29 : SUCCESS in 1h 04m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e77df0dae5a4bc6a04f5b83c1f3804b : SUCCESS in 1h 16m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/973ba66134ab4fe184072dacedde3c48 : SUCCESS in 1h 35m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d4994c22a71a466d92ddc0342f236609 : FAILURE in 43m 01s","accounts_in_message":[],"_revision_number":1},{"id":"122fd316720f71c42899647acfde3571c8b4922f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-10 17:52:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24097b4f6e2a3261985008cb94e1e8e52f9e47dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-10 17:52:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4530d9914f845b7c4cc99edfb40c0bc54ca12d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 19:25:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dede863bbc0e4bbcad778e90777872f2 : SUCCESS in 3m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/249c3ca8feda4f8daca5fa12cab52971 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac335c16f8e44998e805ef90af57dd0 : SUCCESS in 2m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74fa4ee43eb34ec1a7eb723f1496953f : SUCCESS in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/771a9518a8cf466b827afb09dcdf5dd7 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7e1586a49cc4541ab1c4de2d031bfab : SUCCESS in 2m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c9c69e582df41388d6ffd0d228640c5 : SUCCESS in 2m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f31a070a6f7646888f7aab357baf375d : SUCCESS in 55m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee19b1644cee4e8d83e107a739032a3a : SUCCESS in 1h 05m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/02d2a2c957d74a189a903021c413234b : SUCCESS in 1h 27m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9f43d8fcb4f94268b4a963b47e618d1c : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a601a796ccb0bd96cff4f971934704a8b7810a68","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-12 00:52:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"7be9f1811d4eafd865211feb4234dd36b0482d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 00:53:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a4ac151535d24f4594b30dc30c3afff65a7966d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:37:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045ac5ec123c4f1d97f96649eb550dbd : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c073d6c9e8844b598db88714ea3ee164 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ea44616e0a34b51938c6769400f41c0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f24f9da8853d436f8c18be1e943b2645 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66fe9a1ed4354ff1a04b389c6a7c54d1 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/016c6e5f8e5d497e9416e03d5c976acd : SUCCESS in 3m 31s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ccbcf6ede3f34c2aae0a7637b23bcf15 : FAILURE in 29m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8e7b22c150c4897b099eda4f44d499e : SUCCESS in 1h 09m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/73c3763a1f604b59bc21f9b8ac6f7255 : SUCCESS in 1h 47m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2708d06a12b34b78a425e4f916d27050 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"785442cdcf6d9b437af3ab9c3cd730af4fb615c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 11:55:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1cb3a953b35f50972f3c78130f0c63b6d63a145","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-11-12 12:27:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"539f9a25e4df3e14c0871a5e8acefe27f72c9268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:09:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d26416e2e54a91ab6838bcbd2a1e81 : SUCCESS in 4m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/96a6a06669d84f6a96e90d828bfd2274 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3580d3863cf46f7b5b949633232a93c : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47fb75e251334de4b6d3bfea8d1fb7da : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e3bb6e4c83f455aa607a3de2b7bd725 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d89aebe0194f45f9bbd0dde0a9837a72 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1128f00ca2f64852b1d2f0c6ab489f4e : SUCCESS in 3m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/473767cffb7a481580739b7d5587a49d : SUCCESS in 52m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6923dd3998647308c9756b9207c0a89 : SUCCESS in 1h 07m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e25bc9f114414f7683de9e8d5cff4863 : FAILURE in 1h 12m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30c9c888665c4ce0bf2f10be18cfc025 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d97ab232311bd008e147d39a3edc955c1b8cd9cc","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-13 04:24:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ff26e50bd4ee4f5530927cdea702b1ef399a9cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 05:57:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba534d79732416d85c1dc9f153490fa : SUCCESS in 6m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/91d3fa1c609f4277b78398739acf9fa7 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9dc21ae030c420cbe9cd6be85f42180 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e343d79615be41e285491839dd5e1bea : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2398bf05859d4ab994298b103c4ff953 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1a97515bace444e86c6b9b9d9cb568f : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df30ba6e319c45069e5b8104cdccf69b : SUCCESS in 5m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35a2029a9cc748238ba3a1916042a1c4 : SUCCESS in 1h 03m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f7ec5a104d24bd6a0e42a6293ee6b83 : SUCCESS in 1h 27m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7341c89903c14ab68c1c27829dc4d1ae : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9db0d241899c2dbf48f839643f563c51fcaff5d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 05:57:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41f9d5b3dec31b864ed5b192428b6ab82942fd92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 07:38:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a733ba4dc871e8655447c28bd4a231fb225d647c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 07:38:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1fdc83a86f249af91ac1dd1419ba743 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a1441cb504428a9c2f5fa98cf9fb8a : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67a7005e991141bcacd376f75a1468d6 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80fdd6543e824937b394fd3a913793f9 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/353da87b62694da9be0522e9ca1ae503 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98a3d5e577ad487387c9a97ad7045148 : SUCCESS in 3m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4196dfc969b4eb4be6e395f361b04c0 : SUCCESS in 1h 08m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a45f288a861458a819dbefd872a3ba1 : SUCCESS in 1h 39m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/286b0765d9674a248819c00c2ec36f87 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3ea655a5f3c9397500fac894ebb7d1eafbe3acdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 07:40:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b83b7ca09f4eb39a15940cce8381b4 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca652db8e871871124c3fa7fab666f2835907bfa","revisions":{"ca652db8e871871124c3fa7fab666f2835907bfa":{"kind":"REWORK","_number":1,"created":"2019-11-09 20:57:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/15/693615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/693615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/693615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/693615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/693615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/693615/1"}}},"commit":{"parents":[{"commit":"ecd5f77bb01785a10aa08b4e2a9bff741f5c6953","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd5f77bb01785a10aa08b4e2a9bff741f5c6953"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-09 20:45:50.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-09 20:57:35.000000000","tz":-360},"subject":"Handle relative paths for local check-requirements","message":"Handle relative paths for local check-requirements\n\nExisting handling was looking for the local path by moving up the\ndirectory chain. With a relative path this would result in an empty\nstring for the requirements directory, then fail to find fail to find\nfiles in the system root.\n\nChange-Id: I0b77859eea4d2841ba148623b5bca7d1d165c89b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca652db8e871871124c3fa7fab666f2835907bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca652db8e871871124c3fa7fab666f2835907bfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
