)]}'
{"id":"openstack%2Frequirements~693653","triplet_id":"openstack%2Frequirements~master~Iafa7fc5be697f3786094c69343e26abb8671cc18","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Iafa7fc5be697f3786094c69343e26abb8671cc18","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-11 06:23:46.000000000","updated":"2019-11-18 01:24:46.000000000","submitted":"2019-11-18 01:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693653-1574040220314-3ed51160","meta_rev_id":"226c5bdcabfae9fa3f56b6150e6600b6600a170a","_number":693653,"virtual_id_number":693653,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 01:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-17 23:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-17 23:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-17 23:53:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-18 01:23:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fba0f07bbd81a48bf7232351d667055445ad7bb7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-11 06:23:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bee7fcc75ce8085595a183e0c0a9749be813c1fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 08:11:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3bdf672eecc4498b5c06d1c041146b7 : SUCCESS in 5m 35s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fcf1324754e044868df074979de5fead : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb7217a02cf4d69b2349828633dcda1 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f309a17fa65045c789443b82391e8a7f : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d58d4843e144ec58230fd453ab73c03 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b59a78107a4c40b7baf9f80a19cb65b3 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9acb68d529d74eb89d85f6eff7c259d2 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9765e7801664a06bccfbd16b76f8ebe : SUCCESS in 5m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9c27b9a068bb4b79a4ec8066859d59c4 : FAILURE in 10m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be4e11e56de7489ba17552428b25211c : FAILURE in 17m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bd04431ad75c41a08975a5be9944f709 : SUCCESS in 12m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a2f7bd65b50420187fea2370337079a : FAILURE in 13m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9651c3b500f647f194b8d0b9b62e4813 : SUCCESS in 8m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3d02ff135d824e2582fddc0800b2270b : FAILURE in 6m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f5030ce1da1843dfb3b2f752c525d718 : SUCCESS in 12m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3a464a107a0142a1980815ecc16fa2c7 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/120572ea43e049f88d8d2ca2328a96f8 : SUCCESS in 8m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/96552498e55048d1b604365e460997a4 : SUCCESS in 20m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3756059ea52432ea31c50c4babdae7b : SUCCESS in 12m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/958888c2ef904e7c9f41634dbaca6265 : SUCCESS in 24m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7437445d718148a48eaf18ebda28db20 : SUCCESS in 24m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/70e881d54b51421e9efe87aac636f561 : SUCCESS in 19m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d8ff4603f92145b780d04f23ea0e3fbd : FAILURE in 14m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9ee99fe14fbd4da59684d5017556ce6f : FAILURE in 11m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b1843aee5fc4aa698ab54282ac2aa49 : SUCCESS in 21m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/351408f9bbb34bfc875bdc6a0406554f : FAILURE in 5m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/06d5521750894cf9819de765c6c22f1e : FAILURE in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c08967fbe190437588abcfa66d25bba3 : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9cd951a40af54eb7ae4d5911708f86b5 : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74a276da3bac4581be01847bc5a8178b : POST_FAILURE in 11m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ade290d0656347598b3127b8516ef911 : FAILURE in 42m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90c3baeaea65403992cfed0b3fc4a9b1 : FAILURE in 1h 27m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3a302455066940c88a44945f6edce6c0 : FAILURE in 42m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/914ebb946c654106befa2e2162b7bfcc : FAILURE in 45m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bc749d6acc1d87d04b6fed1f4ff6aa85c4b023c5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-13 06:19:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0e08892ef3aaba68e0125bbf408fca7e242b0e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 08:01:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d0b248c85d40df95347c4af2c26fbc : SUCCESS in 6m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/381038713cdc450dbebbe9fdb9b45824 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5bbc5a23dc243afb511d960e0cbe97f : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e8ca99e6ae14401affadeb18c577e6e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ab857713ae41cca7829a47495c1cac : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a3bc72b74144c91b6a5ceaf70792f71 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e0752ed74de46e884f82342e39f7c28 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf8a4f51e8264d16af53a1ee61772f4e : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e09c3bf8a33b45c6b0a1a83be75855d7 : FAILURE in 8m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0144cc823694a7a9be67849433c17a6 : FAILURE in 7m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/109fb314277346d0b6d4c58c4d7dea3c : SUCCESS in 11m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ad524de579b46739eb02c473babfcb1 : TIMED_OUT in 41m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/935e66e5130c4c1e8439f45212898d9c : SUCCESS in 6m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9e8aedfcc1864d53bfaddac673f6f33b : FAILURE in 7m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/88ca69e46bb54421a104541773a5799d : SUCCESS in 10m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9b25f8dc87d94e5589508da95fe38131 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36ce2d77e67c4380a04efa31d442a09d : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/27fc431f7cf645a3b0418a43e88c03d2 : SUCCESS in 18m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c25b6818808c48f297df8318acc741c9 : SUCCESS in 21m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b6167dfcdf4241bfb25c6b984b5c6584 : SUCCESS in 30m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c6ef1b872d114d62be26fa85ccb51268 : SUCCESS in 33m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/500d1dcda0a4468b92a43f9a30ba4a9c : SUCCESS in 20m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/578a976c4c5d4b1f9dae3af68308fe98 : FAILURE in 19m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cded8c273f22474586a1e315c7c16e88 : FAILURE in 18m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b9b7c69ce774fea87b634060d498602 : SUCCESS in 17m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5288bd23911e4a00bd7b3a89b850061a : FAILURE in 6m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9506aac68d5748cb916b68290fedde1c : FAILURE in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e415aaae955444a8a2ebadfb47cb223 : SUCCESS in 7m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b86bdd52fccc421da5cdcd2e53b26b22 : SUCCESS in 9m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9bb8945e34cb4d3bb3ebeab14a094016 : SUCCESS in 9m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/891e70dc52304c3caf8f59586313be0c : FAILURE in 1h 09m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f1e60d096f343f38bd8896d111ecd06 : FAILURE in 30m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/823ad5ebf0fb41cdbecb70fa56f713f5 : FAILURE in 1h 10m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3634444884187f886a04c11fe29ef7041c295c50","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-14 06:23:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b0d6b246b4496727dbba06224954cd212054b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 08:06:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d40784b48b34c148ccf75e42e002b6a : SUCCESS in 6m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/542c516a16d7420f92ed0ddf42743866 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/021a0da9a635457083589ea0ba267294 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f200ac6299a14ccca3da3d9beffc9de8 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e698c2489a4a0ca602dbce4a53f165 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97b2e1f1afad45e7b2508f3cf0393177 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3261f310d0448f6bdb8c78f857bcede : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef04bb11b900469eaddc71e8c1bba02a : SUCCESS in 5m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dc657755172a424e8ee23f3c23971ccb : FAILURE in 9m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e0c95fce40f04fcd8bbdff8f5513f531 : FAILURE in 8m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a050c2a908884cb6baa563b10210d56e : SUCCESS in 11m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1f50be1d771845b3a571a49056314ff3 : FAILURE in 10m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/13a2a15975774b5a8d00aa79f4f9374e : SUCCESS in 6m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4e02269cdbe4df5bf7434b6d76367ea : FAILURE in 6m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f7238e3cc6db42358c860b6c8783a94a : SUCCESS in 12m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/569a8945796242a0b93350517ee2faa1 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/733c09fa513b486d9299f77c6a26a6dd : SUCCESS in 8m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/32cc2eb708014bb0acd7f83a3aa39f59 : SUCCESS in 21m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b865a6510e344e2aa9b3a060b63d1ac4 : SUCCESS in 23m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/62ba3469a8b9408e9d337469618745ff : SUCCESS in 20m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ee00d49574d494db9b761ad58309b77 : SUCCESS in 20m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9763ee3ece7d47c1be84e168abffebb6 : SUCCESS in 17m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9131809c63d48fdabc129addf083f39 : FAILURE in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/61796296d48241719fc518032aa5e27f : FAILURE in 16m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9674016e48e94e2ab342abc1c355bbe3 : SUCCESS in 17m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a4e708325c4e4156b8b4dc00c8f99e3d : FAILURE in 6m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4d49824c732d4d37b92b3979279243a8 : FAILURE in 6m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75cf6b62879d43acbff88f297779d959 : SUCCESS in 7m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7c70d842786044079035ee4c5767cfd9 : SUCCESS in 12m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be5ead4ec9d440c9a3ecf7733169a5e0 : SUCCESS in 12m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71bfcb1a3b484cb39405f53fd8b9c53a : FAILURE in 1h 10m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/65222f6a3f764572a097e5ea459a5bef : FAILURE in 31m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb3b0352e7bb48f786823d2bfa368344 : FAILURE in 49m 27s","accounts_in_message":[],"_revision_number":3},{"id":"e40df196198c45cb998318ff65a2becacde35938","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-15 06:10:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b7fe70f673c3dd3bb1bf3fb723d3aa66b38cf7db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:09:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08ecc35af7b4e8e9bd5a8a846a39dc0 : SUCCESS in 5m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d7c98b3eee324f0ea7ed6ec8b9dd428e : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc8f4558acf45fe9ab96474313cc9eb : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa94ec07eddb453eba2375d8799b0f68 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87db692dc68a42b69fb267c08afaafeb : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1917846d00ab4d588f16b2f50ed72008 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98279e243b3743feae801c72cccd4367 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e0f97e20f68481eb5a47b9316f6628f : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/990b63e7febb48e4981b2ce2595b20ff : FAILURE in 8m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/031a7a95a79548bb91a7af88960d965f : FAILURE in 10m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fa14016367824a8aa6447bd438002342 : SUCCESS in 15m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/260d120c4deb471aa439607f3e9668c2 : FAILURE in 15m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dae9db7522cf46beb9974c26ab18253e : SUCCESS in 5m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad71a5492b804da48c71d99f9e2007af : FAILURE in 6m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f1187414f53f41f48283d39dbbbea8ba : SUCCESS in 12m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db3b1223ddfa4218993aa0fb88e8e1f8 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b4706fb9bc843a3bee16c57a9ffb760 : SUCCESS in 12m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fb0e666b20e34560a81b7ac72f379e73 : SUCCESS in 20m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db73759bf78745ebae2445c01c8c0ff8 : SUCCESS in 17m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/884407ee55fb499392fb3b9d8452eeb2 : SUCCESS in 26m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f246d85edf0c401cbd49e07d232dd2d5 : SUCCESS in 27m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/89a18547fb1a4ad0abba3edaac242dd3 : SUCCESS in 19m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ca89d0a37e324e75b2112463bfa27a9e : FAILURE in 15m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2379e5167e774b7e9cf24395bc37989b : FAILURE in 21m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f468d99a42724ae09d13b6b4f05caa77 : SUCCESS in 24m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/90f785196cb849d9a7a6a437e0a2335d : FAILURE in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1022438a817d40d4b4f0cd7a10afd9d7 : FAILURE in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a9b8a0a3eef488783bf053c477ae50e : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d2d06afb9f940b582413564ebb00295 : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96da113a4e9843dd8caef789387d8ab1 : SUCCESS in 10m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a3da6d7aa624fc6ba7d6f9040e501cd : FAILURE in 1h 23m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8cb17fc1b3f44a55a58ad706f1bfb728 : FAILURE in 30m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c940f35437ee4da0af2032e22ec64c7a : FAILURE in 1h 08m 25s","accounts_in_message":[],"_revision_number":4},{"id":"63d0f02cc8f46974cc204ac996f6f44ab93f04fe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-16 06:10:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3ef917c545bedb4ce49013e5a6e4849343c24b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 08:16:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abeb78f5fc444efa1a17bb3c446ef87 : SUCCESS in 4m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e3262a5902584ec0b79c8ecab064068a : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7788b2cf5d8c47a394d33db5f1be34b6 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6cc257914224981bd4aecbfe75b9da2 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2afc82441d404dfc853550fc0370856a : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/980c843578d249dd8ae4bd214d70f302 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb0cff0809394c6f98ef468b8d13062d : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/934d1ca447d548a7af4c3498b0625352 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/76141eae4d824ffb8a8c309d2eb19c60 : FAILURE in 27m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bbd0adfd33a04d8393d522595ed700a5 : FAILURE in 28m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f00d5fcb96d447e992fc07315e75a230 : SUCCESS in 19m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f437d07f39e543f39c5d8ae8b6f1b102 : FAILURE in 17m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/788a8c89596349b18b9ac7f592e93f9b : SUCCESS in 11m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b8c666ab89594424a23d854d2d63f069 : FAILURE in 10m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c54499f72e3d4bdd85fc913f4917b2e9 : SUCCESS in 9m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89c3c537a67e4f059bd870df362e3f14 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/325128d883eb424783c2207171d16df8 : SUCCESS in 11m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f8990b8cf0e24da6a97837444911a79f : SUCCESS in 16m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/60fa59ca1763489fa59f5e1070ffaf99 : SUCCESS in 13m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/92a9bfb9faae44ae8897874e8e7026a2 : SUCCESS in 29m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d4176316bdc427bbde01540110ce516 : SUCCESS in 29m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/375d3ccbaf5f4c8fb21dfbe892a6378e : SUCCESS in 13m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/48d023b288cf4ccf97ddafce96f4182c : FAILURE in 15m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/04b2942326a24ac9b815b50ea5e31322 : FAILURE in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e9c458e257ae45f79a479035af76f115 : SUCCESS in 21m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8a0976637169402bbd4c6a80038a1b27 : FAILURE in 5m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3b1e49ea8a5e45ceaa4c250397006b51 : FAILURE in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f89b0c00e4cc436b91482b26c6623d7e : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8ce4e39112df4eab92090ac207ae09ca : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b3907deeeda42a183a4839b5133c1bf : SUCCESS in 12m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdc77428bbe84ba3b30ef01ccf427ba0 : FAILURE in 1h 12m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a07a705610949a0aa8fe60eb0525943 : FAILURE in 32m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7f86d57568b5463088ac1a0cb4cef9ba : FAILURE in 50m 28s","accounts_in_message":[],"_revision_number":5},{"id":"da19e4a2f7bb2a91384c4623dd46a6f4b5c4b180","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-17 06:25:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"73a5b682809b8cdc30054440958d8a220115a48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 07:52:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ff8370ef324ababda7b0fa74e8ddb6 : SUCCESS in 5m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7ff1a7dd699644068a994736e3a1d25c : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f013bb27ee4657b4dc14cfb5d84c50 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e71dbdfd77d429a84cbe1f9e3620910 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93504486dc254c1780254721352809ef : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef38e51bd08f489e9c96f05b61c69802 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a6908b214b441cda2d5ae40dc4e0826 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35ad0561a43442c8a9f90258bdea700d : SUCCESS in 4m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/576fe75a531e4b3092720b7fba6443b7 : FAILURE in 8m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebd9f99d942b48b7aab214147eece8e1 : FAILURE in 6m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ba23ab037a8543c6ba2ad20557aa10a2 : SUCCESS in 11m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cd7c112ed9b847a7a4ce014b8b2b3bbb : FAILURE in 9m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/75177f52007b4f8d9032b24dcf0b1f45 : SUCCESS in 5m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/297e988fa6a84623b53203a8c792cdc5 : FAILURE in 4m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6b03c9d5a48e4284a3d3e0178477d49b : SUCCESS in 10m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c24f558674f34518aeb6dd5fc322846f : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f25dd6a14d1c44ae98abd89403542161 : SUCCESS in 8m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d7c378571f3742d6b871880198e3d68b : SUCCESS in 20m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/91498e1e1aff4fbcaf7e85b1ca8887c6 : SUCCESS in 15m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f51963b140904321812cd8a383aae035 : SUCCESS in 34m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4e237483ac3c492bb252fb3c148b7ed8 : SUCCESS in 24m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/52e3efab05424b038b8636702aece38e : SUCCESS in 13m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7153100380d4a7fb13d3c9a2664a8bd : FAILURE in 12m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/464fe11453044fdcbe34f082cc56e10f : FAILURE in 13m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3664e8954ddd461093461d6633e956d7 : SUCCESS in 20m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/75cebbec3142435f9cd1d7974c1e5353 : FAILURE in 4m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/41fcdaa49bc847ccb41acf21e6a1f2d8 : FAILURE in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/008c03b180f4412aafa436348211f2c9 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aff2c4fc96a14ef2aeae2405b56830aa : SUCCESS in 9m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/850e822b1e9b4086a3c003ed2683b431 : SUCCESS in 10m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6a26bdd8be8406298d8a89be8bd6540 : FAILURE in 1h 11m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/081c5dc238e64297a54d5ee9b3cff00f : FAILURE in 29m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cdbd54f432e6494a83f8bcce40035ba9 : FAILURE in 53m 28s","accounts_in_message":[],"_revision_number":6},{"id":"3650127e5ccdab7ddaa40b895e4b73158e5a6f10","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-17 18:56:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"59f0d9745cee4e84851f45322641b7ef62f2b1ed","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-17 18:56:20.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":7},{"id":"003b6c6be70698d33a14d44622851cd3ee04d1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 20:06:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6c4ce1130b4072a9ca786252c03823 : SUCCESS in 3m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3213f9a12ed64b1488ff3dd5fb7f4611 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab7fd2445e04b1da6bf74033e7e34d0 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3ba4351f73a43a4be902defdae51ab6 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/136054fdc88c4a388262a7dab05d266d : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0497b088334c4437b382587d24ec1d2f : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a82f1adba63e4030bd458e1fc6cc17a1 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cf706df9a394e658049f10e1060e77d : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9af9fa00dcd44cc2bfd4b93c02b347a7 : FAILURE in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a393e9bcd3484a74aecf2f1c9bcc8586 : FAILURE in 6m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/33406194f86547a2bd1f42a26945fa1e : SUCCESS in 8m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ffe342b4f0074c2f971acacbeafa87ec : SUCCESS in 8m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5f08575307ec4c8681e3186892ca2704 : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da62ae424d7b45fe85559e5ea7ddc487 : SUCCESS in 4m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/43e947371c9b451fb1d25541ea70d15b : SUCCESS in 9m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76d7e7743eb24cbba50844c5b34524e1 : SUCCESS in 8m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0b9eba2d77a4d31903325ffbb917e3b : SUCCESS in 6m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1d20d560d6294597b8604c32977b50a1 : SUCCESS in 19m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bb68b802d8b34f38a101f6da9d5c8056 : SUCCESS in 12m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3fba95b1e99d407caa8fc8dbb07883e8 : SUCCESS in 20m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8bd98b8cdff547a594774140d691d0a0 : SUCCESS in 21m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ee263caae9634e7bac8542e629fe039e : SUCCESS in 12m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df98fc48caf34b7aa0b6f6a5ee8e5a22 : SUCCESS in 13m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d6590f64358b4fd0b17582855ed55f1a : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3edc121c0a5144aa85b3af003aaf0deb : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d33f40dccf0446ad9a9723f120ed2220 : SUCCESS in 3m 44s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fccfff77a98a46bebf95bce32e2035fb : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aee6a176eaeb462db81b5d7642cfb6cc : SUCCESS in 5m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/82f5477912cc4dc0a1f798f34a235b8b : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea9c4a5b7e854c19902d88661fedcbc7 : SUCCESS in 8m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1109d885cad243ffa6c08e9edb4b7de4 : SUCCESS in 1h 05m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27539b9cf6e7493eb9ece06821060965 : FAILURE in 26m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/834db7bdbb6941419fc0527c9da7f661 : FAILURE in 46m 36s","accounts_in_message":[],"_revision_number":7},{"id":"de20e0c4f9eb4392bf28b3e6fc4ce30a2601c76d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-17 20:12:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e578269c00e62a4b7c50f63e77a886c8cb51555c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 21:33:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b08cb6e75aa48e6ba13c7a62077c124 : SUCCESS in 2m 48s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e7a4d432ed15405eb67044e37248ec63 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917d621becea43f6820c9278e708d7f6 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d038b1f90c74c0cae36b60470fbfdf1 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac8f6e82cbd944d8bbe45500f50f8eae : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ddd4df44adf49f09ec522c977c97bb6 : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3059ea49a54d4431833074d5416d0447 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3203312046834759ab552f7e72b7d727 : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bd5e4f3cc5ce4c55858a310dc24a850a : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/62dc4dec47c34f8fa76cdf46bec01eff : SUCCESS in 5m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2052bbde1c684ced8b01c8e7c0aaea02 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b2ac2900a51c411fb7bd0b37208d85c5 : SUCCESS in 9m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dd771d27ca024cb5bc0c4b3dd997a1bb : SUCCESS in 4m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bfc9f4db736e47f5b39a58e50e61749d : SUCCESS in 4m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/662513b6b9c14d628c97723c5d1b9d04 : SUCCESS in 9m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/acf57cb5b6814d7d9b3083fe73ca3046 : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4c4dd2cf2714a99ab498580d49584a1 : SUCCESS in 6m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eed841a280f54f22a1d6fdff7eecc1f4 : SUCCESS in 41m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4397e8543584a89ae9d0d09824ed3d1 : SUCCESS in 11m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/21d984eadd1645bd934a713479fcf31f : SUCCESS in 22m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7d7eaae9afb2418b8dc36f316f91a8cd : SUCCESS in 24m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a260fdd024f14ed091f1ed2bd6c228e6 : SUCCESS in 17m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b0897d6fdc49450fa1e11cd5ae455459 : SUCCESS in 12m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/83d5286f31ba48cf96651b72e069e15c : SUCCESS in 16m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04667793c9b6488ab413529b61103300 : SUCCESS in 21m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/399fe6b6bf2a4e25a3257c35b31c2a9c : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/383bb68a6b9c4acd9186fcc5bfb2ccfd : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcd7dc2dcea34af1b99d8ee8be0429a9 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/16de29b903f745e189dd5163054303ab : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/80b74d90ee7c440288035e07e0700b08 : SUCCESS in 7m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29f928a4519d41a2a0ac1da3a17b4760 : SUCCESS in 1h 15m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/017eea62b18a46dca997e4ba6dab6a45 : FAILURE in 33m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6362d74de4b5422fafbca94f28ca4d07 : FAILURE in 57m 01s","accounts_in_message":[],"_revision_number":8},{"id":"2a867c3377dc064ed2915cd74a3654773c230a21","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-17 22:10:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b4d2a6b10d72365888d31a321a57001bccdb20be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 23:50:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b9b7e7c8b84fd0b9e75fb922791bc4 : SUCCESS in 2m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0f8b91ccad664d3cbcc1b58f6c96408d : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfc9fbbaf5542c7b71ee89d41c2c5bf : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/395b7aaf23d34b93bb53f62ef12eb31b : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/add3aab495114332af37b476598cd0dc : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1961b6e4dea7447b91b90c5eff31ff9b : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/053bfd0df86747eea9dee54f5eb30512 : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/185d3e4c51f748268a05d92ef49bef8c : SUCCESS in 2m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b4a5d8a8582a4489b00e298299b0c44e : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b088d4d538a946079da6bd6efd34dca9 : SUCCESS in 6m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f6d488ca59a243cba018d01d63914c6d : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18cbef62ad7146148d04709990674bfd : SUCCESS in 7m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/65950c1976cb4601b754b7df4767329d : SUCCESS in 3m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7792c693b45e4181a30fa31f2d29d451 : SUCCESS in 3m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/17fc4e97368c48b8ba96b68881e8f87e : SUCCESS in 12m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a19952719a134af69cbe7b119fb6fbcc : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52bd7b5d2d474958a19f970188365216 : SUCCESS in 5m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6f1d5688646e403ca77c12ad085cb5e7 : SUCCESS in 13m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1dbcdba978464237b0a093f5ea00530a : SUCCESS in 13m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8ba6bba261c6436d8c1cc036b66ef01c : SUCCESS in 22m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/021fdd306a724f58a8ba34a735b502b8 : SUCCESS in 21m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/45a2b309c8ed4a2f8e386789d19f6281 : SUCCESS in 13m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/04a4083cd4814ec6a75a689cdfdbe13a : SUCCESS in 11m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c987e18e9ba64bb5bf61ab16d07c35f1 : SUCCESS in 13m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a0215fc0fb04c4d8a4c35bbb0444987 : SUCCESS in 25m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d0495e3ed5144283ba47f62a9a8640d2 : SUCCESS in 3m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/be256f409cc44b42b48d9b6a99efd8a0 : SUCCESS in 3m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cb6954ab4b14252b1ee432cf2349de0 : SUCCESS in 4m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f9b4017fd4384c3bac22ad34117abc21 : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/95014e4321c84d3999821f55ea4cee31 : SUCCESS in 9m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/725505d1bdb0440bb083f7cf19a0563c : SUCCESS in 1h 06m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/550887de077d4712aff6305771dfc6f0 : SUCCESS in 1h 31m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/546d61c35dc6463b9f2fa94025ba6e46 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":9},{"id":"14953d56e61de690b34453a5195c86bbf3bb755d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-17 23:53:47.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"df5df76a3209276c82b70dc532311277522d30e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 23:53:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7b7e6ee2d980e0ddb2ec45e13e7882e530bcf460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 01:23:39.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd343653e654a779acd5031f05876a9 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c3d4ca30e6d42e782a5bc932ba4371b : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69bd185d805a46088b081b50d962ef48 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/337cf38e924f4900ab3989eefd9a1b53 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8c1fb71b4c84b19b3474f7c10d42f72 : SUCCESS in 9m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/392eb2e94116455b9ffa2ba866aa5348 : SUCCESS in 8m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2779733c36cd4c1ab7ff3d22b621eae5 : SUCCESS in 8m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f91d5ae3d3db4078bc66f71c7dfcbbbe : SUCCESS in 9m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ed7aef106034d8ca384c4f0b2b5d572 : SUCCESS in 9m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9fc1cb549df94473912f4c6452d0a85b : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/08414f5c6112415ba9b1fe4ffb99d401 : SUCCESS in 9m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9ab511c556234e89b8648d8732dfa6d0 : SUCCESS in 3m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/22dda56521aa474eab76bdc451a158db : SUCCESS in 8m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c15161dc63ca4e89a8c8d82900536387 : SUCCESS in 11m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/353b9876b39242b3ae2fa9a81cc73593 : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/494a8c32350847b4bb991b12c4a1e11e : SUCCESS in 7m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/46cc3b9a41e445d4a5d2c11d8b19793e : SUCCESS in 17m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f4cf38582dc45239f01f992bd5ba012 : SUCCESS in 19m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e27c4f8c08ba40a79722d8c89bd554bb : SUCCESS in 21m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/546b804457c14cbd933872aff31c1394 : SUCCESS in 18m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dda0138a2a074b739b914e5ee84ecd26 : SUCCESS in 14m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5cc4580593ac4795bba4352d697b242c : SUCCESS in 11m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bfa41dfd7fbb4a6d94dae44b12958f0d : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d20c0efa0b74a4289d6f5b072448e0f : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/abf7abc6245a4f0a970cb826efc9083d : SUCCESS in 4m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c018c2df3b4b4fd69d4cf66642450ae0 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f38c15d8acd64cfeb1f035869da3aa68 : SUCCESS in 6m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b0d940d6f99b4c509f9ffd2ffef41982 : SUCCESS in 13m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1851d40c14584414842e74d5c2b35fef : SUCCESS in 9m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf3fea7624854a719b32063c43593997 : SUCCESS in 1h 03m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f849c26a06354879a656fa0f8c2a9208 : SUCCESS in 1h 29m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0576b36633424e80af4a160b0a5e2dfe : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":9},{"id":"592e60dc277531fc871f0993b24b3655ca30dc83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 01:23:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"226c5bdcabfae9fa3f56b6150e6600b6600a170a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 01:24:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da96d4948754e1babca82d0301ea510 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fc12b9da7985936b03686de43985389a22c65cd7","revisions":{"bc82d85824547aa3ec85354cbe148250818c3fc3":{"kind":"REWORK","_number":1,"created":"2019-11-11 06:23:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/693653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/1"}}},"commit":{"parents":[{"commit":"9d5d92e6ef8f5b02dda50f694642c0dec2acaadc","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d5d92e6ef8f5b02dda50f694642c0dec2acaadc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-11 06:23:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-11 06:23:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc82d85824547aa3ec85354cbe148250818c3fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc82d85824547aa3ec85354cbe148250818c3fc3"}]},"branch":"refs/heads/master"},"9a15cdeb4e6ddcc797d4868b0fc9cc90a17c9fc7":{"kind":"REWORK","_number":2,"created":"2019-11-13 06:19:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/693653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/2"}}},"commit":{"parents":[{"commit":"a9e58f7109545b6441a3c93054737080def22ccb","subject":"Merge \"Remove build_docs settings from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9e58f7109545b6441a3c93054737080def22ccb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-13 06:19:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-13 06:19:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a15cdeb4e6ddcc797d4868b0fc9cc90a17c9fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a15cdeb4e6ddcc797d4868b0fc9cc90a17c9fc7"}]},"branch":"refs/heads/master"},"87fdf85c80eb126bf8105d0a692ee02ee5314185":{"kind":"REWORK","_number":3,"created":"2019-11-14 06:23:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/693653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/3"}}},"commit":{"parents":[{"commit":"4f16a404a0f12768910d1c503191e3f2ff07b488","subject":"Merge \"Update check-requirements for py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f16a404a0f12768910d1c503191e3f2ff07b488"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-14 06:23:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-14 06:23:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87fdf85c80eb126bf8105d0a692ee02ee5314185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87fdf85c80eb126bf8105d0a692ee02ee5314185"}]},"branch":"refs/heads/master"},"06d16f14131d660d05db6c86bff32a793c3b7bf8":{"kind":"REWORK","_number":4,"created":"2019-11-15 06:10:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/693653/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/4"}}},"commit":{"parents":[{"commit":"1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a","subject":"Merge \"update constraint for oslo.utils to new release 3.42.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-15 06:10:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-15 06:10:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06d16f14131d660d05db6c86bff32a793c3b7bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06d16f14131d660d05db6c86bff32a793c3b7bf8"}]},"branch":"refs/heads/master"},"fa4ae18a7e13d5ca6eb0fdc1437763dd0f2d5048":{"kind":"REWORK","_number":5,"created":"2019-11-16 06:10:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/693653/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/5"}}},"commit":{"parents":[{"commit":"1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a","subject":"Merge \"update constraint for oslo.utils to new release 3.42.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-16 06:10:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-16 06:10:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa4ae18a7e13d5ca6eb0fdc1437763dd0f2d5048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa4ae18a7e13d5ca6eb0fdc1437763dd0f2d5048"}]},"branch":"refs/heads/master"},"7d53c79fbe458c287ec0a82036a6ce379033784a":{"kind":"REWORK","_number":6,"created":"2019-11-17 06:25:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/693653/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/6"}}},"commit":{"parents":[{"commit":"1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a","subject":"Merge \"update constraint for oslo.utils to new release 3.42.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-17 06:25:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-17 06:25:18.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d53c79fbe458c287ec0a82036a6ce379033784a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d53c79fbe458c287ec0a82036a6ce379033784a"}]},"branch":"refs/heads/master"},"ca632fa9d37c7ae11cf187903db64a33d1dc9ffc":{"kind":"REWORK","_number":7,"created":"2019-11-17 18:56:13.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/53/693653/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/7"}}},"commit":{"parents":[{"commit":"1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a","subject":"Merge \"update constraint for oslo.utils to new release 3.42.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-17 06:25:18.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-17 18:56:05.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca632fa9d37c7ae11cf187903db64a33d1dc9ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca632fa9d37c7ae11cf187903db64a33d1dc9ffc"}]},"branch":"refs/heads/master"},"d1fdc51f309276b77b8625c5a5c0e738d8537233":{"kind":"REWORK","_number":8,"created":"2019-11-17 20:12:09.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/53/693653/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/8"}}},"commit":{"parents":[{"commit":"1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a","subject":"Merge \"update constraint for oslo.utils to new release 3.42.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-17 06:25:18.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-17 20:11:59.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1fdc51f309276b77b8625c5a5c0e738d8537233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1fdc51f309276b77b8625c5a5c0e738d8537233"}]},"branch":"refs/heads/master"},"fc12b9da7985936b03686de43985389a22c65cd7":{"kind":"REWORK","_number":9,"created":"2019-11-17 22:10:31.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/53/693653/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/693653/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/693653/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/693653/9"}}},"commit":{"parents":[{"commit":"1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a","subject":"Merge \"update constraint for oslo.utils to new release 3.42.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2688e23e98b45df6f87f1df941b8a3cb3e4c7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-17 06:25:18.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-17 22:10:23.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iafa7fc5be697f3786094c69343e26abb8671cc18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
