)]}'
{"id":"openstack%2Frequirements~693819","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I4c858e0790a842c36f9a92edd7ac658220af9ca4","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I4c858e0790a842c36f9a92edd7ac658220af9ca4","subject":"update constraint for mistral-lib to new release 1.2.1","status":"MERGED","created":"2019-11-12 11:40:28.000000000","updated":"2019-11-17 22:06:21.000000000","submitted":"2019-11-17 22:06:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693819-1574028381733-a4121484","meta_rev_id":"7461b17048d6318305e0144468cecd43b91e9fcd","_number":693819,"virtual_id_number":693819,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-17 22:06:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-12 21:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-15 05:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-15 05:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 21:02:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-15 05:03:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-17 22:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1716e787e72d4b186ba4ff789cbba752217ed3e0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-12 11:40:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7c4622249f612cc9b36951c4dc4f011db26d5d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:19:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6df0dcf122af4d0e88e3e35d62562404 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934cd49b020a4aeb8ee67abb0c85a92d : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/614a91b1c4c34752b34759a8071d1296 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9520dfb0e6dc40c6ad0653cf40af1a01 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/018315e3099f4258a8af1582cfd7a9ff : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e129961df7b34afca325484f568356cc : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab035bd0e1274538af5f3be7f28f0e99 : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8d8c6fb5aab74c6cb152aa1636f1277b : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2311f5cf2864810ab97d158d410263e : SUCCESS in 8m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8ba47e0526024641aa96a62a149aac64 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0d23e2d7194846b394ee3618d9b382f6 : SUCCESS in 8m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7b825928e3be4637b6808c2324e40af9 : SUCCESS in 4m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/773f023e40c84ffdaeccc005af9357c0 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cea0d2b1a0ec45c0995fd16531e1c110 : SUCCESS in 10m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e12003f8c334afaafde1cc5d5e45106 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9a5e0c0cbe8442118f218ec0f10d98a5 : SUCCESS in 6m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/36b05175c89e4ded9b300b2e4b41ebbc : SUCCESS in 13m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a84d2eb89c614a93869726b45f9b9c78 : SUCCESS in 13m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c7043fcb4b774811abce44afbe4c9622 : SUCCESS in 23m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/491fe1cf51354151a2b9457c40455065 : SUCCESS in 24m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6f583baf1f804a12b8ae594974582b92 : SUCCESS in 14m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55192e03f89f4cb18e0c73627c7bb6e5 : SUCCESS in 13m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/95db49883fb04f2f9f4c2e35c3e00d10 : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea15e33bb29746669c3e27c927613aa6 : SUCCESS in 20m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b62e04f82434c16b03032022e15b034 : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c5c2bf5c4c79413eb70c40f29ab6f67f : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1a21dbdef09147b5bcc7a1af55259087 : SUCCESS in 9m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/956aab5a964a4a3d9019364d1f2ab6bc : FAILURE in 33m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3c725e80223481f96ca911632fa38ff : SUCCESS in 1h 10m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50aa844804f84a85a4f41095276f63e9 : SUCCESS in 1h 33m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7cd84240858f4a068925467ef2f5d784 : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fe6d6ef4f8318ca0dc2ef385388f18d95bb6b4f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-12 16:23:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05c62bf19986a1a8cbc94a3a6a9fac7e0424ca81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:15:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/deb334899b284211a568f6bfe0e11a3d : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24100cc1e7f54eefa49be307bfeed0ba : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/397175da0d9f4a0f848798beb7db6459 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62365a2a9bdf48839efa08a5c3b5cf7f : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e529593ed81146ad9bd05f3d8cba9011 : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/704d9fa8f85a4454b15ea251a2ff2c53 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2dad19cb2cc74e63a400877f86b1d7fa : SUCCESS in 5m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/75aa64d8e6cf448584a90c0f280905b3 : SUCCESS in 14m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fcc0960310964d87bfb4611d61ab6c50 : SUCCESS in 8m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d06efd79ec374a8d9e11bb3ff6dd6729 : SUCCESS in 10m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/feaab2e9707d4988a097bf5e693cfa17 : SUCCESS in 10m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/47f70f80a4284f69a9a2e2a3fa4b44d1 : SUCCESS in 6m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d259f7658d2f4abf9dbb70cefc1dcd87 : SUCCESS in 5m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ff4c1aecdf414189838619533da31631 : SUCCESS in 12m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89506598d5164dfcbebf1ab164354638 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48fe63c7e3e34cbc9aeaabb92cafba4f : SUCCESS in 11m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5c39fe1cbb874d3f999d6df2b688f040 : SUCCESS in 14m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ad11aac35d5444c3804bb4ddf4297062 : SUCCESS in 15m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9438eb50c3044c7c86a4fe6a9f718ebc : SUCCESS in 23m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8de6f8e2747440149a2d8144f5c6673f : SUCCESS in 25m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/892184d3dcf848acbc93f1721d644317 : SUCCESS in 19m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7d9469ffe0084480b315f5474e9aa3e6 : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9f6320b9836b4223b79f7f47af0f5918 : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d7bda43ab554db5aab6a83eeef6fdff : SUCCESS in 25m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0804388c00034d20bc3a46d3b3758cfc : SUCCESS in 7m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c28f24f5c75409d9bff63c8757337b8 : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf6dbf1ee92648caac6f154da8f460c8 : SUCCESS in 13m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8136959f43e94769b683d868f97e3ad3 : FAILURE in 32m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f356b5446ba8485e95e1e990d9773d4b : SUCCESS in 1h 16m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa8f706acd6d412680b9c0488aa4766d : SUCCESS in 1h 39m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3adc53b73cc54e38801203a78662f49b : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":1},{"id":"81f52291bf0ef80dbefd0e15c94a6440c038f613","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 21:02:25.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1dc39ad9b142cfd347017f80c932480da3457dbc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-12 22:46:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"34eee2744d4e3e95a931fc70367c983d7eaa46c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 22:54:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7a0b8cb3f1424fdeac6e98dfe24825b3 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105f88644c6d458ab41750f55a539407 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c57ab95f34be4120838568be2cf46363 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0d9d0a324f470a8fcba4ee94f286c3 : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/378b6f56575e47afbc6f2da855d5e36e : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4eb1494ea0db4763abde4bf26010913c : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d10d4a6678b84bc08ed8adefe73bc883 : SUCCESS in 4m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f1961e0f3e954edca3bb1638eef73162 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6279d54c36b6404fa1d7fe7bb4c10c74 : SUCCESS in 6m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/be8777eebaa54bd8a32507040d479710 : SUCCESS in 12m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3b16d4ebc1364fcd988295df50286c20 : SUCCESS in 9m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3083f87565c84f34ad74c7f6a397d02a : SUCCESS in 7m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a14e7233be7c4374968dfcf339bf09fb : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9430ea89238a4225ba6d104f3227a374 : SUCCESS in 12m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd31e0f91e40422fa08f83176faf56e8 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0098b77a5e3748908dc944402ff8eb2f : SUCCESS in 6m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b48c7c22d20f41688026b69bdba91913 : SUCCESS in 25m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8dcdfd5b44a4752ac1b00e48f15df0a : SUCCESS in 25m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/462f27c279034365bed50596dd712198 : SUCCESS in 24m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c8073b39efe749ed9595412112ceeb5d : SUCCESS in 24m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1089c4be6e64416cba9a1e6b2f30a991 : SUCCESS in 14m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fae18ce004a0438f9dd293592544bbb5 : SUCCESS in 18m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ae8e622d44fa4f85bc9868f928c59737 : SUCCESS in 18m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7778745743b441f689c13ae740e4a1ab : SUCCESS in 20m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73419a014d6941e8a747fb7bf886a391 : SUCCESS in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b5ccb5a67885426f8575f3b97c23ec5d : SUCCESS in 19m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d6febb57c7c4a2e89e8a7cf05bba0d1 : SUCCESS in 8m 33s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6a535483d8f141d09557dbbc94994d59 : FAILURE in 39m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c292fd4ac81f40339048b43b1f2c9663 : SUCCESS in 1h 05m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/889459984e6d47cdbcec30286c5a5cf6 : SUCCESS in 1h 46m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f5a7399cf7404740b689592e1e0c7da2 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7455dc5925d242ffc495d9dd48f0a3e8b1285b0e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 03:15:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e68a65a39d8f6dc1b765c5b387f629b17c5cf5d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:01:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1c466f8565c74aedb608b33c842e3367 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b97801824d04f8f9ef0914fd5a45103 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44a6106976484728b8e677798d3ba1c0 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/845d5bac9e4c424e9ee731378a368096 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/580d7f331ceb405db1b6cc30e92e6a65 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05576b2faf6d4e09b30033d64e75bae6 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6dde6c7b479d475abed4236fd913018d : SUCCESS in 4m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4b9b25077e94441d9796b414a581e13b : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2dbbdfadcf2e442c937a3682fd4dd13b : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e4cb21b543214b9197ef47bfd9a8c17c : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eaa2f6a1fe2e421cb7d7c3cb6c29a20d : SUCCESS in 12m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f29fb2f8581f4574807d3d8cf3a3c698 : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/64c43e7a701c4314b88c0bb7cd5cecdd : SUCCESS in 6m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/314036853f544ce1b81ee2179f849bbc : SUCCESS in 10m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3bf6d8f34f2c433490f1da161687e931 : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d448656c83644eb49344b042ef2481ca : SUCCESS in 7m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/aaf007f3d5724a18925dd4bd7018f4be : SUCCESS in 24m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/52973d84634b4bd9926a5d4466f85849 : SUCCESS in 17m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b7c4fb79ac7d43c4b42100ffbee20e83 : SUCCESS in 23m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a633210bd1e7480f8c2b670ce13f32c0 : SUCCESS in 26m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2eefb183de054d889f954b9f16cae464 : SUCCESS in 13m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bd50ecb2ac47413389b25bba52aaaa52 : SUCCESS in 15m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3612e3fa20be43bc8d28732f7d010c10 : SUCCESS in 19m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02fa9a7f3cae4e808439dc4d885b5f78 : SUCCESS in 18m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b04df366c1f542ec8c16ec1caa40666e : SUCCESS in 4m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/77ec73244ad0461fab29fb9070bcc2fc : SUCCESS in 15m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09e6c91599d44dea9548b7c062e11db5 : SUCCESS in 10m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f425f5c3a43410197c2a5fa978b401c : SUCCESS in 1h 06m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/725549e23cba43aab9a742197680724f : SUCCESS in 1h 43m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd45b316f0504424938d00a334c54d1a : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fa1d77361369b12953aadf734622ead8dda1c629","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 05:03:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68ebbe8d99f408f1f36281656df7d6f950c53355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:04:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"789f49181ae2398f2ca9184e6107c4c47318ee11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 12:03:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2387d01149d24aeabe8ddd4be318a228 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18be470691f34f2aa9a7777ef27b85c1 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2303aa514e7d41bbb3ebb40038c8ceeb : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f37df156c1dc4a28889a72a5ac9ea733 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e9b64f0343e45699bb7d502a3b3f6e4 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc9de42b38f047ddaf16033c900db1b5 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f3bf334e64364717a79afd14f901cdee : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/08cecdd8642b4abcb9b14a7307ea0aff : SUCCESS in 8m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/57f427a370a7408ba754e1815792f272 : SUCCESS in 12m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7521614a850840a9a275f139c931d0cd : SUCCESS in 10m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f06f716076434b1c8a2824aad4f3b83f : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3cb054db1fee4510812dbf1553ca834e : FAILURE in 7m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e664ec83b70f4d2ab58ae0306bcadb30 : SUCCESS in 15m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/daea73c36a524f359f67599b21bafb62 : SUCCESS in 13m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ccffc20e4e14529a7667140fd601500 : SUCCESS in 12m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/25b5eaed91264b56ab00b1360f007a87 : SUCCESS in 25m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0968472ac18a430ab96a0c20784a82d5 : SUCCESS in 22m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8e25769c0b9d4730a74f1eec087abd90 : SUCCESS in 32m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b67cf846e0e34ef396186feb0cb2b87a : SUCCESS in 25m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9238284c00184ee392ef9244bbd19e85 : SUCCESS in 22m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6587723ae8c644da9d4e56cf06c914c8 : SUCCESS in 22m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/79f55e79ae634cf3876ec20dc642a005 : SUCCESS in 14m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81d7348d27e64741b3c8879d93d4110c : SUCCESS in 20m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0007160ba24249b3bb36e5b73e6e0fe8 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/57fb8f6090324328a7fc9a0daea7f785 : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a49b44e8e2434e189c03bec4b4f3e4af : SUCCESS in 11m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f1e1a60541d465bb49ebc4e5614d9db : SUCCESS in 1h 16m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f812b73bcc9d4d689670c6007ab82326 : SUCCESS in 1h 59m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d35779a766c344e0852495af89fdf905 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3c6837d3e470f85a13078475582e54f83b71e96e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-17 18:51:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7d5feaa43255acf8eaefd1f9a1320c904a891bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 20:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bf4b619581074fca9fa6e275276acff2 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a08896c5b174ab8987b05e526df9877 : SUCCESS in 15m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66e8ceb9750342ec9179e97160066ff2 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b089eb4694344bbaa43a494239cdfa11 : SUCCESS in 2m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/506674c7acad40988a863b7fbd202fcf : SUCCESS in 19m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2b81784ca2c4cb38662959cf20c3f15 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ded664650f84bf0848c973fb4667a26 : SUCCESS in 3m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2c7be5ebc7e843499f8e2cb62d3962e5 : SUCCESS in 22m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae4da4b1aa78429daf3ab2846402ed51 : SUCCESS in 23m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3f7867ba930b4cac8866f6dfd975bb42 : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f7a581213c3a4e9b92e4e187dd793f3c : SUCCESS in 7m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2575db8ddd634e1e9b8b27beafa873d3 : SUCCESS in 4m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6648d634b7324b6bb035e5824ea70423 : SUCCESS in 3m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5713a8cfcfa84f509561174a0890e4cb : SUCCESS in 8m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88b52f9f36064ade95f1a9f1e8e2f741 : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5080e385cd2c49c2bf249ac1ebf9ea93 : SUCCESS in 19m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ee8e97013e724395a2be5541fdfdbc22 : SUCCESS in 12m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9382c395a31d47e1b7d34c40e4ea2114 : SUCCESS in 12m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/67f5b971dab54010918c021ec8908e7b : SUCCESS in 22m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ea20e5f306440048c9de75bbdb247c8 : SUCCESS in 24m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/120effcb4aa14ad992bbf42e53959343 : SUCCESS in 12m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/63649ac3da7f40ce890587dcb6205e65 : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c176ae172f714786a08ef9bf29ad326f : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9844d6987a3b49bb8def3b99600cd62f : SUCCESS in 23m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6be6554ad63a4bbf95d3a903444d871b : SUCCESS in 4m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c11925b85fd94e5dbb8c07ca80dda14d : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9dad5a60ad754672993f651aa74ad74e : SUCCESS in 8m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da1419e0bc144fa78274f4ad3fe04859 : SUCCESS in 1h 05m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fed5f249a3de41db827402811cb0ca77 : SUCCESS in 1h 37m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d2224289d45d4e098fdb9759e7b3956d : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"76196be094d1a2bd451d6ac278f83161d0a45ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 20:36:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a208a048d83675a3998a40ea62b2788441a699a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 22:06:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b2791cc2ed44acb75fbc6c0c2c1a45 : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27c007f6b711406286d64a8ee0a2981b : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66c8ae741bb24db1add437d170abc862 : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/258006ecb4c44749a560b641f77b5a1b : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07159702595d45518d0e0e4c11490091 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96c168153c8841078ca74519c0f573d2 : SUCCESS in 2m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/102e607cede4490db61212820254666d : SUCCESS in 6m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e70b000ed9f74bfeafdf1d9590fbddc4 : SUCCESS in 8m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/665888b750d149df92be6cca3918c3a2 : SUCCESS in 10m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6cd989e157304e00a698f50168119095 : SUCCESS in 10m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/19207bc0ec69458ab614126bcc48d88b : SUCCESS in 4m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c20164b82eb346939f32df3cfaa0c145 : SUCCESS in 7m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3ba1f3a51dda4d1499d466b4540851b3 : SUCCESS in 12m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1eeac8369f1d4736970a304d99b84dc1 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd3cd2638fdb40de93ea147f3ad11f04 : SUCCESS in 6m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5f7de1ad0abf414f9abaf98ac87d54f0 : SUCCESS in 13m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90880b8395314381bdda2011c1649843 : SUCCESS in 12m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cf15bab884484d7298cb55394b05cc91 : SUCCESS in 22m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8d83ebeb48084440bec5c0e2a66b9ebb : SUCCESS in 21m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e524f2ca22ea42898a160bd5a5307b5f : SUCCESS in 16m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd315d3ca4db44c3b7ce785893ae96bc : SUCCESS in 11m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/85e6383eeeaf454c9e99aa1309e262fd : SUCCESS in 12m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6ee8786facd4265b62b39c4cf650dd5 : SUCCESS in 22m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c19d308c972f44ef9e14134fddea238b : SUCCESS in 4m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ba824c60d8a046f7b1c73d28b13ed96f : SUCCESS in 12m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/77dee5b418264dd49c58732e6de06504 : SUCCESS in 11m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad4c6df27c9443e59d0691fefa6068dd : SUCCESS in 1h 03m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/254362932de54112a0b892783db9020d : SUCCESS in 1h 24m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4073fd198d804e14afbcd0e71a957923 : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"98cda1cca3874aa9b2ae7347a230592c3593d338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 22:06:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e22fcb6e7c4654caaa362be544a70e60b8907f7","revisions":{"0e22fcb6e7c4654caaa362be544a70e60b8907f7":{"kind":"REWORK","_number":1,"created":"2019-11-12 11:40:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/693819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/693819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/693819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/693819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/693819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/693819/1"}}},"commit":{"parents":[{"commit":"17d99de9237dded3b27b25f1844d7b8276f12087","subject":"Exclude sushy 1.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17d99de9237dded3b27b25f1844d7b8276f12087"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-12 11:40:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-12 11:40:23.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 1.2.1","message":"update constraint for mistral-lib to new release 1.2.1\n\nChange-Id: I4c858e0790a842c36f9a92edd7ac658220af9ca4\nmeta:version: 1.2.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dougal Matthews \u003cdougal@redhat.com\u003e\nmeta:release:Commit: Dougal Matthews \u003cdougal@redhat.com\u003e\nmeta:release:Change-Id: Iea7280546122199e4ebc603be473b470f2d817fe\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e22fcb6e7c4654caaa362be544a70e60b8907f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e22fcb6e7c4654caaa362be544a70e60b8907f7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
