)]}'
{"id":"openstack%2Frequirements~693822","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ifa9fc84ec3664c9b2b1bf228e12903ba6402a631","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ifa9fc84ec3664c9b2b1bf228e12903ba6402a631","subject":"update constraint for os_vif to new release 1.11.2","status":"MERGED","created":"2019-11-12 11:45:30.000000000","updated":"2019-11-15 11:48:21.000000000","submitted":"2019-11-15 11:48:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693822-1573818501420-752aeac9","meta_rev_id":"9ed4ae3f61f71956c0e7116cf3f682c8822ca2ff","_number":693822,"virtual_id_number":693822,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 11:48:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-12 21:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-15 05:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-15 05:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 21:02:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-15 05:12:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-15 11:48:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9989d48c56ebf79582bd871375fa39f70d1d47a9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-12 11:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d6dc32df0ed6be96a584f0acc08edc7bf2cb62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:39:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/45011352caab4126a9ba727122acbb01 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b1c20f4ad44cf3879d5c052626bd36 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3504817856e34b8ea8118b76031fb096 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a95ce1edd2c745fa9c817b9750878a13 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/917f033a3012422aae6a75a9dd896cf4 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef06a3fd6910402992d1872fe008aea9 : SUCCESS in 4m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dbcfe71201b349ff9e33ad9dc0a1684f : SUCCESS in 8m 30s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/55cfa4c483c543339edc8c53e72d32cb : SUCCESS in 8m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6d76d4404bc946a6a2221ea0913e7fc0 : SUCCESS in 8m 22s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/12a0c2308f9e4a5db6216c869fd29baa : SUCCESS in 9m 32s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b9304a912e92476c8d44955b1f75e5e6 : SUCCESS in 9m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2dce7cdb3e2b4c2abcb49a9046675722 : SUCCESS in 5m 35s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3e49e3671bfe410ebf15b8b082393ce4 : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ce95aef4bda64a348570db8b5fc4b33b : SUCCESS in 10m 08s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/4ddcc8c32b2d4944a255b3790b6c498d : SUCCESS in 8m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/94e1639d46014d4eb3297f6e66bf62d8 : SUCCESS in 13m 25s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b38822bbac9d452e9f123628d519c331 : SUCCESS in 17m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2e51ba8acf03409d89d73c7377f62520 : SUCCESS in 27m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/6ab6634340b5463897719901b8ce5127 : SUCCESS in 21m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/390f8c0b942b418fa8d49e562d85ab07 : SUCCESS in 13m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7f52a9fa2e2e4963831513184e88efda : SUCCESS in 13m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f76f06a0e4d413f9d942919944df81b : SUCCESS in 16m 10s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e3281f36807844189d8930c3c1ba5cbd : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3e7b37cfd7ad40678cbe9907ecea9ef6 : SUCCESS in 10m 50s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/691596ab1b9541179b1386edd7609ed2 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3113b5213f45f788e0d0fd658bbba1 : SUCCESS in 5m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/57fe495367264a6f85375e8de071a53c : FAILURE in 34m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b89ea7e995ee45deb4bcfaad60b2bbc9 : SUCCESS in 1h 02m 28s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e43c3cde289c4bfa860a0699914f8532 : SUCCESS in 1h 46m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/61b626d9dc9a4104b38a72e4b496f75f : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6dc00eebd17255d84c7cb15caba6a41aa7aeaa5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-12 16:23:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2a98008866547b58d52d27238366bac051decc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 17:59:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0df04dd27c554ba6ac4741a68389389b : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ecb1a4a6154ffd9948cd0beeb079cd : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d406d288e02a4a15862833890bfd7e7c : SUCCESS in 5m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4a760264441446e68a744e80da5a3383 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b945fbb48fcb46d7b10c9f842466f93b : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7970e101aeed4b44a8ac0ebdd8d1bc25 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7bb19fc0cb4d400795cb751e77639ef4 : SUCCESS in 8m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f6d13afe81ba4c6aba1e46c1e9cd25e7 : SUCCESS in 8m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e1ccf2a53ff4900906dabfe4a7b3b4d : SUCCESS in 7m 53s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/31f7a257f61d4c789bde7aa5b231ff01 : SUCCESS in 11m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/468d93071a694dc89393ee1026a3ddcb : SUCCESS in 12m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/87dc9439d776462eaeabfffb6a538252 : SUCCESS in 5m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/86e1a84bc7c34a08929afecf9c17b3d1 : SUCCESS in 7m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a044ccdda1f14814a55d69a13d68571c : SUCCESS in 11m 57s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cf0dcaec65f748578b5a1c532730e127 : SUCCESS in 10m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4d7250fbdae3496698328acefa2f0f05 : SUCCESS in 21m 47s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e4688b055b1c4829924432cd95686a20 : SUCCESS in 12m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cbb597d7a6a244b786e6a253feaffb8c : SUCCESS in 27m 17s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/491cade5a49144a5be1bd16947c1b020 : SUCCESS in 26m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/29492371390c46af814b7ccc266e9134 : SUCCESS in 15m 48s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/821eda2164114237a1108b6287b9f5e3 : SUCCESS in 16m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d43fb864c022414e8f6413833391b046 : FAILURE in 19m 35s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/82a8235b029e49d490283a0b899b50c8 : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b008430d5e7e49d3909123805a7002d2 : SUCCESS in 9m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e7b71b3d6427472aa21d4728674deebd : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7506b454ac73473b98e9f6b82e7af68a : SUCCESS in 5m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/13bb2702c5ca469eb5abf56e90250f19 : FAILURE in 31m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/048082e9261841e9963e2b0db468a48f : SUCCESS in 1h 17m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d2f732a09d8546ba8b4d28c1c165dd23 : SUCCESS in 1h 26m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/88cc47e90f0c4acdbcb3ca933d27dc94 : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8684db89697a41c93db9e7fcb698628b7bc094ea","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 21:02:35.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d5dce770bb507cd646d4e0d407e33468e2b180e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 22:45:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8f1c3e2c1e934b7d90771e233589f93b : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8320979485e6412a911cdef7212172bc : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99bda2bdb7724dcc91f65909a80dcadc : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd198f9b80e14d189b889e3997d04a43 : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c46b8853b1114cc1a889ec39b96a514e : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2be343e9dad49768fda8f02e4dae42a : SUCCESS in 4m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c9f4ead0b3f24239a3691a02e3e60413 : SUCCESS in 7m 37s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7219edffcc7e49c085e9f3e1e9e3fd2c : SUCCESS in 8m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da324fc405b449efbf064eea1b6928a9 : SUCCESS in 6m 42s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4508a230e17d4129a9f05976f8a22ed0 : SUCCESS in 9m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/17bc302c7a504320b38510eb4c7c2b77 : SUCCESS in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8f4a0ecfe7674043a2b40d659c05d126 : SUCCESS in 6m 16s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/467c84a8085c4e839441a2473eee76c4 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9a6225cfa3904aae93ba0fe25345b484 : SUCCESS in 10m 16s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f01ed1052ad24501b1db82a301a2be62 : SUCCESS in 10m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/92fd53409f7945c39a18daf9910591d8 : SUCCESS in 21m 34s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/53fcfde4357f47bf9c33dfa296c31183 : TIMED_OUT in 41m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/164a9adf79214041bbb79535629adcc1 : SUCCESS in 27m 11s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/1a15d0b043d048c3a417a1711745e81b : SUCCESS in 28m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b5426dfaa20c4ffc89e140ec2d22f376 : SUCCESS in 11m 53s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/5584283725854bff8d82746a8263f344 : SUCCESS in 15m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e89f007dbc548058f562c9da456e183 : SUCCESS in 18m 27s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/7927f27654b44c048e2c480247a985a9 : SUCCESS in 4m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e462860cbd1747aa8352edd69437b241 : SUCCESS in 9m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0b5c73f3cb544552aff2ad0f1a119898 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4acd4c9c99794157a2143dd25e104a08 : SUCCESS in 4m 49s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b59ea41da589437da0725923887532cd : FAILURE in 33m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3aad975a89624d5aaa0cc2dce0fe977c : SUCCESS in 1h 00m 40s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8ea1b0136bbf40339e8f2c9983f0f715 : SUCCESS in 1h 34m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2ac780a26dfb473b9a48e2cd5f54e7c0 : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4515fa8f9d80e3bccdb5975881176c8421f5c2dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-12 22:46:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fddb4024396340ef2e8d7f59f34cd27c9d5d9460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 00:26:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ca4861dfc71470cb02885c8f32536a8 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af885010f4b47d085cde0a7725ec836 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/747b8b1566f3407dba73531586bfed68 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff37a11aa15c4ede956feb1802e48ce8 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c7cea553c734432a3471abec6284093 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22f3c1866ef84d32b6a99f3fe3d3b940 : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b64a11b360f74701a4c1359bce335cff : SUCCESS in 9m 17s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/725cabb74d384f889097b364adda0d26 : SUCCESS in 9m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/62b00b3d6af447249a517546ba950e91 : SUCCESS in 6m 51s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2a888446f6da40d7bd6d092141e9c0d7 : SUCCESS in 9m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/df0ed1348645421dafc579507c6236a5 : SUCCESS in 11m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/90b6e0d974a44a468c0003f61dc31b94 : SUCCESS in 4m 37s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ee657d5d20ad488e9c1e4fc7898f8b98 : SUCCESS in 5m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/24fc2b2003f6411cb66b8e3104caeaf7 : SUCCESS in 12m 19s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/51250cecd39d421187d2732c05ebacb3 : SUCCESS in 10m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/40a29c5e91eb4ebd95a088c5c525fa58 : SUCCESS in 16m 26s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b2a89a5e40464dcc8ac503690a53fb26 : SUCCESS in 25m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a62681c92b0d4c728d3ae51fbd02a12d : SUCCESS in 29m 31s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4f0855fde7144b6e914c6ba43e8e6ae6 : SUCCESS in 20m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/22d74d9dd89249a2a168948e526dd5fb : SUCCESS in 17m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/a064c04060cf454c9a281a0fec7ce741 : SUCCESS in 22m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c586c2f0e39455cb55b756de211b599 : SUCCESS in 18m 29s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/dc3e1cbdc7634d5aa88226e43a36aaec : SUCCESS in 6m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aee3cd84f335478db80127bcd2f6a298 : SUCCESS in 8m 27s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d3960a02e32f47818f60504e872da5cc : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f9f4cb97fb246deb761dc24ced4890e : SUCCESS in 3m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b5264f3421454070810364aad7c26b6b : FAILURE in 30m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5acc98812962430c9d6eda8d0c2fedd8 : SUCCESS in 1h 17m 46s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/693eb4db139e415c903989938c3b51c7 : SUCCESS in 1h 38m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/99320fcda30642d69d9f9192a89ad95c : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":1},{"id":"87f857cd3cecca8d0295cbcf761d630f29bcc8b0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 03:15:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a66b2c3b6ccebf8cb5ca01731e4adf303d72d9a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:05:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7ed9393877794c1ea3984f72bc24980b : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bacdbee3ffc24f759dc8d1fd64d77b31 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54eb1cc8c2f84192bfd08ab7b336769c : SUCCESS in 3m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/00bbeb1740c347ce9e7e848b1858ca69 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3c4d60ce7ce4e38a49bf9da04b5ab81 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6de88810db04e1185eeab85799e5cbb : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4225d6c12db342ee8d15533d5d9f919b : SUCCESS in 6m 29s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5c52479fc26548c391e8291ecd835225 : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8497ca5abce0458884c2a36f949d291b : SUCCESS in 5m 39s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/98c5efa95ee94903bb800563f7bb354a : SUCCESS in 8m 53s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/5d5f627151b9425386f0b4737ab43688 : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/74e8e486f9f543c3899c9f8fbbcb05c9 : SUCCESS in 4m 54s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/cad50b4b54784aaba43f7206557e1573 : SUCCESS in 4m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/01e3b2419ad343a885077d196e225a5d : SUCCESS in 10m 33s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d8d867d54f0e4ed69fc6e92fcd635dcb : SUCCESS in 8m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a9e4a75d7bd14afe8c944f52de2079c8 : SUCCESS in 14m 34s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/73950cba96564f748dcf41e35a590875 : SUCCESS in 12m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d7621c43817549ae9916cae253af3a8b : SUCCESS in 22m 41s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f814c5318af24a118e1ded05be887912 : SUCCESS in 24m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c3f43d2e76664069976480e8e23ce07e : SUCCESS in 15m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/daaca1fc45fa4c30b388c67d2500a8bb : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2afea33b579e401ea31b1678aca65193 : SUCCESS in 19m 44s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/859f3493ae284422a4e9ff0146bbcad6 : SUCCESS in 4m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/55e8a0ce9d2b40de9d0d026076723329 : SUCCESS in 8m 01s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b4dd7be7c1b848aebcf312b31f932d53 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd5a5d6318e34aa5a7d4da5a6e7e467f : SUCCESS in 3m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a55e731d46e4446bdaa2e6df91262c4 : SUCCESS in 1h 08m 10s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/488e81f99fe8413e998e32ef080801f2 : SUCCESS in 1h 23m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9a57043eaec74dd58d1f842406e71081 : SUCCESS in 1h 40m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fbeb8ebf20562f12762ba5b532c1638cc9242d3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 05:12:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46c3a09d17d6909149814faf939810dc4d62817c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:12:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14a8f05ee882488f8ec2ed313d4b9d9475b746b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:48:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e142776e5145db84c150a1f64cddc6 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9c2a1d1499144419c83fefbac73614a : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab2c8d72ff1147809429104cc8bb9bc8 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0eab887eca44038a076999ad20b3868 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14e21cca8da54545a4330a5638c28559 : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b3056251cf81444788b44167bfa628dc : SUCCESS in 8m 00s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a9b68c609f5e41909fc0578250deff77 : SUCCESS in 7m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0cf08074c15a4cfd94efc81b2c568c43 : SUCCESS in 8m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/4a522b28a77e490ba0cc7ecd38e8395b : SUCCESS in 9m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2811b37ed3a34c138f973ccebf0fde76 : SUCCESS in 5m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d0a97178c0ec4d9fb65da5919ea68df0 : SUCCESS in 5m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dc0266ccc2ae43bc8c74734a73d8bf77 : SUCCESS in 11m 45s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8846d88a14fe409a8ce92ce310e91e13 : SUCCESS in 10m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3ad734ba05d04a709dbcc64b3c2c8752 : SUCCESS in 25m 27s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/896de9d0204f4dbeb35d07416342f58a : SUCCESS in 13m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7b23c251b01947899af54eb679c59790 : SUCCESS in 28m 54s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8e4b66d821014c0089b7b5cd40ff9e51 : SUCCESS in 29m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b395d0508d974b86b9d95c88a03775de : SUCCESS in 12m 45s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/04169fbf73744be39838632ff63d42d0 : SUCCESS in 12m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c412bc7dc894298bd6299b75e18dd13 : SUCCESS in 18m 35s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/3c1c808a38e4495f8e39a160d64a04b7 : SUCCESS in 4m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1e827d4279b54393bb86e88d3de8a2ab : SUCCESS in 8m 10s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/be7ff7cd17a743bda32d5131eccfcaf5 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13dfa1127ac4473ea8c95c16848397ea : SUCCESS in 4m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5aad0733255045078c47c33040b1ae6f : SUCCESS in 1h 02m 27s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/cf870a2ea9d046049911893615f6a38e : SUCCESS in 1h 29m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58f934882b2d461aac880bb6d0d4caca : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":1},{"id":"875fe27aa2febb8ff41b28b2b70d06ad48e54db4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:48:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e39edc1536981507966f64c619606226517a7d1","revisions":{"7e39edc1536981507966f64c619606226517a7d1":{"kind":"REWORK","_number":1,"created":"2019-11-12 11:45:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/693822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/693822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/693822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/693822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/693822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/693822/1"}}},"commit":{"parents":[{"commit":"6d4f6a08eef526a6af777b2abc99633e912e351d","subject":"update constraint for python-openstackclient to new release 3.16.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d4f6a08eef526a6af777b2abc99633e912e351d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-12 11:45:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-12 11:45:28.000000000","tz":0},"subject":"update constraint for os_vif to new release 1.11.2","message":"update constraint for os_vif to new release 1.11.2\n\nChange-Id: Ifa9fc84ec3664c9b2b1bf228e12903ba6402a631\nmeta:version: 1.11.2\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I2347ac7bebbe59c42f3f92640bcc7fdadadabb15\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e39edc1536981507966f64c619606226517a7d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e39edc1536981507966f64c619606226517a7d1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
