)]}'
{"id":"openstack%2Frequirements~693885","triplet_id":"openstack%2Frequirements~master~I29abcba929995f9307c392110160d889a29b47c3","project":"openstack/requirements","branch":"master","topic":"requirements-integration","hashtags":[],"change_id":"I29abcba929995f9307c392110160d889a29b47c3","subject":"Remove requirements-integration job","status":"MERGED","created":"2019-11-12 17:03:02.000000000","updated":"2019-11-13 02:45:07.000000000","submitted":"2019-11-13 02:43:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693885-1573613010876-64a7fc44","meta_rev_id":"e863aac9343fd31508ff2135e2a28be19691b431","_number":693885,"virtual_id_number":693885,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 02:43:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-12 17:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-13 00:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-11-12 17:03:50.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 17:03:50.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-11-13 00:58:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-13 02:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72ceb920a386caeeb302ddf6d5e3151b12b6240c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 17:03:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4151c6112408f335149426cf15e0ae0f847a3a5a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-12 17:13:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7feaefb6dd9b3bf5d474912c0e44483aa00806b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:36:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2984d05963f42a597e3709628a9cf5c : SUCCESS in 5m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/67da574430234cedb9e858634f2ccba7 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01ec4ca27a6406fa7bd36a479c69685 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77caf282ac8a4e55992ae8ccfc53bed3 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3892b6d6c53346c3a0553bd82a332343 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73cb8aaa1135425c904a5445f51bd8be : SUCCESS in 4m 09s\n- build-wheel-mirror-centos-7 https://zuul.opendev.org/t/openstack/build/6e88fd1618924549accbceb6c39aa16e : SUCCESS in 56m 37s\n- build-wheel-mirror-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98c73d80e8e542deac77e4444ac11c1c : SUCCESS in 58m 58s\n- build-wheel-mirror-ubuntu-trusty https://zuul.opendev.org/t/openstack/build/15ee2a814d924f09a8772a08f69b08a8 : SUCCESS in 1h 01m 16s\n- build-wheel-mirror-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/90b43297b36f4ac180fca5a624838aac : SUCCESS in 59m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4115f1de70b54c8cb136cdb4723f2dcc : SUCCESS in 2m 59s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/6c3cdbfe87304f3b99362f827fd753ee : SUCCESS in 5m 16s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f53ababae00a457584530e88e31abc86 : SUCCESS in 1h 05m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/349f2dfc16b8481c818432b13257b5f0 : SUCCESS in 1h 31m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ce924b67062b4695aa6ee4c9a7517cd7 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"06d173a26877eeb02d43fa693f343d02d28e9cb7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-13 00:58:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd54ef0f666fb97b657fc3483469eecce247b270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 00:59:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e13ea774c36870750bbcf45c5439140dfdab59a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:43:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79c025dcc798042021ecc7b050cafaa2e28fbee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:43:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7238f44feb4cbb95348d3e801a452f : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b0b3ec0949464b93e0903ef72e28d8 : SUCCESS in 2m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef18486b6b724cf09ade0a5c612ad218 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86808d91b9b54ae2a1d3d37e82887308 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e3b67c648d948cab95c0cc38c9c01d8 : SUCCESS in 3m 14s\n- build-wheel-mirror-centos-7 https://zuul.opendev.org/t/openstack/build/9931a86ae71c4f4898f90db7ec127d14 : SUCCESS in 1h 06m 51s\n- build-wheel-mirror-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a13069bf85514233bb750df04dd08996 : SUCCESS in 1h 09m 06s\n- build-wheel-mirror-ubuntu-trusty https://zuul.opendev.org/t/openstack/build/c5a71bb89bc642038c8b9aa373f4b607 : SUCCESS in 1h 06m 36s\n- build-wheel-mirror-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ac08b76935f74fa69f088a692acafe65 : SUCCESS in 1h 07m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fd168932810443a8cc0059dcc37c2d7 : SUCCESS in 3m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47deafbdd6a846609c682b4925452603 : SUCCESS in 1h 11m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e25410f9090945c383caa507552ef02c : SUCCESS in 1h 26m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/384ab77fcfcc4c5893958c2188c0a567 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"582c5fe0778cccd19d3bcbae06c7063cdf541107","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:45:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2047068c4878405883156d4d2ba96ad9 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58d622019d8e92b301699787180aafe16cde53ee","revisions":{"58d622019d8e92b301699787180aafe16cde53ee":{"kind":"REWORK","_number":1,"created":"2019-11-12 17:03:02.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/85/693885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/693885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/693885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/693885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/693885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/693885/1"}}},"commit":{"parents":[{"commit":"68422a77419c84f563ee2acf7387d1ed930d7bfa","subject":"Cap gunicorn for py27","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68422a77419c84f563ee2acf7387d1ed930d7bfa"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-12 16:56:02.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-12 17:02:50.000000000","tz":-360},"subject":"Remove requirements-integration job","message":"Remove requirements-integration job\n\nThis job was a legacy job that attempted to verify each project could be\nindependently installed after all global-requirements had been\ninstalled.\n\nThis job is no longer necessary since we have much better coverage with\nrequirements-check and other jobs that make sure the requirements are\ncompatible. It is also a legacy zuulv2 job, uses the long deprecated\ndevstack-gate, and runs with Python 2.7.\n\nFor historical reference, this was added originally to PBR back in 2013\n[0] and moved to requirements in 2014 [1].\n\n[0] https://opendev.org/openstack/pbr/commit/378261a8cba818a29b7e3d33b917de5ed0751579\n[1] https://opendev.org/openstack/requirements/commit/8e073738db06086f949d2be377999356aee2d8d0\n\nChange-Id: I29abcba929995f9307c392110160d889a29b47c3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58d622019d8e92b301699787180aafe16cde53ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58d622019d8e92b301699787180aafe16cde53ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
