)]}'
{"id":"openstack%2Frequirements~693963","triplet_id":"openstack%2Frequirements~stable%2Fstein~I29abcba929995f9307c392110160d889a29b47c3","project":"openstack/requirements","branch":"stable/stein","hashtags":[],"change_id":"I29abcba929995f9307c392110160d889a29b47c3","subject":"Remove requirements-integration job","status":"MERGED","created":"2019-11-13 02:47:43.000000000","updated":"2019-11-14 19:00:43.000000000","submitted":"2019-11-14 19:00:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693963-1573758043333-64a7fc44","meta_rev_id":"4c2534ce633df536770af633ff505efc69792535","_number":693963,"virtual_id_number":693963,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 19:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-13 12:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-14 17:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-11-13 15:43:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-14 17:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 12:08:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-13 15:43:21.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-14 19:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"910981d713d42e2e863df5780c6536327119168b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-13 02:47:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5356a37976b14938d7374577c494ca18ac4b0bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 04:02:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fb4278d64df04573a8dd24a74734ac3c : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6535a422c4c425c9a94800048724120 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1eaa7ab962e44b2ebb14d208d2eed9c4 : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/056bf42b81874fcdaa97bb1b8da48749 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0860f980722a477ba8281e89a1601454 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab9a58524ece44dcbad6f13f1972d9b5 : SUCCESS in 3m 03s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/e349780ca8ba43c6b775f7f060ef033f : SUCCESS in 2m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a3eb8e02e3c40e99ea9cde26436b87f : SUCCESS in 1h 00m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4817170e266442458d95ce0db7163695 : SUCCESS in 1h 13m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4384d557bff24da98d3a796702a066b4 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1abfff4e3d0968ef16cbef6952db1abe3178766a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-13 12:08:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24aec369c4e6fd93dcc5c967407ba863ba08595e","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 15:43:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"da21fb55718ceb26b5bf5ec97514aa658865b6d9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-14 17:07:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6883d3191a56806842a6ee58671deb9456955f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:08:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"104517eeedba25dc76b3fc2bca64d5a87582bd02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:00:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8102535e6ce84e898f163f1baec7333c : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f992549d0204e55adbfda546282575b : SUCCESS in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6d70a71c662046ec92d4ba3269185ccb : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d5fc48ec67468d95a431bfd73a201a : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f356704c955c40fa996d984a768e547f : SUCCESS in 5m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6652d721c6474ea9ac592ed8f2648674 : SUCCESS in 1h 11m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf3795b2e7df4306be61177708b661c4 : SUCCESS in 1h 44m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5639d020405b47588ed18a1b646a15d5 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4c2534ce633df536770af633ff505efc69792535","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:00:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"442891e31e28bbc430be722ba50058ceda9e2308","revisions":{"442891e31e28bbc430be722ba50058ceda9e2308":{"kind":"REWORK","_number":1,"created":"2019-11-13 02:47:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/63/693963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/693963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/693963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/693963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/693963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/693963/1"}}},"commit":{"parents":[{"commit":"e28c429e77f597091f011448e9a2d47be9d69a52","subject":"update constraint for python-swiftclient to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e28c429e77f597091f011448e9a2d47be9d69a52"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-12 16:56:02.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-13 02:47:31.000000000","tz":-360},"subject":"Remove requirements-integration job","message":"Remove requirements-integration job\n\nThis job was a legacy job that attempted to verify each project could be\nindependently installed after all global-requirements had been\ninstalled.\n\nThis job is no longer necessary since we have much better coverage with\nrequirements-check and other jobs that make sure the requirements are\ncompatible. It is also a legacy zuulv2 job, uses the long deprecated\ndevstack-gate, and runs with Python 2.7.\n\nFor historical reference, this was added originally to PBR back in 2013\n[0] and moved to requirements in 2014 [1].\n\n[0] https://opendev.org/openstack/pbr/commit/378261a8cba818a29b7e3d33b917de5ed0751579\n[1] https://opendev.org/openstack/requirements/commit/8e073738db06086f949d2be377999356aee2d8d0\n\nChange-Id: I29abcba929995f9307c392110160d889a29b47c3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 58d622019d8e92b301699787180aafe16cde53ee)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/442891e31e28bbc430be722ba50058ceda9e2308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/442891e31e28bbc430be722ba50058ceda9e2308"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
