)]}'
{"id":"openstack%2Frequirements~694017","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ia4477ceb00560955e74bd0b080138f48637a2704","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ia4477ceb00560955e74bd0b080138f48637a2704","subject":"update constraint for mistral-lib to new release 1.0.1","status":"MERGED","created":"2019-11-13 09:20:20.000000000","updated":"2019-11-15 11:48:12.000000000","submitted":"2019-11-15 11:48:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694017-1573818492210-a3a83055","meta_rev_id":"85958ead094007c404c79ff0f09b4885abd83e60","_number":694017,"virtual_id_number":694017,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 11:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-15 04:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-15 04:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 04:49:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-15 11:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fc6a624333c4a6f77227ef651c79fcac8ba9bea","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-13 09:20:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbe8e5d9ebf3e8c99be29dee378086ffca71ab21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 11:11:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6a786aa16137439baaf9c995acb94a6a : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9116e308e612451bb74d1f43b09737ad : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7d0ab32002d44fc97528e235c2fb4de : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a2c54b7b17be4912aed4f41627ab41c7 : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfbd443b2a92475aa9f93454822bd54a : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f04cebbe5bb4c6c9da1bc4d71a42862 : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/36700af51ca9443581b963b6bdc134ab : SUCCESS in 7m 40s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1e7a9a3896154bc592990103d3dfc567 : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d605ae2acc7460aaf28ddbb2da81a10 : SUCCESS in 7m 08s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5b08699cddb74423b5ec79af34ee0fca : SUCCESS in 9m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6b7e43d7292d4ec69c42d3ecd1371dcc : SUCCESS in 9m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/92eca381b6a54004893af25a3ef0ceb9 : SUCCESS in 5m 19s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9c7d5aace49a43a8b2142e4a2303f5c6 : SUCCESS in 5m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1804322235414c458020ba8ecdf80169 : SUCCESS in 10m 01s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1f0ee76449c74f498e734bafc7ed82bd : SUCCESS in 9m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cfc540ff3efc4f35af161ac73483cf1a : SUCCESS in 16m 18s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7897c94005e143eaa0310b775a3a3fc5 : SUCCESS in 14m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/70cb743f963042d2981f74943cbaeca2 : SUCCESS in 26m 48s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/62359179819a4c4798b279f97cc40356 : SUCCESS in 20m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bae583cc557d4e08b47684563586563b : SUCCESS in 14m 09s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/c3d0f562056a403fa60df88924147aee : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ee43d12c34e4fd58ab28d0c582e476b : SUCCESS in 16m 27s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/afbb303b8cb6411e9768dc064f9ed63e : SUCCESS in 4m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d393158ffac9472f8dc8ff23884f4f9c : SUCCESS in 8m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/522deb4c722f443687d0117be1b432f5 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5126a6015af6422aa210b3d4cf6f735f : SUCCESS in 4m 50s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/56848781804b40d6bfa3d22a32826ddc : FAILURE in 30m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/12b9e3f9c6754ecb9d7c1103fc18fed4 : SUCCESS in 1h 06m 02s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/30bb0754f3c84502b847f6abb060b80c : SUCCESS in 1h 43m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/68d4c0e119db40f881247c8a2a76d6ad : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"94a7ce3b215cafb62b3e6a4d49ee9bceb9e06242","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 03:15:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"766f5aef227e710130aaff39ff4b40992f80e423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 04:45:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c88c104d2ecb431397a30397aeb9a4e1 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd400e7ddd1489fa395e56504f72f57 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f41889b52f824da7920a4850c6ce073b : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f488f530abc445989d903ebccc2ca232 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b03225fc1fbf48a0a3c704a74374e53e : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75cf1209287444ffab15cef04cb592e4 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8e4607ec609a4fa1b15c112f6a7479d9 : SUCCESS in 11m 40s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0d5704ef78904a4ab5c45e17d97590ae : SUCCESS in 7m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72ff06be8eda46658c5b08083a6383b2 : SUCCESS in 11m 17s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fbf1f2d08a7a4f2dabd21f6e62ab1784 : SUCCESS in 9m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9a163c203eec491c8fa2291951f423fb : SUCCESS in 8m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4828523d81f447c897e1b2233608c6cc : SUCCESS in 5m 03s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3b794f32302d4d96b12ca4ecdf51f18a : SUCCESS in 4m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/678577dda00d4181803f43ff2cf7ce80 : SUCCESS in 9m 47s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/81318b66b2d34a82a93d7aa283451561 : SUCCESS in 9m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/406e5d2f64254c9a88fbd8becc9e2430 : SUCCESS in 14m 37s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/96a59de5a83646bda50631be328165f2 : SUCCESS in 29m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7d20c7bba7774eb79e08f1e9dc874de5 : SUCCESS in 29m 06s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/22511c39ea41426c8a9abbf68dc45904 : SUCCESS in 26m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ff2c81a0cc5e491b82fb05c5073bb606 : SUCCESS in 13m 29s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f5512048a4dd4323a26b04be6d772b09 : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c3a541575cb4415a0f303cc47157572 : SUCCESS in 19m 46s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/1e90dc9f94da4a9c8a6f555173f4a2fb : SUCCESS in 5m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5cd18895c66a41c2b050895361577ecb : SUCCESS in 10m 55s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/95a7852564764d18b6e93b7afb5f4be9 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/081818bda4844f16beeafbbfaeee2e21 : SUCCESS in 4m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe0437030486444e9da8d98a880089d6 : SUCCESS in 56m 31s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/a29a670650b84a92ac2d644ed72134ce : SUCCESS in 1h 23m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6b1347da3d254f42a1aeb5efaf4be51b : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3a41a7ede7df8a913095715a7b25768f11cfc2d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 04:49:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5103574bd707e69b26b91bb1f892527eeb46737c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 04:49:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2835e4aa24f0c0d32446061bb623d556de65fec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:48:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f3155804e584fd38561ee05e5cc31e9 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c49ef4ddca6148479cc8201348d08352 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/82c4b54964ca4a50b903ac5547409ef7 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4a7b70bd67e466f9c59040abaadb1fa : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10da8edc85e348e1b5238b5183ccb13c : SUCCESS in 3m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9c043221ffa54583a099b5e17d41beeb : SUCCESS in 6m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9e8130669b0d4520943aaa14221f5672 : SUCCESS in 6m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0df31308ad9446d18d5a18b48866d4a0 : SUCCESS in 9m 26s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/2e382ff6bbbf497baf07ef4ba296454a : SUCCESS in 12m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d9e06fb774a44841bc512349f0c5bfd5 : SUCCESS in 5m 10s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b5a0741f9b2a4f0fabc544420448942c : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7061b15055524bdba129897c89195011 : SUCCESS in 11m 55s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c78614e407684ea8be4ac33e26cd4a56 : SUCCESS in 10m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a0d7f05f9b244d01b25d75c7df7bad8e : SUCCESS in 14m 34s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/708c394b232d4acf8833649de5a3e621 : SUCCESS in 15m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1eba35f1f890491a8a1acebc4e0415b8 : SUCCESS in 22m 39s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/95309bbae4404c6d98464a20ae7457f6 : SUCCESS in 25m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9acece5934ef477ca83106c72d24e67a : SUCCESS in 15m 25s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/33d91476b9444537b60a8db87c25f9df : SUCCESS in 15m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c38e0dcc4c5448b3b3db7134a36f719e : SUCCESS in 18m 02s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/24797eb65ecb4055a8d4c7538523117f : SUCCESS in 4m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8447989887fc4de28e76090b972d7b36 : SUCCESS in 8m 40s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/38ad78582ca04b1d98943b69e72078ec : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74374dea68524fe5899cd8b4fe9ce743 : SUCCESS in 4m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6116f648f49e4085af8ee715e9653109 : SUCCESS in 1h 02m 18s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/3c0ec4405f654bedaa0de9f15527803a : SUCCESS in 1h 18m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/409a6908ff864b2a8c80135e2ba16eff : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a98e48ec32fd32176a47d325a69e0b55ec2e546e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:48:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a3f5f763bb17b8fff600da02ea7706fc04a2978","revisions":{"3a3f5f763bb17b8fff600da02ea7706fc04a2978":{"kind":"REWORK","_number":1,"created":"2019-11-13 09:20:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/694017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/694017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/694017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/694017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/694017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/694017/1"}}},"commit":{"parents":[{"commit":"6d4f6a08eef526a6af777b2abc99633e912e351d","subject":"update constraint for python-openstackclient to new release 3.16.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d4f6a08eef526a6af777b2abc99633e912e351d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-13 09:20:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-13 09:20:15.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 1.0.1","message":"update constraint for mistral-lib to new release 1.0.1\n\nChange-Id: Ia4477ceb00560955e74bd0b080138f48637a2704\nmeta:version: 1.0.1\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dougal Matthews \u003cdougal@redhat.com\u003e\nmeta:release:Commit: Dougal Matthews \u003cdougal@redhat.com\u003e\nmeta:release:Change-Id: Id4dcc5c39767eb3610c888f18515a67bcd20ab12\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Vieri \u003c15050873171@163.com\u003e\nmeta:release:Code-Review+1: Andriy Shevchenko \u003chuang.zhiping@99cloud.net\u003e\nmeta:release:Code-Review+1: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a3f5f763bb17b8fff600da02ea7706fc04a2978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a3f5f763bb17b8fff600da02ea7706fc04a2978"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
