)]}'
{"id":"openstack%2Frequirements~694068","triplet_id":"openstack%2Frequirements~master~I68f640b4ec843e946047f25c37c5277f09de28d4","project":"openstack/requirements","branch":"master","topic":"new_release","hashtags":[],"change_id":"I68f640b4ec843e946047f25c37c5277f09de28d4","subject":"Update constraints for amqp to new release 2.5.2","status":"MERGED","created":"2019-11-13 13:25:45.000000000","updated":"2019-11-14 08:31:10.000000000","submitted":"2019-11-14 08:30:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694068-1573720203475-fe519b9b","meta_rev_id":"85eeba46f8d0ccf5f3176d7d73a9d62cc298f52d","_number":694068,"virtual_id_number":694068,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 08:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-13 15:45:49.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-13 21:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-11-13 20:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":1,"date":"2019-11-13 20:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-13 21:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2019-11-13 20:50:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 13:27:21.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-13 13:37:00.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-13 20:08:29.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-11-13 20:46:39.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-11-13 20:50:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-13 21:08:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-14 08:30:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f92968f50e3867d977f7a12335e9745a6ddf199e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 13:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64dabff45e8f94a4214bbf09b40c87ac0b0ca489","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 13:34:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8630b8c20125ddb09efd042ca26cbcf622b3ec9a","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-11-13 13:36:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"58b9c15d0b07f91390c27e642807c1e39fe2fc8d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 13:49:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"cf53945d55130e8d9aecb846eca61ab2f4fc7428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 15:39:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b07e7a096a417ea7a642f19bec869f : SUCCESS in 5m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/347d10d2178044f88549279ab96da2df : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1813f456d64d5489d7e67a6f068010 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb54736a85e64bf9b190b6fd75987a4e : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5df766cdf8024cc99081ea122685fb70 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bccf4690a98f4e2cb3c7d03e57ef1c97 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1b1d81d738e4b909386ea0f28b442e9 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b0c926418444840b42fa2fcd398332d : SUCCESS in 5m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/675f5045c51c4f90a0c7d9a8ba398c86 : SUCCESS in 8m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/220640dc8e1843a594c2968e5e367b4a : SUCCESS in 8m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b0d849dc0b174b839ca5a83d8c88ba5f : SUCCESS in 11m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/42f63bc4ead84b04b7c3cba42edb096d : FAILURE in 11m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ec5736404f26476fb81f52a1de7bd896 : SUCCESS in 7m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d79d9e62af624ea2b14d46f6c0728ba6 : SUCCESS in 6m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/270fa61d346c414999719a20a06f1a38 : SUCCESS in 9m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/315accbcf90247e7adce5f902e727211 : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/772635782f8c4864820133a24da2b604 : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b0700193a6b14d6f8bcdf83061961d1b : SUCCESS in 19m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1dcd91f21e4a478da24e159e5be1b8e6 : SUCCESS in 15m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a53385c58b1d4775827d5cfa3b698d50 : SUCCESS in 23m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5799cd62ebb0427e9b16ff5d1f60345a : SUCCESS in 25m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8d65866c84db4446a603e0ef970bf6d5 : SUCCESS in 14m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1e7a7b6990a34a1ab43d07764693eb96 : SUCCESS in 12m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5395de2c8797489584376336047fed95 : SUCCESS in 15m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e66f8b57b3047ad8b17b53b2792fe32 : POST_FAILURE in 32m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e77b9b45799e47b998614456367fef5b : SUCCESS in 7m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/00a8ffa5d7fc402dab204cfb39971ee9 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de25c1d4a69f4694bd71a52bf6021171 : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8fac80c5487a4f768137e735ce68c4f4 : FAILURE in 14m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a59da604155c40e29c61cb50d6636e22 : SUCCESS in 12m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b050711f67f4511b52b459942861bb2 : SUCCESS in 1h 16m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/62c3532e01874a0ab5edf7ff26a241e7 : SUCCESS in 1h 34m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b369da228ed5496dbb0a42baea01eb45 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":3},{"id":"21cb1950d48d0d3355f50280998fe6d5cd0457a8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 15:45:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"630b71b30e017875ba0f618cbf1730512ddf20fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 19:43:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e7068e25c74af4a816776e68e50121 : SUCCESS in 5m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e88a4871d10147c49416913d4c5e11d1 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63741ddbd4d4905a8e54b263b5b4042 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a0123ee54f44d38a75f052d52eed7eb : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2e2903ce8d0490f826d05cabfe2722a : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a514297c321f4de5bdddc1140d0bb8db : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6404655bf85846028d8c2bf27b5dc8bf : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a028cced5176440987f839fb20910416 : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e17181bc216d4536b80083b82b8c26c2 : SUCCESS in 9m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d26eb4324a5c42668800e860fdb66226 : SUCCESS in 8m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9d379649621344d1bd7cf15c7d73d144 : SUCCESS in 12m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db562230a80a4a8489866abf9232f765 : SUCCESS in 10m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/019a33a7189d430986d35c561850d167 : SUCCESS in 5m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4cf747d71d23416f84608ef1336d4508 : SUCCESS in 5m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a41da95034314cb0995e54526a8c8253 : SUCCESS in 10m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/badc3e4cbcb64aa7be40568f032edaab : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74234496b8a94ce1a2423baeb70c5a68 : SUCCESS in 8m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4ca7c02cde294e4692e4b1db926cc1aa : SUCCESS in 14m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dd0cbf9baac84ca2b852aa09b69eaba6 : SUCCESS in 16m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c1073cbe6bfc4f128d9be14e52b41202 : SUCCESS in 23m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/463d43c3e4544d43bb6269f8189b9191 : SUCCESS in 22m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8d587c6a7ed4dedbf810a1931be085e : SUCCESS in 15m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c27bcadf48e14f61802e4b16f1a45242 : SUCCESS in 17m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cf73f7a621624400b68e41e5966a8400 : SUCCESS in 14m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0210338753af4ebbbabbe0119b508070 : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9fdb7c8d3a0c48f196a423ee7f377855 : SUCCESS in 5m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e458ac324edc44e0b7d5acc4dc279f52 : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f014ebe9ec94ce69dd91c856c7c78a2 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a188a7ba2b684bc1b2885aa2b82c2220 : SUCCESS in 12m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b7251fe729e45d4929b62deb3e565fa : SUCCESS in 16m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcb58c53a8b248d3b4f005874d0705fd : SUCCESS in 1h 14m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/edad6aa4dac84b93acf04ef29718da73 : SUCCESS in 1h 42m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/15ed9a1ee0c5403ba621f27a941eaef1 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f09fb0e86be96bade1374ee93b48bfed119dc0cf","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-11-13 20:08:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"56c954996a932a78a0b6d92599367c089cd63dd5","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-11-13 20:46:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c21535547a9546638c62e350ecfd0a2472659616","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-13 21:08:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"04ac2f969879a4ac737f7d9f8fc8d0c2d97d889b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 23:56:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b0bcb0712b8977bf58c98f5907b03152a5ae4834","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 08:30:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ce4b215bf9b3b9b240167108eb4e5650a4e884a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 08:30:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2fda08ecbad4564a8952f06894f29af : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/399aefa3641349b095c62ff1937c8079 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f96563e00af144c7a290221ff76cbe98 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1cda1a944a4cfe98aeed90aa48f49b : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff667aa773e945b184623b8e0fdb97d5 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5f019a0d254467a936c649afa2bf895 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/65ca5af0c0e74d6b9fe54537e4d0cfcc : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a5008b95554e4d3990c3245818fd2cbe : SUCCESS in 8m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e055cc7ad782428382745e4bff3083a7 : SUCCESS in 9m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1634177604c54ea0ab1e28836a2325a0 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/30e09a8ae83e4359a9bf053d4ba364f8 : SUCCESS in 11m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2c81648c78cb422190cbdcf59fc7cfd9 : SUCCESS in 5m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9b182b6964a449b78e6ca42cf33f4fdd : SUCCESS in 7m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fa4ec4a46fa644cda4e1cbc621e668a4 : SUCCESS in 12m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ca411494b0c94ad3818d610108ccc6d9 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e60e1baa1bc45389228cf93eb2fb049 : SUCCESS in 9m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7328335b7e8b416d8bb32d597e279406 : SUCCESS in 14m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b6274552fa774f31b7b63fa88d5f049b : SUCCESS in 19m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4b9c966632c3436592bf2a1719f99613 : SUCCESS in 23m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a95e9d088fda47a5931010056af6edee : SUCCESS in 24m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d0a7d98ea70447d805a34cb5176e16d : SUCCESS in 15m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/371ca0a3b4cd43369587ae54a51bb05c : SUCCESS in 15m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8c1f2c7b3dd14017bdcd127bc785a2a5 : SUCCESS in 15m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d875f8454e824727b937c159fe1ae2bc : SUCCESS in 18m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/008699683c1149a08b6958948a3f859b : SUCCESS in 6m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2c2e0f97d7fc4622bc9b0a30acf75ccf : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ba0aa40755244b892fe73531b9bbbb3 : SUCCESS in 9m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b2636c3d21cb47e49c5194f424c9d6f0 : SUCCESS in 12m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69a95b25e1414e6a8961bfb9dc3fc630 : SUCCESS in 12m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6664e53bbcef4624a7111dfbf9574c74 : SUCCESS in 1h 17m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/68bd014fc7934fc2acc3cfc216b1f33c : SUCCESS in 1h 34m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/590cfb3c49c3449394c1a8dbb05ca02c : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":3},{"id":"cda5b22c6424505546562808bbec095df6abefad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 08:31:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649125e8060947ff825d73b81159f5cc : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"520cecfb71b39dfcf4630e5daba5beec8d2ced80","revisions":{"5541beeff01cafe23ec80e4c14d5bf128f3550e0":{"kind":"REWORK","_number":1,"created":"2019-11-13 13:25:45.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/68/694068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/694068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/694068/1"}}},"commit":{"parents":[{"commit":"a9e58f7109545b6441a3c93054737080def22ccb","subject":"Merge \"Remove build_docs settings from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9e58f7109545b6441a3c93054737080def22ccb"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 13:22:26.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 13:22:26.000000000","tz":120},"subject":"update constraints for amqp to new release 2.5.2","message":"update constraints for amqp to new release 2.5.2\n\nChange-Id: I68f640b4ec843e946047f25c37c5277f09de28d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5541beeff01cafe23ec80e4c14d5bf128f3550e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5541beeff01cafe23ec80e4c14d5bf128f3550e0"}]},"branch":"refs/heads/master"},"5e71891976037fe36b8a4e7de8cd01a6992377c3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-13 13:34:00.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/68/694068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/694068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/694068/2"}}},"commit":{"parents":[{"commit":"a9e58f7109545b6441a3c93054737080def22ccb","subject":"Merge \"Remove build_docs settings from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9e58f7109545b6441a3c93054737080def22ccb"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 13:22:26.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 13:33:56.000000000","tz":0},"subject":"Update constraints for amqp to new release 2.5.2","message":"Update constraints for amqp to new release 2.5.2\n\nThis fixes a bug in amqp see this\n\nhttps://review.opendev.org/#/c/693704/\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1852058\n\nChange-Id: I68f640b4ec843e946047f25c37c5277f09de28d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e71891976037fe36b8a4e7de8cd01a6992377c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e71891976037fe36b8a4e7de8cd01a6992377c3"}]},"branch":"refs/heads/master"},"520cecfb71b39dfcf4630e5daba5beec8d2ced80":{"kind":"REWORK","_number":3,"created":"2019-11-13 13:49:53.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/68/694068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/694068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/694068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/694068/3"}}},"commit":{"parents":[{"commit":"a9e58f7109545b6441a3c93054737080def22ccb","subject":"Merge \"Remove build_docs settings from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9e58f7109545b6441a3c93054737080def22ccb"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 13:22:26.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 13:49:48.000000000","tz":0},"subject":"Update constraints for amqp to new release 2.5.2","message":"Update constraints for amqp to new release 2.5.2\n\nAlso update kombu to 4.6.6 that depends on amqp\n\nThis fixes a bug in amqp see this\n\nhttps://review.opendev.org/#/c/693704/\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1852058\n\nChange-Id: I68f640b4ec843e946047f25c37c5277f09de28d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/520cecfb71b39dfcf4630e5daba5beec8d2ced80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/520cecfb71b39dfcf4630e5daba5beec8d2ced80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
