)]}'
{"id":"openstack%2Frequirements~694083","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I81658ed225d6da3b98fc63d8454562a2f27a2995","project":"openstack/requirements","branch":"stable/train","topic":"new_release_train","hashtags":[],"change_id":"I81658ed225d6da3b98fc63d8454562a2f27a2995","subject":"Update constraints for amqp to new release 2.5.2","status":"MERGED","created":"2019-11-13 14:26:10.000000000","updated":"2019-11-25 23:36:01.000000000","submitted":"2019-11-25 23:36:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694083-1574724961473-2b47506d","meta_rev_id":"fcce92e27b64e6675e47fdfdaec41582ded1fa5c","_number":694083,"virtual_id_number":694083,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 23:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-14 17:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-11-25 19:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-14 10:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":1,"date":"2019-11-25 02:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-11-25 19:41:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 10:05:03.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-11-14 17:11:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-25 02:26:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-25 19:41:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-25 23:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8a3247e6b25685ae1170ada7630864bfc287623","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 14:26:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3e3e69a8dde29958696d3415d05cdcdaaa03da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:14:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b775a202732046fca260109009987a9d : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6323d6e33e4993aa23f536ee37b2c7 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85dccbf116c44b0c9902bca58088d6bf : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/772c60ee49a14df88acc7db6393fa6f2 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b405b2084f64cc0a64945182547c6e5 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fe57f148fd74c3eb79dd80254c1661e : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/522573800eef423ea46da623d0b7db73 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7eed40b004564bc3ae0b79c79ddacf0a : SUCCESS in 10m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/83990a1912f548788f90323cb9655320 : SUCCESS in 7m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/062bb062c4ab4f6aab9d4d21f6df8de5 : SUCCESS in 10m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6818db8d3e624df5a24b3b9cce39fb0c : SUCCESS in 9m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5bcf4c1bf8ac4a86bbd612bbdfd8dfd7 : SUCCESS in 6m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a04cd10352544fa0825e85847790e165 : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/80e88b070d0b470bb5982a32252d6915 : SUCCESS in 10m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3715c472417441e7b00bdae89f743e57 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ca38eed9b294c7daaec213530b722ea : SUCCESS in 8m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4b3efab6cc4b449e8dd0b92a4a3a0999 : SUCCESS in 20m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/571a3a642ff94edbb813584360c1af46 : SUCCESS in 23m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/babde4b39ae54b8c9772d7b48353b0fd : SUCCESS in 24m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5e51f75708ec48f6b57d10c118fd7dff : SUCCESS in 23m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7bf6b01e73244fdda4c9277bded42027 : SUCCESS in 14m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37f1ac0f965b4c13b3d5568c4e8a0e1a : SUCCESS in 14m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/19d9849566014384963f03e47295215c : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b72f53e2e78d4f8bb455027957e32392 : SUCCESS in 24m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d244e06cc1c4417a8438d0fc1677a97b : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5e31b914b25f48b0a5203c90ec6ddd89 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e147860225864b08852fe47e50c67aec : SUCCESS in 12m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/903009ff06644d3087ae629c40276f49 : FAILURE in 35m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4d1e728549746239b343f754a37a982 : SUCCESS in 1h 15m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ce352ad5b0b74e45974e375e26193d04 : SUCCESS in 1h 29m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7779bb3f44db47e8a39d27bd69c539ce : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5cf01e972afa662d89008d31aa63bd88fab24695","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 17:59:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"756e62a2d239ebbee2ae24eed62b67cad0dc9279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 22:51:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ac12eef6769403fa8c83a38f040c386 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58bab3d814045289c3641bfca0609dd : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/635c74bad03840d09d1b257a93ba4a5d : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea6bae978c944ce82fe0e73905eb622 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45b98c12bc53424b90f06cb66a9c7cd5 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0063dac4f62427a8493df16a730c9bd : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8790a869dcff418fbdc150aaf5b5fc08 : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/48d12ff4d35e40b9b87047ae49d203d6 : SUCCESS in 8m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/353b238d4eb44889a2cc05794a57d376 : SUCCESS in 8m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7f1c35a5fa774cf78524e9ac73529ea7 : SUCCESS in 10m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4be1de31fd0e4e09b7fc2d05b52a8712 : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ab1e01d55e8649b79adad7557e7e7d5a : SUCCESS in 6m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13803e235271495f8c47a5b1de2c261e : SUCCESS in 6m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e9f237a9ee73483b9be8e7396147f8ab : SUCCESS in 14m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c97d0f882ed1480ea7cd019c1a52e1cc : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c34491f86f7f4367adb443394a69abae : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b460799e3f974e9ba382b26f397ccc86 : SUCCESS in 16m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/31bf872e39dd4292b6477abf5a10f379 : SUCCESS in 16m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9d0d0b13a88d4c8a954e394d9caf6cab : SUCCESS in 22m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce3f415b1e594c8b9aad8b15c36b9495 : SUCCESS in 39m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/87d357ee553d43df86d325a754ae445b : SUCCESS in 34m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/45ea3fd213ef42f2a58d53db55b3f978 : SUCCESS in 14m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e00511b91f904e7890bdc560735009b8 : SUCCESS in 28m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5610320e1b6d41139fbd1e79ccecef62 : SUCCESS in 38m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e634bdca68654a40b3b5a4f53c480b75 : SUCCESS in 5m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3b58811a5c0d4365824ba0f7ef0504c5 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3a8f217765e542f29099493518fab979 : SUCCESS in 17m 19s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9380ee7bfd6148f997c71b5bd606c53c : FAILURE in 33m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff3c4a7f80974ecb95821b58c6ed526e : SUCCESS in 1h 10m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d5399e5edbc14dbbb2b07ae0dc24f5cd : SUCCESS in 1h 48m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/afc03e78b490410c991227ded0e6f417 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c372d559958b18663de5c6fa9ccb39edc2970962","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 06:52:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05c7cd113b0a07b50380af70a602d37e53b9adb7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 08:26:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d93d9d45224404aed6e75ce7b89f7da92f438b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:04:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/40671c34030a4da1aaa42f79bba2a6f1 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0cda7e56944dad9cce6fa248aae592 : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a191da7dda740f0acbb32972565cd30 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87f8a230100a4d92801f0876de63566f : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f13ebfe8bacf4d93b76fa7a47de884f7 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd83accf67984aaf8460a028eecaed0d : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48410529bafc4e99beba73be2d1ea9e2 : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/735a7159a24a4e05ad297a4fb62e8ff0 : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54a73b7c6d3a462a90861ca09b13ce64 : SUCCESS in 7m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8fd4a25c919042acbc92ebd4498ef4a2 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/510c779302754bd9a1eccca199ba4e96 : SUCCESS in 9m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1c735df0bad94c528006bd540a4aa539 : SUCCESS in 6m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2cbe6807559c4b049130191c3140e507 : SUCCESS in 4m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f2d4324cedc048399a889f0dc4ade0eb : SUCCESS in 11m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/493a9a9a40fb4833847e107e8b19c3ca : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87c5e8d955d34f508e7ecd70b66536fe : SUCCESS in 6m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7527f3394d8c42589ceda3d92aaf25ce : SUCCESS in 14m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/650f94e1d9754ea49740cc3e3a28749c : SUCCESS in 20m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e2d66ec666d542939f3490d2ac9a7e54 : SUCCESS in 19m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c2f14681117415486258ade234521f0 : SUCCESS in 18m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/72b3e5c834984a2b9fcd269e65053dc9 : SUCCESS in 12m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f7e0c78fe3944473a02a6e78c283b446 : SUCCESS in 16m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3b15c52d18884927b73637892388e4eb : SUCCESS in 14m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74b8a6337df74fd8a2f749888f1cfdaf : SUCCESS in 18m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85ddb348694a466ba72c4cd7748526fa : SUCCESS in 4m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/76d4f11f3bc243eebc78de67f3bee09f : SUCCESS in 10m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/340ffea294a044b1b773f7dcda90ca00 : SUCCESS in 11m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ab301e4983874a5fa7b3956b70e2c82b : SUCCESS in 57m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a718b4ab13b54acabd049f62baca0780 : SUCCESS in 1h 14m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8fb15e47aae34477baeb19c2a02fcfca : SUCCESS in 1h 31m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/98e241c5b51a425fb9bdf1b2b5579174 : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d55635281b65b9bd3b48d62d2fdf5947bf58bcfc","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-11-14 10:05:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"60ea0e70b543f28ca283ca75e017d32bb462a228","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-14 17:11:50.000000000","message":"Patch Set 2: Code-Review+2\n\nwould be nice to note that the gunicorn thing is because upstream dropped py27 but meh, that\u0027s going to be a regular thing","accounts_in_message":[],"_revision_number":2},{"id":"caa71cfe175980f56bb62bac9f17c19ed350c7f0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-25 02:26:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b80f3df93a1f56013660e8888a08a14d15ba5f91","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-25 19:41:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"255c971b3f7a03f69ac4d8be6cb07d7b961d2a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 19:41:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d0974a09ec19fc03fa8e3f926813def9a6f3f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 23:36:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b335c2c6e1404664a9b3430ab12bf184 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/156b4b48907d4563b382879239ca36cc : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86b5c09ed5aa482bb5f858b548472e3f : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88f60911dfbd4dd899e6f6022e6ecc74 : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f070e26bd4d84c63b83471f7d8080ac7 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c41af1fd79e14ed396b94b76a4c4ad95 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f7a2d2f4c0904a9cb7f9de271f909fd4 : SUCCESS in 7m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af267172fc1545de880b02c816bc5c37 : SUCCESS in 7m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d646859b92c342c2aeb712e0ec718dc5 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a02da3e75504aa299b5d83316684c65 : SUCCESS in 9m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/610ca3b32ee4471783d11346c2bbfe3b : SUCCESS in 5m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5bbaf2be2f81469daf22a625b6fbf82a : SUCCESS in 7m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9a40cf0ecd9846e7abaf16ded760a9d1 : SUCCESS in 12m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8c39c7ea18c44c6af3243206ec03c21 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ae38d85106846e3bd7eb05a9d4efc81 : SUCCESS in 6m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/73257246d626461e960e4639bb76ef54 : SUCCESS in 25m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/99fa34473b47489d9e153d9ff484db16 : SUCCESS in 13m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5e4bb34ae1504a348dccbdead7dc454e : SUCCESS in 26m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f1da357fe8db416c9e6016e851473a81 : SUCCESS in 27m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2aac019e11c64ea3bb1d941b76792259 : SUCCESS in 20m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf23d1f6f38b4df0a76915fe01765113 : SUCCESS in 14m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ac7c62c6d5564433941f2abe8de78bc6 : SUCCESS in 13m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/894922af7687422ab166dc9cdc5b0a89 : SUCCESS in 18m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3abdd2887c0b4894a863747138c56a48 : SUCCESS in 6m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/59e62ad228a345e5bd9e4db9773e8bd8 : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bc9f3a3c6cfe4df6bd13ac71ee12e0be : SUCCESS in 9m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/259d703e957d417589d02f308d5547f5 : SUCCESS in 1h 10m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b9e3757a9c2f44cdbfcaf76c2207407d : SUCCESS in 1h 43m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/315543ced1ac4b47b3ea50e04d38a24e : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dd47e5662444871d21f3c08ca4ad6a7d572bc93c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 23:36:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"932a906348331d1e706e5bea829cdb3033cffd5b","revisions":{"e10b1b97c89786305b797f654977990123f28b0e":{"kind":"REWORK","_number":1,"created":"2019-11-13 14:26:10.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/83/694083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/694083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/694083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/694083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/694083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/694083/1"}}},"commit":{"parents":[{"commit":"17d99de9237dded3b27b25f1844d7b8276f12087","subject":"Exclude sushy 1.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17d99de9237dded3b27b25f1844d7b8276f12087"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 14:25:56.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 14:25:56.000000000","tz":120},"subject":"Update constraints for amqp to new release 2.5.2","message":"Update constraints for amqp to new release 2.5.2\n\nAlso update kombu to 4.6.6 that depends on amqp\n\nThis fixes a bug in amqp see this\n\nhttps://review.opendev.org/#/c/693704/\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1852058\n\nChange-Id: I81658ed225d6da3b98fc63d8454562a2f27a2995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e10b1b97c89786305b797f654977990123f28b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e10b1b97c89786305b797f654977990123f28b0e"}]},"branch":"refs/heads/stable/train"},"932a906348331d1e706e5bea829cdb3033cffd5b":{"kind":"REWORK","_number":2,"created":"2019-11-14 08:26:46.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/83/694083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/694083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/694083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/694083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/694083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/694083/2"}}},"commit":{"parents":[{"commit":"17d99de9237dded3b27b25f1844d7b8276f12087","subject":"Exclude sushy 1.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17d99de9237dded3b27b25f1844d7b8276f12087"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-13 14:25:56.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 08:26:22.000000000","tz":120},"subject":"Update constraints for amqp to new release 2.5.2","message":"Update constraints for amqp to new release 2.5.2\n\nAlso update kombu to 4.6.6 that depends on amqp\n\nThis fixes a bug in amqp see this\n\nhttps://review.opendev.org/#/c/693704/\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1852058\n\nChange-Id: I81658ed225d6da3b98fc63d8454562a2f27a2995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/932a906348331d1e706e5bea829cdb3033cffd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/932a906348331d1e706e5bea829cdb3033cffd5b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
