)]}'
{"id":"openstack%2Frequirements~694093","triplet_id":"openstack%2Frequirements~master~I80eb63c2ed01659990ca5506be7e0e246c9586ba","project":"openstack/requirements","branch":"master","topic":"tabulate","hashtags":[],"change_id":"I80eb63c2ed01659990ca5506be7e0e246c9586ba","subject":"Add new dependency for tabulate library","status":"MERGED","created":"2019-11-13 15:24:54.000000000","updated":"2019-11-28 00:42:06.000000000","submitted":"2019-11-28 00:39:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694093-1574901581493-81677534","meta_rev_id":"6378fbcef8a183b3a13361002017a5503bc8b2e4","_number":694093,"virtual_id_number":694093,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-11-27 22:01:34.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 00:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-15 03:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-27 22:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-27 22:11:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 03:21:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-27 22:11:15.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-11-28 00:39:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8dc659834b4941a5103644fb03c2a0ef309db3c7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-13 15:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e54e27f6da27ba539bb99da11b6bd60d5e83f860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 19:27:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a173d31e81574d0984141b175c027bdf : SUCCESS in 3m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/63471b920a6a44d5957f174fc205f502 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2185b63cfd464d9ba8592c629f897944 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/394a31c902ac475aa67dff106bb24a3c : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/939dac1b5b3c41b0a0da0f4d8fc0c62a : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a612d74c91e948e6b5f4242ef29a657a : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15b920730f3245e6a9103d32b1a28cbb : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc19cc2641d64523b63240c5c5745dd7 : SUCCESS in 5m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/245de70062de43c495f1ba4b040f7a3d : SUCCESS in 7m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e7ac2b04476c4ad6b5a4de2a4c1a29b6 : SUCCESS in 9m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a7c8be539c2e4c00812d8105d8257867 : POST_FAILURE in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abf60085724c4b169d90364eebb1e46c : SUCCESS in 8m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4cd459f62eae4e7abb6c362ff60e1d20 : SUCCESS in 6m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3940706d70be4d888316309aadbb25ff : SUCCESS in 6m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0f5a6d2227cb45a2bc602b172481a995 : SUCCESS in 12m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f082211bf0324a38b31f82bb24a41d90 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6de2d51076db41a2a3c5d9005e89e3ac : SUCCESS in 6m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/03f36b09a42c41298615cf4f37e6b0cb : SUCCESS in 13m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/355a2b004f0944c3997858cc1ef82552 : SUCCESS in 14m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1d1d03cdcf994c9aa17fa026a24b56b6 : SUCCESS in 27m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/693840a7c8f74fe285d64af22a4f09b8 : SUCCESS in 25m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/813b95f53e664d56afdd7d4cf2c89aa4 : SUCCESS in 13m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84294331df4e4d33a9b42213dd7add9a : SUCCESS in 14m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/66a9824a3cb24f26be13ee16271840e8 : SUCCESS in 12m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c199cf36b168448da9e47bb714249ad2 : SUCCESS in 20m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/64dd06c7acf24a35bb609da44cb384a1 : SUCCESS in 4m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/75728c147261410d883574c2206d2be2 : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35c833c8f8054bdeb8e16b50c7c1c419 : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f9204afd7abf4f2c9178501d1ca0dd85 : SUCCESS in 13m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85d379129bc34d25981e93d0636a3eb3 : SUCCESS in 10m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61fd0fc629444820aaeec14e210f0acc : SUCCESS in 1h 04m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a5fa10b046f4909830358e3108ea3f4 : SUCCESS in 1h 45m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2a312f654b5646bdbadcb5595089d7e0 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8fc256cc16bb7077b1b6dfef02b0915b89539d20","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-14 22:03:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b481a92cbb76904d85fb18737726f1dd466267f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 00:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7288ff1cd88043ab9dacf52c5280c393 : SUCCESS in 2m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9708ea0596c44de6a77cbbdc930c6bf6 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60081e4bf8e41bc895f0f078967c006 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10fbc891224545589a88e29359207cd2 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64051dbe23784f61b2156afe0526a9d1 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c228f20899314a3995f182ce7028294e : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e57c2ec3da9490289e97a23d0926d86 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49f947fe93624c16bc96ce6c46a51989 : SUCCESS in 2m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/71b2ae3cc2b54637ae6d293ab39ebfc0 : SUCCESS in 7m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9429d15378147538e01ba4283f1b69c : SUCCESS in 6m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ecd637c4931e4f4d85fd074a01341c4e : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ba2992bb6e745d9b55c9551cf5cadaf : SUCCESS in 8m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e7c6391ce1be42d0b2b7be269c612ef7 : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e9b886f7ad8b44939abc48cfe681b058 : SUCCESS in 4m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fd0a4cd9de60461e9555a58a6b361228 : SUCCESS in 11m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0c8e70ad956b46e9bc996d608c045f94 : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13f82292c1a7482c84e4308231a5422f : SUCCESS in 6m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b4e6e99bee824dbf9a7737af8514c6ec : SUCCESS in 12m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f4d5f08a0c674c93963f6581dc49af0c : SUCCESS in 12m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/70092ab7da2d4b25bd9e401d94e9f876 : SUCCESS in 22m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe74976765d64574a002294b67210016 : SUCCESS in 21m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4fcfabd795e04c2594be4a98a7c24500 : SUCCESS in 13m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/215aec9d3bfe417aa4e5b2446650780e : SUCCESS in 12m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1a084f7142114519a55091762ab1946c : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d38ce64c8fe0487a93d0f129da9f1735 : SUCCESS in 22m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4d31c15a5e9e4d43a6f9b53e4b0687c2 : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b54d09943b5946baa542b6c2e044a82a : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5c2fe586cf04da98ce7f69e66d012e3 : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/923f8c0912a84914aac2a375bc977c77 : SUCCESS in 8m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aee28a951db2474dadff3dd3401e7e1d : SUCCESS in 10m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/323eebea0e744f2293e37020fdf94b0a : SUCCESS in 1h 02m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/87baf2d5afd74cbb9b09b8e365970908 : SUCCESS in 1h 55m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ddc40125cb64a8bab6959aed83a057f : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"625c075e9e7413c08f2cc286ac859e3599e8272e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-15 03:21:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70e3a0e412edd9cebf44528614878d0b82543299","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-27 22:01:34.000000000","message":"Patch Set 1:\n\nUsage example: https://review.opendev.org/#/c/694101/","accounts_in_message":[],"_revision_number":1},{"id":"b0e2dac5c15e6e3d763ac67a22bc0dd456d989d0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-27 22:11:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good to me.\n\nLooks like it\u0027s well used across a variety of python versions \\o/","accounts_in_message":[],"_revision_number":1},{"id":"9534b089151985d4c1ad7bc5f24ae18efe516d02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 22:11:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4057e367df07d5547e52dbc12e7106055532131c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:39:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5249522b4646d1b5919117f365daa8 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e432ef8af94d7486ca25cf4e0ba66c : SUCCESS in 2m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d7aa6d86bc54c58bd9ce2b658a000fe : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf3c87169b6a4e2797bc146e6ac3b1c6 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/219bd347fddd47fba2a870f8cbd69832 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8235d1029f6149b58a672d5498751554 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d70d829bb38b4850ad9a89c935b305d6 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c4cb2a5827754146b7c16db79454f741 : SUCCESS in 11m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c09eb04664449d8bedde2a42be76a28 : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/07bbad9b187643e59aaf07b117f2fa8d : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9286bdd4e63449fca0b0c0be493de4c2 : SUCCESS in 9m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d854219badbb475bbbb64246cead8aa9 : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/168ef69bfd284e169522172762f7caa3 : SUCCESS in 4m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/97fcf37441ef47a5a3eb12b083b2ede5 : SUCCESS in 9m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d073107e23344acf88226932f7c7d5cd : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7f8b59c85f049bc9da688fbc428d514 : SUCCESS in 7m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fea0baaf9d1b4ceb82c0f2e581c19beb : SUCCESS in 14m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f23dd22d715417792b6c175d33c3c5e : SUCCESS in 15m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/368f0204770c4f019eba5b2a14bf414d : SUCCESS in 25m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/56075bad82e443aa8637fce78883d70d : SUCCESS in 17m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8598799917f4431ea7d7567d9657a92e : SUCCESS in 15m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fa80b56870b74ef2923529d3701dc245 : SUCCESS in 12m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/02e4fd8df55545e89d147153782312c3 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/093eac71b60247b7b58ec25c84bc1628 : SUCCESS in 16m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cd10a1358f83490099eea1bf8a830fa4 : SUCCESS in 3m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5dbe113d7e7a4dcdae4dce6b9cf260d0 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9650e74fe974f8ebebd5559c98c51b3 : SUCCESS in 5m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a9eb25c45091471bb07577eaff99044f : SUCCESS in 8m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe9004eaa1b641548bfe81f9365a949e : SUCCESS in 9m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79db9fba5eda41909d276fe6246501a1 : SUCCESS in 1h 15m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d7ffd39b0a54fd581d4da8de9a2af6f : SUCCESS in 1h 18m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5634d88e8c4540d287ecbcab9bc0e7a9 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8a683869f993e9ba4aa5e69f23ceb38c451dbbac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:39:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"863b5b598030230b5eeeced26f0b5191a3a4f061","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 00:42:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930916be1ddc4d9e82ca7a2bf45cc9a3 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bbb17864bc64a0165be068e19189ea4a7b85f83","revisions":{"9bbb17864bc64a0165be068e19189ea4a7b85f83":{"kind":"REWORK","_number":1,"created":"2019-11-13 15:24:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/93/694093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/694093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/694093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/694093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/694093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/694093/1"}}},"commit":{"parents":[{"commit":"4f16a404a0f12768910d1c503191e3f2ff07b488","subject":"Merge \"Update check-requirements for py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f16a404a0f12768910d1c503191e3f2ff07b488"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-13 15:12:48.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-13 15:12:48.000000000","tz":-360},"subject":"Add new dependency for tabulate library","message":"Add new dependency for tabulate library\n\n* Is the library actively maintained?\n\nThere have been three releases so far in 2019 and the last commit to the\nrepo was 2 days ago. Issues appear to be addressed at a reasonable rate.\n\n* Is the library good code?\n\nCode appears to be fine, and there are unit tests tox targets for all of\nour supported Python versions.\n\n* Is the library python 3 compatible?\n\nYes, package metadata declares 2.7, 3.5, 3.6, and 3.7. Tox also has\ntargets for all of these versions. Contains compatibility code for the\nplaces where it needs to handle things differently between versions.\n\n* Is the library license compatible?\n\nYes, library is under the MIT permissive license.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nFedora has python3-tabulate. RHEL has python[2|34|36]-tabulate. Ubuntu\nhas python2-tabulate and python3-tabulate.\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nYes, this would be an alternative to PrettyTable. PrettyTable has not\nhad a release since 2013 and is starting to emit DeprecationWarnings due\nto Python changes. Various attempts have been made to get a fork going,\nbut it doesn\u0027t appear any of them have been very successful.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nThere are quite a few OpenStack projects using PrettyTable that would\n(and probably should) switch over to using tabulate. Some existing\nprojects using PrettyTable that would be candidates for switching are\ncliff, cinder, election, glance, and many more.\n\n* If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nN/A\n\nChange-Id: I80eb63c2ed01659990ca5506be7e0e246c9586ba\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bbb17864bc64a0165be068e19189ea4a7b85f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bbb17864bc64a0165be068e19189ea4a7b85f83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
