)]}'
{"id":"openstack%2Frequirements~694198","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ife0dd98d26447603e344a6ef08495bd1ee3adafe","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ife0dd98d26447603e344a6ef08495bd1ee3adafe","subject":"update constraint for oslo.utils to new release 3.41.3","status":"MERGED","created":"2019-11-14 04:42:50.000000000","updated":"2019-11-15 03:10:20.000000000","submitted":"2019-11-15 03:10:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694198-1573787420365-e0020214","meta_rev_id":"743e1e4b3b866ce28480e0b7ca94d0e299e73152","_number":694198,"virtual_id_number":694198,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 03:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-14 23:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-14 23:44:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 23:44:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-15 03:10:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d4011c8c2b346c502f754839c8cd1d8e0278585","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-14 04:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c492b26006a83086378680b94c746c4be795b82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 06:34:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0ef67803d3c84cf0a070f42071a2faf9 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dac50810b7142e290125194ef78b618 : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/650901cb39ef43fb94800aae458cd53c : SUCCESS in 9m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f3a4d181b9a4da499768eb0009dbbf8 : SUCCESS in 10m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11046853e3c14039a36969ea70b48c0f : SUCCESS in 12m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94449011985d4ce6a039827da02dcb9d : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ace10c266aa410ab21d6cf34255aa6f : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6d258a3fee1c4410945f6b5eed248aef : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d97d5a5157994e40beac4c1cd86fe15d : SUCCESS in 7m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b115f17baae14eeb84e28054c98b79e4 : FAILURE in 3m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f2b54f93d4934d83be20fbff6e3dc88a : SUCCESS in 14m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b5390b51fe1f4ad7b14d709c170b6f17 : SUCCESS in 5m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8f0571f81fbc44b9bc4d62623f59cccf : SUCCESS in 11m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1882c447ec5d4b348da5890da9363ee3 : SUCCESS in 8m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3152889a3e0b4b44869302ed109319d4 : SUCCESS in 7m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d0201ce6fe84c4180f3d50d94793b60 : SUCCESS in 6m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4499a8430ff840648e8a2fd7e1949a01 : SUCCESS in 17m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb5e2a1b058141b09ae8e4780304be96 : SUCCESS in 19m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0cad1f37d2dc48939ebda5750286e0d1 : SUCCESS in 21m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64aaf208b1b2429aaac0beba5ab41e0c : SUCCESS in 25m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4b0651f4ad67407b9e49c1f90c0342bb : SUCCESS in 12m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9155c41353ca4f4f9b38c8aa24ab13c6 : SUCCESS in 17m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/63cd825a22d444cfa85e3f797651bccd : SUCCESS in 14m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ba81272170249ecbb19e9bf3dc6329e : SUCCESS in 23m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e91d477fc70a4c07a37240c58e8f149e : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1e25daa6f1ea4899bd96c56e40fa2f6d : SUCCESS in 11m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/01839b27e4ac42c49e48830b7904931e : SUCCESS in 9m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a0573aab07d2407cafa5f9cc9744a884 : FAILURE in 31m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7acd8f3714ce4efdbbeeb1d13087f59e : SUCCESS in 1h 08m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/82c2323fd1314fc3bd5f0c40987f508f : FAILURE in 1h 39m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/96c86e4bd8ef403da4c7bdd80e6d7116 : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a3276c2bffacb27999493bfc6bd1322bcacf3923","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-14 22:03:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88ff16623da330123d7e6f175a9de070ef94e775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 23:43:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4077226c30314452b11197c8812800d3 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf71e4eb3a724c0d91daf79f5af07e69 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3182cdfc5ed74c3d8f7edb2df936c1f4 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2733204fc6b44e4e991cee3163eaa16d : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc900bb0c8474c82b4681e78d9053b8a : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec761977d96e4bfa9ebdc9b0c27fe3b4 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dea565b16fce4a77ac893fa8ccca9dc0 : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ac7f3ebab21847969096c29eb5586efc : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca43a92b546847e98bb7d1bac840e489 : SUCCESS in 8m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9b08917144b8489e9f3c272e0351eb27 : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fcb6d2ce02a74dcab30e7fff21ef67cd : SUCCESS in 10m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/da594b106ecc457b9ad22e215866757a : SUCCESS in 4m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67be35196b8444a3b479b970018c500a : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/761b4ecfa5a64d548ae11d2ac6564dfd : SUCCESS in 11m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/86edf791fab540c0ad32e409aa682f25 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8548b5016a064af6991f8a5b861a3363 : SUCCESS in 7m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dcc7fb5d3a9c4a10bdad6f7175bdf91f : SUCCESS in 20m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bab6eb6b1a3546a490c1e116405e50eb : SUCCESS in 17m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2639e3ba376b44f8a7342b113dbf16fc : SUCCESS in 23m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/071923997020411da758915e4220054a : SUCCESS in 23m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/06538a90cb48488e8dca31b6ae15b346 : SUCCESS in 13m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0a9ff0e206347c5a34a79a5c58c87bb : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4c0f071e8931438a9745a02b7617818e : SUCCESS in 11m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1f167b9aa914c819a944f8f44b6109d : SUCCESS in 18m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5284244acdff4c968996c2219f8b3e54 : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f28b5f407b4243a4b10fa5fea82064a2 : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba31a925cc644097829f835a04d44ab4 : SUCCESS in 10m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b59ee82327b9409ca3059b83bd11d80b : SUCCESS in 1h 06m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27f3c64c22254d489b824bdb99d4397f : SUCCESS in 1h 33m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0959dbd8a1b84e87bf1834edb8ea060b : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"66db9184228e737807824e4ef8f0d06d9345d5ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-14 23:44:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac9ddb482b9f42c1a7cb5979b3448b297f2cdb75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 23:45:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b19d04392b5e277cd59247487604d11809a4a31c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 03:10:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178e7fd7c4b14f2ea2702ea27cc915e2 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2c40c39bf99405aab2ca56e937ba1a2 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f1ed8944959471ebdb90f434e711b3e : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bdce34accfe4622bef74d67b501f93a : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89a701085dcd4312bd4dd6739295fb52 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a56d61c35584d448984e2f3766531c0 : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/39a9f80649d344d888b49d4926afed5d : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac91940c829043df94dc60d7c1b12da3 : SUCCESS in 7m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c8a1d3523ddd461ba07f23070c5fddd7 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7a1fc166f8c45ba947683313f1b86a8 : SUCCESS in 9m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/60b62da3b60b4545bf777a1c5dc41838 : SUCCESS in 6m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b57a086d86114e76a604ec3d44a5b52b : SUCCESS in 5m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9ab6ac019d9c4a119722482135df5bfc : SUCCESS in 11m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f2f22235fbe64c2690a75318b4baf6ba : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ad9803afeaa44f49152cd9f20b53a7b : SUCCESS in 7m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d6c22306e0684e7795af674f55d8e301 : SUCCESS in 13m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c862e609fe6448f3a0c5e0b033ce45a2 : SUCCESS in 16m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7a1bdb7273bf4fc4bfa570a7eaee00fc : SUCCESS in 21m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3e8079a893cc4945bda47c6ec9480a22 : SUCCESS in 22m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/069886181f404adab321e858607a9333 : SUCCESS in 13m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/23b095a2e7fd4a79bde1766f01a915ed : SUCCESS in 12m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/279cbdaab4904f4a9782adfaf12adcbb : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30c1a5eb06cd4f4782d30980934c392f : SUCCESS in 18m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbfbf2284f1d433d91a116ebfcea9c0e : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/62e3aa90f640480f80662a2ec037c7ad : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b7462987cc5a4304bc5a8a98abfcad6d : SUCCESS in 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2e7e1d0172b44a3b8c449b33c546303 : SUCCESS in 1h 05m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1fa6ab0020474c21b82a13be3c870405 : SUCCESS in 1h 32m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e4690b8e20d4ab69818d03336c50a0e : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0022700131b3d96b48c7c63b751cae6330d86ed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 03:10:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37b4e18d038b6e59979a50b12dd7927eef47e2b9","revisions":{"37b4e18d038b6e59979a50b12dd7927eef47e2b9":{"kind":"REWORK","_number":1,"created":"2019-11-14 04:42:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/694198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/694198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/694198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/694198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/694198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/694198/1"}}},"commit":{"parents":[{"commit":"17d99de9237dded3b27b25f1844d7b8276f12087","subject":"Exclude sushy 1.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17d99de9237dded3b27b25f1844d7b8276f12087"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-14 04:42:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-14 04:42:48.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.3","message":"update constraint for oslo.utils to new release 3.41.3\n\nChange-Id: Ife0dd98d26447603e344a6ef08495bd1ee3adafe\nmeta:version: 3.41.3\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Iba1ba1ca02b68247728d8dbb78c7e472e4cd12fd\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: yanghongyu \u003cyang.hongyu@99cloud.net\u003e\nmeta:release:Code-Review+1: Andriy Shevchenko \u003chuang.zhiping@99cloud.net\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37b4e18d038b6e59979a50b12dd7927eef47e2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37b4e18d038b6e59979a50b12dd7927eef47e2b9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
