)]}'
{"id":"openstack%2Frequirements~694248","triplet_id":"openstack%2Frequirements~master~If5de7301d998bc0ec5034afd853426ae393d1064","project":"openstack/requirements","branch":"master","topic":"py3","hashtags":[],"change_id":"If5de7301d998bc0ec5034afd853426ae393d1064","subject":"Requirements-check fails","status":"MERGED","created":"2019-11-14 10:58:51.000000000","updated":"2019-11-14 21:58:35.000000000","submitted":"2019-11-14 21:57:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694248-1573768649757-10f32a19","meta_rev_id":"b290bcb1e802e3917fd24f87f500b8d3594a8900","_number":694248,"virtual_id_number":694248,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-11-14 12:50:21.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 21:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-14 12:52:15.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-14 14:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-11-14 16:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-11-14 13:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-11-14 14:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-11-14 16:09:00.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-11-14 15:25:08.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 12:50:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-14 13:54:29.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-11-14 14:22:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-14 15:25:08.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-14 16:09:00.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-11-14 16:16:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-14 21:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3887f2c623f460f65cac7dbbc0e85e4fc2744229","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 10:58:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0acefe58517d36348d0691dd0d91490c1be1030c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-14 12:49:11.000000000","message":"Patch Set 1:\n\nis this tested with a depends-on?","accounts_in_message":[],"_revision_number":1},{"id":"72696a02b749f366f9d00997fb70a15347e64bb6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-14 12:50:21.000000000","message":"Patch Set 1:\n\nI\u0027M asking since the jobs configured for this change will not test the changed playbook and are therefore no indication of whether this works.","accounts_in_message":[],"_revision_number":1},{"id":"cb9e51955e2e9e2533a218cfa7c0a9b0552fe81c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 12:52:15.000000000","message":"Patch Set 1:\n\n\u003e I\u0027M asking since the jobs configured for this change will not test\n \u003e the changed playbook and are therefore no indication of whether\n \u003e this works.\n\nYes here https://review.opendev.org/#/c/694233/","accounts_in_message":[],"_revision_number":1},{"id":"efd395ad67a24b0ad806efeca176cee60f430431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 13:01:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9120ca73af44a39946dc208ad15a9f : SUCCESS in 4m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/337f06d47d244bfea38f11e42e5bd633 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b0cdf1621340ef94aff81a096d63bc : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7601cded83b2446da68c3f2533f3df60 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74325bfa724749bdb29f93773cff07ab : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2dd2742c25f4e699dee3fee1c56460f : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c80e432dd84474e9641e09c8f1e8c13 : SUCCESS in 5m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af6c4b385b97409081785a82accda1a1 : SUCCESS in 1h 06m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ead9547a86de4d9da9151291c8c9d102 : SUCCESS in 1h 59m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b6c8ae6e9944f5c80129767803e6d07 : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"59a77ee90471503b56665e5e695ee325f8c3ffa7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-14 13:54:29.000000000","message":"Patch Set 1: Code-Review+1\n\nyes, please","accounts_in_message":[],"_revision_number":1},{"id":"de4eef477b6da7b42defec6e43cebffd160465b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-14 14:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis last change was partly to fix an issue with the job trying to install with pip3 and pip3 not being present. That was done from within the script though. Maybe we need both to be safe.\n\nWe need to get this fixed though, and the current pip usage obviously isn\u0027t working right, so since this is fairly trivial and is blocking a lot of folks, I\u0027m going to push this through. We can always revert if necessary.","accounts_in_message":[],"_revision_number":1},{"id":"3a72894eb39c9a8247b9901bf109ece66803fc40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:23:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2682b04987c99c67818d03c3e81fa231931a5364","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-14 15:28:18.000000000","message":"Abandoned\n\nShort circuiting gate failures.","accounts_in_message":[],"_revision_number":1},{"id":"dba1323979be77ce87db4501a90559d9bf9f06a9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-14 15:28:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"cfdc8b8aa3f682cbe2afcfb8e34f3b8096f90269","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-14 16:16:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"061842a3fb4faf47e702d2d7458e7df258c2e8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:37:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d86ec4624249f5bb57bcef8705944a : SUCCESS in 5m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/432e367b0eb74368ba4ece3341e776eb : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b752cf0694e43638f3428de5ea9b755 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9fea25bced44770a5e0b173d3d9baeb : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ab8f9917b0b4973addae208d08663ec : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/484971c783d24aa0b7a141a40d5f57c3 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/509d18541ff84b68b8c158bd4552527a : SUCCESS in 4m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/890b527e984b45808db9e502ecb85b95 : SUCCESS in 1h 11m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db23381f5f3244ff9a33ed7f4f3c99f6 : TIMED_OUT in 2h 06m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f9838276c12e4dfb9f5bc8d328a49ef0 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":1},{"id":"44ebb990371e9ea7d2f8d7614c5d4ee93a9b974e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-14 17:40:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c050c0884cb6653cc52d3c1944aec697e8d546c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31829fd928b4485834d53012ba1b841 : SUCCESS in 4m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/80272eaa55264eb383acfc44bb2dd903 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488d2cd6be0f48889963c007538d41ef : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e6765e11d8d46ed9fcf95f7e6a3b998 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e93d0df1a8134646965200a20f071c8b : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/343c6dae9ba14ca581b28c6584bc10d1 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37a4e96ed9d34571950843f4cb3223bb : SUCCESS in 3m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af4694d078ea4ef9a47f46c20a6fe656 : SUCCESS in 1h 15m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ae07102e5a2d4914b230947b12e62527 : SUCCESS in 1h 36m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/315fb700eae14482b61a596fdb2de53d : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"855da7c2896114e927df589d04301b46977802e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:58:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95baa1e72f99344d8efb9044d9618022fe407c5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:57:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0276cca474874f889f1615225621dd9a : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f20ba360b654251b710d1e6c5ee912a : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60feb37750ae484b98717f1433e18722 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d7c122c91024c1cb4da4b72009e69e1 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07f364ddb01e4199a1a4064bee67a394 : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2355bcf6fbaf41228de8d0be1210b771 : SUCCESS in 4m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb8aa619b6d64b2a847068a51d828b12 : SUCCESS in 59m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f44666e4e232450d89369afa675c0985 : SUCCESS in 1h 56m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8a5f33083aa44e71a728beab7d4a88cd : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c9f0ec218f6c077b6765f5037fd9b7c812d0d0e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:57:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fbbb5b01d796b5c27d1370f81fb7baf539459ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:58:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bc1c11972343aaa84580b3f807e898 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c270c8dd53f2aa37d93e21d353f885682cdcf08","revisions":{"1c270c8dd53f2aa37d93e21d353f885682cdcf08":{"kind":"REWORK","_number":1,"created":"2019-11-14 10:58:51.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/48/694248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/694248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/694248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/694248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/694248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/694248/1"}}},"commit":{"parents":[{"commit":"5620d2c1f9abc6c278c13f3ef867e4136923efbc","subject":"Merge \"Requirements-check: use local repo like any other library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5620d2c1f9abc6c278c13f3ef867e4136923efbc"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 10:58:33.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 10:58:33.000000000","tz":120},"subject":"Requirements-check fails","message":"Requirements-check fails\n\nopenstack_requirements is installed in python2\nproject-requirements-change.py is run in python3\n\nsee:\nhttps://openstack.fortnebula.com:13808/v1/AUTH_e8fd161dc34c421a979a9e6421f823e9/zuul_opendev_logs_91b/694233/1/check/requirements-check/91b72b0/job-output.txt\n\nChange-Id: If5de7301d998bc0ec5034afd853426ae393d1064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c270c8dd53f2aa37d93e21d353f885682cdcf08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c270c8dd53f2aa37d93e21d353f885682cdcf08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
