)]}'
{"id":"openstack%2Frequirements~694716","triplet_id":"openstack%2Frequirements~master~I78cc2a10c2008302a250cf85ce6fddeda1739ba3","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I78cc2a10c2008302a250cf85ce6fddeda1739ba3","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-18 06:15:53.000000000","updated":"2019-11-22 23:18:32.000000000","submitted":"2019-11-22 23:17:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694716-1574464640125-a496186d","meta_rev_id":"74df8247d1ba335c4304859b95d9efd7108e9475","_number":694716,"virtual_id_number":694716,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 23:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-22 19:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-22 19:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 19:30:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-22 23:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0471c4252d877628431ebd9c567dd4f5bcfe118","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-18 06:15:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b27fb7aed7bdb9a2cb7165ef8cae94fa1a51f609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 08:42:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9342d42712654772a9593c855f4779d0 : SUCCESS in 3m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/16aa848f94ab4dc2a8c1fb62e931c624 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf230ae783444d639760ae4b65413da5 : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d327bdd2bb24863b46aeead814fa078 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43791b9464aa4fcc85e6ead02fa5d06b : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/053ebac7c5464e0cb18a3d51c99a50e6 : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59278b64bccb4253ab1bf0d9d50556f9 : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac2920d3a32f444a98d2d9d508c2c638 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f8dd3e2e5e6c4668be925477c1a4b9dc : FAILURE in 7m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cce15339b45749fd813439de0cba08d8 : FAILURE in 15m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c5de9145b2ed4d1f9a210c360319d7f2 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/536c1a22ee9048a0bccf8256d446df27 : FAILURE in 8m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/433a16f81e1e451abce2179094ce4ef5 : SUCCESS in 4m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1fe6a0b335d240288007ca797ce97085 : FAILURE in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/efd07ed14cfe420e841e5021fb0032c7 : SUCCESS in 9m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/05279ca309e348dc89dc3f29cee188a4 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98db65b05dc04cca993f807c1a00af48 : SUCCESS in 9m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/efa36a04836544af9b5ca2e7041f8f01 : SUCCESS in 12m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a26573d5f34f411d86b3d373871803da : SUCCESS in 16m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5581ef7e38d04dc885b61ae6ee021a45 : SUCCESS in 24m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e7c1547e31bf4b2799aff78a23ac09da : SUCCESS in 23m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4099a41cbb5945f0b61e8449406ee1fc : SUCCESS in 14m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/143c6df6f4a0418b8cf8f26f5606f104 : FAILURE in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/546cc43b3c6e46eea54e6e532b2f5681 : FAILURE in 13m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67cc150486464c859485e8dce7dfe8dd : SUCCESS in 25m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/29ca42a31816415293fbc2b67c3f6a20 : FAILURE in 6m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/82162fed8bf448da98da1243a3e208c6 : FAILURE in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74f183593b68401296f22542606fd37c : SUCCESS in 5m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1015ec61f2f742618a86e4e87826dac6 : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a6ae8a0dd7194953a61b4d7c4ab2859b : SUCCESS in 9m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4df9477a503142e2b45710d67fd2340a : FAILURE in 1h 05m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/119f906612014199ac2940d2adf7a662 : FAILURE in 32m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f709c83928cf40c984016d358557b5f8 : FAILURE in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8844b50801822184be5139be5696a809246173a6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-19 06:10:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7f8d344077e03daf0363365044cd5393cf698b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:30:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cfece4837e454abe79fcb5b427e48c : SUCCESS in 5m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7a62f3d2d1d74f8cbdc524550557de19 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/757646f2f0724b9b85794467c4e34690 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d1d455d598b47c0af99098336550ef2 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92eb135d36b5440bb907f66770378d7c : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62ce5bb6d41943d8adb91c1c96c7e9d5 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79a97c03b90b4530a792f8aed5d5a1c6 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf2baa745ee5414192f6a916cb68cabc : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/911840c59aa843468d7f8469fcfa198f : FAILURE in 9m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a7a34dfcf40649cf9c33d854cfccd8a6 : FAILURE in 11m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/791990d930bf4d988d08df4a9a149336 : SUCCESS in 10m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa40c1457ce14c9e988e103db5f7f3f2 : FAILURE in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0084bddbfd2e497e88f1212455bb65da : SUCCESS in 4m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb192f0a3fcb44d4a5581bd36ed3e063 : FAILURE in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1f99c126691647cfbb63be93e6656a46 : SUCCESS in 9m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9af571d76c044c8aa44d215896462d90 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c103dd1e0ce4462e85a0c094b6c27fdf : SUCCESS in 6m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/565c6048355d4d79818ce8e5aa31460f : SUCCESS in 12m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06c5328810c14f46ad62fa65a812c28f : SUCCESS in 13m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/73620978b4924abfb4e603f86c442283 : SUCCESS in 24m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/334bec8bce194dbda24edb2ec4df1b26 : SUCCESS in 21m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e5430a5f96d54c56a613c2d361276bda : FAILURE in 5m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1e5a1c1883f4bdeb01fecc41392014f : FAILURE in 14m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/90b9361db3c74ef6a6bfffe1024d5350 : FAILURE in 13m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1410f9afd724646b50e7c198e5cf844 : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d8705adf39fb4234a00d31fcf03e7d14 : FAILURE in 3m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/09c54360bc0347e2983df0f03f01d549 : FAILURE in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9768d517b5fc45688963ca0421f15b2d : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e0dae0befddc4744976fd621f1a11732 : SUCCESS in 9m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2515fa5b0d1349afaa035dda3c89a21d : SUCCESS in 9m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ff9667831714175b9436d3adf7ef6d0 : FAILURE in 1h 11m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb860bbfddff4f0a8142353aca275355 : FAILURE in 17m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe65fd68c272475e9ff416692bae24ff : FAILURE in 46m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4daaca7427d2fed4b33fc426256ece250acf5ad3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-20 08:03:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7590c8df3764391f5093c1db172c38cb59409d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 09:20:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc47432489004eadbc87d2257e8e705d : SUCCESS in 3m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b6117377873b490592e3db810ab254d3 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e49078b6de634bc2be80a2c4c94c5a94 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/947353d65e8e493c998e1a60e1d773f7 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d2f0dfc50b48f98e4f13085ef7d407 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23033586e8b44441a28271845e777bd3 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b32ad1d944e04b359823837cb1a3e75a : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e5aef0bbe9a44d7bf72d8b8b0c8d80b : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/75317fa431b24595b42af795be59cf7e : FAILURE in 12m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7ab0a47a6014287ba5e8cc153c37129 : FAILURE in 12m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5f443cd3950f41639d9e2760ca481820 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea1f84bdb8db4f97a921c7940ac66caa : FAILURE in 8m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7a135a5a0fbf4e91a643da2f3df402f1 : SUCCESS in 4m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b655734863d241a3a966bde3f8fc5581 : FAILURE in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/365db0c431724905a2aa41020a64f888 : SUCCESS in 10m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce92acd1b77040e0974fd42e81a9e7fa : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77f1a633475b4be794c1148ea44ff92d : SUCCESS in 7m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/73c86fda3e7149ceb7be37a0be4f1c8c : SUCCESS in 15m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3f0ab8d4a2aa42c4a98862cda9950782 : SUCCESS in 14m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0e7e8dfd410c4c04ba8377c970226d60 : SUCCESS in 22m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9fdb95ff71624f55927fa20e6f7997e7 : SUCCESS in 23m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bc2b48ccb3f94981aa06c8f089403e64 : SUCCESS in 13m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f467530cf874843b3f2481e5dedc8da : FAILURE in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cc5975b2425c45aa88f92fe7bb814343 : FAILURE in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c58b159974d46b1b475d255f864b14d : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/442f26e73e0548b99d1af6d259cbf5a7 : FAILURE in 4m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/05c0870b33954bb591a6f2f01298f555 : FAILURE in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58e6402925a54256bff4243b5fb175f6 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cea1e893de6846e6acaa3ef9afdc5b1e : SUCCESS in 8m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/270787dbec4942e4b258e493baf468ec : SUCCESS in 9m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/65d50d5d35474085848d1baf56b33579 : FAILURE in 1h 10m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf3861768c6d46a091b811ecfc47b7ef : FAILURE in 32m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8c0ecc53e0114ba3a63660e819876ced : FAILURE in 57m 56s","accounts_in_message":[],"_revision_number":3},{"id":"d7811fba412912e9f3a88fb4f8ced8723d3b50ab","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-21 06:28:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5b1e23b82b9c87fb68f83f04821b6be74d40dd9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 08:05:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9108ac533c604052b99be9bf77395393 : SUCCESS in 5m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9698be3e373340bb8c638dde919c3d61 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2fb1b7a2583406ea02b68f30b512002 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26923f7716d340bfa0f961a4ac994723 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac3e1a80d0e492bbec167528ffba84b : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/411a8b5098944092b33656e9e6a68b66 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e28e25b3813f40bab7708ced191b3945 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3973cc4f8e6e40b5973ff498d687ec4e : SUCCESS in 5m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3989fc86e6524cf9b48088816091d9ce : FAILURE in 13m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/360e863760614592a0790ce11b99dbd6 : FAILURE in 10m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6a2ab90d06c241488de5c959500b960c : SUCCESS in 10m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52570120b97242309fdb5b77c1e42af5 : FAILURE in 10m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b172a1bae7fa4d798d133173326baae9 : SUCCESS in 5m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6c1a91748085491299e60f15eb7c7a27 : FAILURE in 6m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bdd44d3683084605a48087d2ce28b020 : SUCCESS in 10m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f741647246c4abea4e86e2b887b659d : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e667186e67143078b0619361df15922 : SUCCESS in 13m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eab1f30115e0450694d420d85e8d87a9 : SUCCESS in 16m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2ef0150427544b16bebff4b6de731f25 : SUCCESS in 15m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5c65e746235641c2932e30a9afee5a46 : SUCCESS in 24m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66200ea268ab4617af312f1e59aecc3b : SUCCESS in 21m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b415665699f944db9ed588124ca76917 : SUCCESS in 15m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7080f4a5d3574210b1122ddba1b8d3f2 : FAILURE in 15m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e43553e6ada148ce9fffa9d6876bdc6f : FAILURE in 16m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e79dbf9c446644f388f540f8844743b0 : SUCCESS in 15m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f8b03e1965b648ba869aad8df052bbfd : FAILURE in 5m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4109ccd2cc3f4525a50e2d06a39b5263 : FAILURE in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42a82ab64b55423fb6f66f0fc585a981 : SUCCESS in 6m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4ca1713f38d94882a7f030066ac935a9 : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9a0e6eddad2545e888363039ee794b5b : SUCCESS in 13m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/000e4e52ccb145a8a5bb32483fef9733 : FAILURE in 1h 04m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ea0e6820ed744f21b626abf53df153b7 : FAILURE in 1h 26m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e8bbbf56de144c19ac80a521deca0389 : FAILURE in 59m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e61154f2a1b4d37632e6064f0ed7c1951c4ed78b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-22 06:35:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"efcc69498ff17608b7d73723a4ae1100be65b342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 08:30:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca0df738cb8492fbee3e83408413058 : SUCCESS in 4m 25s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6547a44b9c104b6e82787780babf717c : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/357de702f6454919b7280e9515d7c7fb : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75e2d149e0fc426b857103a7f5f37ef9 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/783dfeb2bc2b4608b8d0b2c891d2d595 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7aa433c43394c5bb5f97fc2ce0ef59d : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44c6822d508a4eeaa5b01e9a2af3930b : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4b3080556594d88af267937611217bc : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c6a9f38785fb4a1ab5a9cecac80efe50 : FAILURE in 11m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9f0944d7705f4ed89dcc7dd694cb21f4 : FAILURE in 13m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c8cdc09e7a7e497a87c56671c4667407 : SUCCESS in 8m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7fbfcc64de654aa582783b88880c1cc2 : FAILURE in 7m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4d97c933450e414da3d4fbcac10f4420 : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1bf070d5eb6b4e54bafd6e13fc592f14 : FAILURE in 5m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3521e0d3bcdf4d46ad7527ea8a24507a : SUCCESS in 9m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1ff32d29f044c1a9e2e742bf4ef1891 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/056dfd1616b54f0899d0718fa1547f75 : SUCCESS in 7m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1d78608339b442f59d03e1052dd77931 : SUCCESS in 14m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a0459a6ec598449687c05cb6698b9d1f : SUCCESS in 12m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8270abdc4f354350ab3f44d0eb691381 : SUCCESS in 24m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fdefcd1e2db46b687cce1b6ab4e6716 : SUCCESS in 21m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a4c2ae7695034a40b890fd1bb7139b0c : SUCCESS in 14m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/19c0572fbc50424aaf904a14e3005072 : FAILURE in 14m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/18dc53e7951f4bc5bd7bf4a41695de08 : FAILURE in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/209d5b1dc17a49abb7900cfa53c39f6c : SUCCESS in 19m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cbe8b4d5a131425ca185a2ff9884dead : FAILURE in 3m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/86379e204327433687cb2f3b61144ce6 : FAILURE in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd880407c9f84b8bbd0dc2f105db9cab : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f70bce0c6a54a7eb210d1e7bf40e819 : SUCCESS in 9m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65b796310834499e9ad7a2e6dc8d3a61 : SUCCESS in 9m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f982b466465b4be7b148213c3958f999 : FAILURE in 1h 20m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9a5a6a823c20457f9e5d52d675b10ac3 : FAILURE in 1h 02m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/befe59cee937405581ac383fd2531a0a : FAILURE in 48m 52s","accounts_in_message":[],"_revision_number":5},{"id":"8f9805ee2c4a656e40bb329a8c6752edda888c2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-22 16:12:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1aa251013baea8f686f9d4ed0cd2890926cd6d04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-22 16:12:53.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":6},{"id":"f4517641bc15162398660f025b935a6656ee38d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 17:58:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca53c6c4e4d04c28b9dd9f468a791e3a : SUCCESS in 4m 36s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cf4ffe63d41144369ab89d1a79b74c25 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c62043b6aee4da7bed90494c15b6d0a : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca9a0ce201f346cbb7b7c9daded7b42f : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b1cc00cf38b4935a7e871cc052a03c7 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/859699412072466ea92962b4cf52c04f : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d38c18dc6f6146bcbce31a9e9bdf5b02 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/741cb6e3a9d84c2c81761f9cfea25326 : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a9c17900224540168bfe9ef1a5a69178 : SUCCESS in 12m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87286aff95d3468f87ae54f1d2b1fd3c : SUCCESS in 12m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a84d71f1710d4ec99c6a95205b1e6070 : SUCCESS in 8m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d8b1ba6bef0464aba268ce891cf11b3 : SUCCESS in 9m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cca75a62effc4e4eb8cacd8c9e80b924 : SUCCESS in 5m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3fa9c6d7ea6d4a809ff0dd7d346f9945 : SUCCESS in 4m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a79e34b2ec26471982b071e1a904515d : SUCCESS in 9m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/278c2391962f45d8896231b651dee334 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bfa541ebe4a4b3180c39df673021727 : SUCCESS in 13m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e0b7f6996042401e9ed77b45921979d8 : SUCCESS in 13m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a62686aa16d54fc3b890f73bfbacd01e : SUCCESS in 11m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a71f9fb7d2744dfabf7ce5b0e612ab53 : SUCCESS in 25m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bd7c9c6406ae4805ad6e817f8a2c542c : SUCCESS in 23m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c9013235374e46658326cd0599feb1dc : SUCCESS in 19m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae1014549eca413a9f5fcc39898fb613 : SUCCESS in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/23fdc5c3696a414fb2b712627b69af92 : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68e1916eeeca428d98018b7f68417cfa : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/04f174aab6b94c48b9f4f10be46ff497 : SUCCESS in 4m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/461f0a89cba144eb8b968b86797adccd : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30658deb57304b3a90841cad90111463 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/083fea8954d34797a91da4683817e1b7 : SUCCESS in 11m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f2ccf03b9e794b3fbd36cde2333fca5a : SUCCESS in 10m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/01bb1f08d57943faa4817cbff515f7ce : SUCCESS in 1h 15m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f49c21ac45340f3b79c3dfbcc93fd9f : SUCCESS in 1h 26m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ab8af2e1f174af7921b8ce94bb6d71f : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":6},{"id":"db24b9ee0576da73058a375f41df7b9212ffc3ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-22 19:30:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7188eeefa507373f6fda53e9d7dd2dbaa9d06aad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 19:55:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c500636b02228febdd606d55e54297524f28ea06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 23:17:18.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b868cc81c0940888882324e05b58dd0 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619b06d74d72498ebe20915bb8ff972f : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7928c704b1a41fcbbfcaa860281c753 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b880b690c704061b9b8748a741d6665 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e121ffddd8a34293bf60e5a33f4e9bac : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f04aa316be9f48d8b5c8174d22b40102 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce907ed48f0940e4b68c667051a845e0 : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ba8e372ddef441dab40a440aa0b94686 : SUCCESS in 12m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e517894badc45ac9fc2263af411f690 : SUCCESS in 15m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f919ccc9ace14fe3b9166571e363770b : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3d194310b093455dbc6cc18dd707f621 : SUCCESS in 9m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e887b95cdc534a53bff7441d7c7f933c : SUCCESS in 4m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/782ae4a9a70e4ddd8533d4b6dd8a1e76 : SUCCESS in 4m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7ae663f9c6794b03841d61a09acd3733 : SUCCESS in 10m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa4d81f4492945e98101b1d56d5a9562 : SUCCESS in 7m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e6c9309e1e44072b4bcd49f1ec05771 : SUCCESS in 6m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1b78abe60d854b3f9f1ed8aa10bd4c33 : SUCCESS in 11m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/48b107afe1424290ad4d39bcc20e9981 : SUCCESS in 20m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/784f9bf47fc14849b7d8d5013a9c91ae : SUCCESS in 23m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ce1a31c796749cea73ce710109237c3 : SUCCESS in 22m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d556195f4c743188bda1ef3a7b25969 : SUCCESS in 14m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/04888b4d60f943ad884070c72498ba14 : SUCCESS in 11m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7b9c758f97cb4c5ca4510005369221ea : SUCCESS in 14m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81c68d1737564cb89c112d4f3ec4894f : SUCCESS in 18m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e2827dba03c249acbb07d8dc43864ed3 : SUCCESS in 3m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/29683666fc9240f2a64bf92f811613fe : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7761e0642a8e4117b9b459e965187a82 : SUCCESS in 6m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3b7db75b268c4478a95eaf3ffd329921 : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ca20603c3fc9431eac5a156df33b9a3a : SUCCESS in 10m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58c24bbfaa784a8e88f0c04d600f7643 : SUCCESS in 1h 18m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d3a50e624214b93bef094dfe85836ca : SUCCESS in 1h 16m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6fdc1ddc09d49bb9789ce91187b43a1 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":6},{"id":"58527ecdbba7e0cf5483177a5429839ee50db4c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 23:17:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"74df8247d1ba335c4304859b95d9efd7108e9475","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 23:18:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6219b134ac9a49a0900e020fde58f957 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"81b8bf06674d0e4ab04980bc3011b59d713910f6","revisions":{"b2e29b3145607915f873d859b2fbfad91fccd000":{"kind":"REWORK","_number":1,"created":"2019-11-18 06:15:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/694716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/694716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/694716/1"}}},"commit":{"parents":[{"commit":"fc12b9da7985936b03686de43985389a22c65cd7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-18 06:15:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-18 06:15:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I78cc2a10c2008302a250cf85ce6fddeda1739ba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2e29b3145607915f873d859b2fbfad91fccd000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2e29b3145607915f873d859b2fbfad91fccd000"}]},"branch":"refs/heads/master"},"bb1f76b5d739cec228d9982df03401afa86fc2fa":{"kind":"REWORK","_number":2,"created":"2019-11-19 06:10:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/694716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/694716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/694716/2"}}},"commit":{"parents":[{"commit":"fc12b9da7985936b03686de43985389a22c65cd7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-19 06:10:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-19 06:10:40.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I78cc2a10c2008302a250cf85ce6fddeda1739ba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb1f76b5d739cec228d9982df03401afa86fc2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb1f76b5d739cec228d9982df03401afa86fc2fa"}]},"branch":"refs/heads/master"},"eabbb0866f304ad392d354b3eca07203402ddf6a":{"kind":"REWORK","_number":3,"created":"2019-11-20 08:03:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/694716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/694716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/694716/3"}}},"commit":{"parents":[{"commit":"fc12b9da7985936b03686de43985389a22c65cd7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-20 08:03:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-20 08:03:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I78cc2a10c2008302a250cf85ce6fddeda1739ba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eabbb0866f304ad392d354b3eca07203402ddf6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eabbb0866f304ad392d354b3eca07203402ddf6a"}]},"branch":"refs/heads/master"},"10bf30988ba9f90a6947c8ebb4f343fbc83f6be7":{"kind":"REWORK","_number":4,"created":"2019-11-21 06:28:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/694716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/694716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/694716/4"}}},"commit":{"parents":[{"commit":"fc12b9da7985936b03686de43985389a22c65cd7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-21 06:28:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-21 06:28:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I78cc2a10c2008302a250cf85ce6fddeda1739ba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10bf30988ba9f90a6947c8ebb4f343fbc83f6be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10bf30988ba9f90a6947c8ebb4f343fbc83f6be7"}]},"branch":"refs/heads/master"},"fa1493208d1c73debecc516d3775161574e0ddbb":{"kind":"REWORK","_number":5,"created":"2019-11-22 06:35:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/694716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/694716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/694716/5"}}},"commit":{"parents":[{"commit":"4ee21ab9693ddb44666b59d35ef4d656e6bb3cc8","subject":"Merge \"Stop enforcing py2 requirements match starting with U\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ee21ab9693ddb44666b59d35ef4d656e6bb3cc8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-22 06:34:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-22 06:34:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I78cc2a10c2008302a250cf85ce6fddeda1739ba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa1493208d1c73debecc516d3775161574e0ddbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa1493208d1c73debecc516d3775161574e0ddbb"}]},"branch":"refs/heads/master"},"81b8bf06674d0e4ab04980bc3011b59d713910f6":{"kind":"REWORK","_number":6,"created":"2019-11-22 16:12:42.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/16/694716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/694716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/694716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/694716/6"}}},"commit":{"parents":[{"commit":"4ee21ab9693ddb44666b59d35ef4d656e6bb3cc8","subject":"Merge \"Stop enforcing py2 requirements match starting with U\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ee21ab9693ddb44666b59d35ef4d656e6bb3cc8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-22 06:34:57.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-22 16:12:29.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I78cc2a10c2008302a250cf85ce6fddeda1739ba3\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81b8bf06674d0e4ab04980bc3011b59d713910f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81b8bf06674d0e4ab04980bc3011b59d713910f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
