)]}'
{"id":"openstack%2Frequirements~694829","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ib90f6588b64aa1d06f33d5572dc93384ff280e3f","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ib90f6588b64aa1d06f33d5572dc93384ff280e3f","subject":"update constraint for python-openstackclient to new release 3.18.1","status":"MERGED","created":"2019-11-18 18:38:52.000000000","updated":"2019-11-19 02:35:38.000000000","submitted":"2019-11-19 02:35:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694829-1574130938671-ca8dfbed","meta_rev_id":"a26adcc4375a50784e43736c741d6bb7bb598848","_number":694829,"virtual_id_number":694829,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 02:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-18 18:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-18 22:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-18 22:39:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 18:48:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-18 22:39:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-19 02:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5de0be16219c7cf1aaa8870da009536a671b3375","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-18 18:38:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70f2bfbe8679694d5e9779e2f86f92196be5be6f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-18 18:48:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56cf496c6da4cf979dd66b21338f08306f070d95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 21:31:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d8e5603a64334b3fbb3c6a2fd9c77f03 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b64cc86c1a4c74b1e56ec87ba1e975 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/880e6a4abad646dfa38d14066ead55af : SUCCESS in 3m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8cbf60882d9c416da55232c6435e863c : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/291c58d50f2f4f838012ef4ee3625c71 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cd82097d7144845b07576a0d2442a98 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d301ead02c3e4ad69b7e624331d0fca8 : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f31ac6698b0d44dd9eb73609967cd1a7 : SUCCESS in 8m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/12388bf67a6647bea7c40493e43e6918 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/25ba8a241599425b8e8782d6ac612f41 : SUCCESS in 6m 24s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/85060997344848fc8a566f686df876c2 : SUCCESS in 10m 05s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/985b1c6cd9f0435e813e9c675baab44d : SUCCESS in 10m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9eec7a4abbf84367b976cc9f8616822d : SUCCESS in 4m 43s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ae9cc5b1c3c74b2cabc613d3ffd7d81a : SUCCESS in 4m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1387ed9be6db4896a3a7f4434725d3cc : SUCCESS in 10m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a89219d8da34421e99849e7f22dff41b : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83f70b605fcf45afa20aa31a3152a629 : SUCCESS in 7m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/36cc70c85bb04e0dbffbc4943d6987ed : SUCCESS in 22m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e3228979a0b8476b887cbb4e3e46b227 : SUCCESS in 23m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3e0f9a0c3cae44e48976b13dd3c0e89a : SUCCESS in 21m 46s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/81019ff163ac4c2dafe994f99c325370 : SUCCESS in 24m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a46daa0757414933b13ee1f9a0aa1db4 : SUCCESS in 12m 57s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0ce1404c643444dd9fc420d75a122d49 : SUCCESS in 14m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48ebdc19ddb04768a5edc3cd61e0dbaf : SUCCESS in 22m 09s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce5196e2c465489fa28499fd8508f854 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/15bc4e0e31f34aab9b2afaede059a362 : SUCCESS in 10m 11s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b90454e6149e4e5da8c09cc05eb14e5d : SUCCESS in 9m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/532e10749e854dd5a3d1f92db7d4c36c : SUCCESS in 1h 15m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/279c80ccc10643d88d93175f88b247df : SUCCESS in 1h 30m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2bd724a193f5411b85b2b1df56b38ee7 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b4df9ffec0fcb0ff9211fb6c6f770290deb186ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-18 22:39:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f74c9d1c25be40fd64f8e1a82924205b67a0aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:39:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c425e261b979294f2df4bc6297138d212dc52b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 02:35:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0c09d8063441049157434f5cabadac : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4685568d743d47178cb935e659b69658 : SUCCESS in 3m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c2e3da403994433bbe39d3dbb03bf71b : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f28e18389444d295c3b77c1a34df96 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb94ee77b29a4aec979d41d09d0ef90f : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87e550262f0149d4a0ff13060c288ac2 : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2b6c1eaecbc245dc9973a46855366dee : SUCCESS in 7m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/83c41223584a47a6bff3c725a1699778 : SUCCESS in 8m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/edc7cf95be45408296f98a0ea411c38e : SUCCESS in 9m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/eed4423ef9b34ce9930705bed48caf48 : SUCCESS in 11m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/618fb82c958b4de684fe4f06d5ee61ee : SUCCESS in 4m 56s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/5be5186c2c9848d5ba5911d034961a51 : SUCCESS in 5m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/23fc06d22bb347d0954e70b34e53ac91 : SUCCESS in 12m 21s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f6ab0944df704cfa92857009311f4fec : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6598472ef59c48d4bf3e9932fd3d5238 : SUCCESS in 7m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/273179e699f2426198b252fdc5d39637 : SUCCESS in 24m 12s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/2697f7076e004477ba29bc63c714a5af : SUCCESS in 28m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6f3db07ef7d2499e8223b70cc8e92121 : SUCCESS in 29m 21s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/187d14738eb440fb8b65b004970a74e6 : SUCCESS in 28m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c5c0dd15042c4a9f8cc0e125379354a2 : SUCCESS in 15m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/59fce350d054426dbcf75979167b8b6f : SUCCESS in 13m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/280c350fb5fd4cbdabe715d5b161f7ea : SUCCESS in 21m 12s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/cc22120e905e4cfca676349d2e3d6c30 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9c7358df86a046889a9c13e5455ebf94 : SUCCESS in 11m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e5d0a2dacf6e4740b9da894c74212251 : SUCCESS in 11m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a327d20ec7840ddbc222b7e4a1a31c0 : SUCCESS in 1h 05m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b017a8e9ee9419cbff373beaa01354b : SUCCESS in 1h 31m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a5a1e3e712124916aef3794239165060 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"402f4582df997667433082c387c3ae53cdb6b21d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 02:35:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4895a30aea623c343525ef540a7f0125e3fb7c15","revisions":{"4895a30aea623c343525ef540a7f0125e3fb7c15":{"kind":"REWORK","_number":1,"created":"2019-11-18 18:38:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/694829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/694829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/694829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/694829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/694829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/694829/1"}}},"commit":{"parents":[{"commit":"df494f4811a7deb8ff0d57601f6d3e6e5a01dfb0","subject":"Merge \"update constraint for mistral-lib to new release 1.1.1\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df494f4811a7deb8ff0d57601f6d3e6e5a01dfb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-18 18:38:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-18 18:38:50.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 3.18.1","message":"update constraint for python-openstackclient to new release 3.18.1\n\nChange-Id: Ib90f6588b64aa1d06f33d5572dc93384ff280e3f\nmeta:version: 3.18.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: melanie witt \u003cmelwittt@gmail.com\u003e\nmeta:release:Commit: melanie witt \u003cmelwittt@gmail.com\u003e\nmeta:release:Change-Id: I348886db00a292468d16ba8bb46a67a648872e9c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+1: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4895a30aea623c343525ef540a7f0125e3fb7c15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4895a30aea623c343525ef540a7f0125e3fb7c15"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
