)]}'
{"id":"openstack%2Frequirements~695020","triplet_id":"openstack%2Frequirements~stable%2Fnewton~Ic58bd79a192e6624b63a640cb5189d6d0f816250","project":"openstack/requirements","branch":"stable/newton","topic":"newton","hashtags":[],"change_id":"Ic58bd79a192e6624b63a640cb5189d6d0f816250","subject":"Fix the cryptography install.","status":"ABANDONED","created":"2019-11-19 15:52:27.000000000","updated":"2020-02-16 17:25:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00e1c195f8944c845931ca8f63813a8d8815f63e","_number":695020,"virtual_id_number":695020,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-19 16:03:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-20 15:13:08.000000000","_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-2,"date":"2020-01-20 16:57:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2019-11-19 15:56:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-11-19 15:58:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-11-19 15:56:57.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 15:56:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-19 15:56:57.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-11-19 15:58:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-19 16:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-20 16:57:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9492664273f26616d2ce1ec304b463ef9b4f83f6","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-19 15:52:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6196ee8ed8773d3f41b29ca370672680b9e0a9c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-19 15:56:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd02e039d20240d8e91096bddc71c77e8cfe4b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:03:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9286622e206f4e55bce0cc93877aa3a7 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/312719897a4546e982acda3116d5adca : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/feb941639ac24a5e97d432d508668682 : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/eb1753307b754c9687808e2a011cecc6 : FAILURE in 9m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e9893a643e4054526db02a0182ee1db6d00161a8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-20 15:52:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e485d2de3b1bb892132c7a9f3589b88777ceb73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 18:36:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a917dfae61ee442bb140c795b7619dde : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d208ccf63fcc4535aff1c995f1155779 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6c2801b47d24ec8851f1ca48a7f0a26 : SUCCESS in 5m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ded6da54c1144e1f807a6759c41ce3e0 : FAILURE in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"96c23e3fa37d865512636ae3c79ed025be0539a2","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-21 13:05:25.000000000","message":"Patch Set 1:\n\nrechech","accounts_in_message":[],"_revision_number":1},{"id":"ccbb5d42f6faf1e1354a5ca8c217e7935a7c2e57","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-21 14:18:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5dadb0680afc5ed23cd4dbc11bfada0fbb015ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 15:10:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51c7f1e0e164fa0ae25f37cd137c454 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba8cf03042804225b8efe32a96df7627 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/266a7412918a4e8cb8c5ac2fb95d42d3 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/38425200f46047c7a5d8a537dddff2a4 : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d3ed2a4351be461601c8ad9931d2a107cba17b12","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-20 15:13:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1fc807be9378f6952d99da10f6a18c4ce07c30a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 16:50:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08707508534f4642b241d8f220fc5dad : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67ae99f303db46ddb1e7a8f5605a47b3 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a1687be9a564c09b513bf9c100c509f : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2f5905de27bb47c6a2c703d3f3c66412 : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fee2e6efc7130c83292e3f3860e88123935900f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-20 16:57:55.000000000","message":"Patch Set 1: Code-Review-2\n\nnewton\u0027s dead jim, if downstream\u0027s need to use a changed upper-constraints there\u0027s a variable they can override to do so, upstream considers this closed","accounts_in_message":[],"_revision_number":1},{"id":"fd9c00e9924e11431120788fdddc8a27130ae400","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 17:25:51.000000000","message":"Abandoned\n\nnewton\u0027s dead","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c48196b81767f4bc85caf612c957e4f07f301272","revisions":{"c48196b81767f4bc85caf612c957e4f07f301272":{"kind":"REWORK","_number":1,"created":"2019-11-19 15:52:27.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/20/695020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/695020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/695020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/695020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/695020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/695020/1"}}},"commit":{"parents":[{"commit":"87909ac4f0112e8a5aed42da3d6fcdf4ce356d64","subject":"Merge \"Replace openstack.org git:// URLs with https://\" into stable/newton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87909ac4f0112e8a5aed42da3d6fcdf4ce356d64"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-19 15:52:14.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-19 15:52:14.000000000","tz":60},"subject":"Fix the cryptography install.","message":"Fix the cryptography install.\n\nThe newton version is end of life but is still supported for the\ndownstream. Currently if you tried to run the unit tests on newton, you\ncan\u0027t. The cryptography can\u0027t compile. With this cryptography version\nthat compile and unit tests run well.\n\nChange-Id: Ic58bd79a192e6624b63a640cb5189d6d0f816250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c48196b81767f4bc85caf612c957e4f07f301272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c48196b81767f4bc85caf612c957e4f07f301272"}]},"branch":"refs/heads/stable/newton"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
