)]}'
{"id":"openstack%2Frequirements~695216","triplet_id":"openstack%2Frequirements~master~I2eeb951f900a9d6916e57371d5fd7b8368cfe327","project":"openstack/requirements","branch":"master","topic":"zeroconf","hashtags":[],"change_id":"I2eeb951f900a9d6916e57371d5fd7b8368cfe327","subject":"Update zeroconf constraint to new release 0.24.0","status":"ABANDONED","created":"2019-11-20 15:10:35.000000000","updated":"2019-11-24 16:48:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a5ee9f4783ea91443433c591d0f90636daf0bdc","_number":695216,"virtual_id_number":695216,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-20 19:33:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-20 20:01:32.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-11-20 18:16:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 18:16:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-20 19:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d9d1f8f8706e6805608d17780280fcda94fd338","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-20 15:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49bd3fde3ff666081edf889f7ae5f085723748dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-20 18:16:45.000000000","message":"Patch Set 1: Code-Review-1\n\nany reason this needs to bypass the normal bot update?","accounts_in_message":[],"_revision_number":1},{"id":"be55bfa8a8d29ad1ce913ffdf45c38aa13a05261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:33:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c5748c34424531a66cf6639b8224ae : SUCCESS in 5m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9bbf59b15d2d4a548a283149dea15fcd : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/575dc720700f4fcbaa637bb8fad033d1 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73ea517b21fd413e99316446e3ad02de : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2a3338d90e4b4689b35dbcdf7b14aa : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66331f1d3fb14825b9668ddff11733ce : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d21486901b0438c8ccd41fbf76a6987 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39b430aef91a48b18b1f2d0c0237d940 : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cbeb15a0b99a4d07aaeb7d9c6753ce4c : SUCCESS in 8m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7e985ab4fede4ecbb987f8f895c39cc0 : SUCCESS in 7m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6fe5db24750d448e81d44fbf08ece02a : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7129d92bd78a48699db94d6230bc86b7 : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c550cf5c49404ba29d52e3837a8a41c1 : SUCCESS in 6m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/23be7fa45b144895b68a753cb592a319 : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cbb6cc485b5d48df8cfe9a21c453a04a : SUCCESS in 9m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d125cf91285498cbb29e9e753ad4865 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f264e8bec79e46d48c7dfd53c489d860 : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/94e861482e8945fab97befa589984749 : SUCCESS in 24m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1224a9be8164b56b0300e4fdef5b5a2 : SUCCESS in 15m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e6530dc3b8074552ab1e6555c89a94af : SUCCESS in 22m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6cb4b32b15b0453085483abf41b40943 : SUCCESS in 24m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/96c67759ab454c0181181a38675ce799 : SUCCESS in 13m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9a57b328b0645f18167be8be28a353a : SUCCESS in 15m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/da90e51d3854415a8d02c640a21a7f7d : SUCCESS in 16m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/210d8ac8902f47489171d27d9cb7c478 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4024548b942b412ea5d8244fc3d82e9f : SUCCESS in 4m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b9611cc5bb10432a91b1220af6500a89 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b26377c98714fef9446cfcc240a79bc : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7e897343922c47ec8932dcaa7178645e : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9c543d8032414e9d914b23af426b917f : SUCCESS in 9m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a98daaa3e9cb49fcbd89423750524954 : SUCCESS in 1h 27m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bd6fdbf9127f4a0f94a12212744ccf10 : SUCCESS in 1h 47m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab0df5b82b7b4942839a2c89f65f937e : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"406ad673ef21c51051b00a1ed9adbfcc155c2703","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-20 20:01:32.000000000","message":"Patch Set 1:\n\nJust not sure when it\u0027s going to happen. And this is handy for testing with Depends-On. I can abandon if not needed.","accounts_in_message":[],"_revision_number":1},{"id":"30bf51c344a123ae088dbbf8c1db8f190201de6a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-24 16:48:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"732515745d0cba86dc36a52cbce3d948577e06e5","revisions":{"732515745d0cba86dc36a52cbce3d948577e06e5":{"kind":"REWORK","_number":1,"created":"2019-11-20 15:10:35.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/16/695216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/695216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/695216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/695216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/695216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/695216/1"}}},"commit":{"parents":[{"commit":"fc12b9da7985936b03686de43985389a22c65cd7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc12b9da7985936b03686de43985389a22c65cd7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-11-20 15:10:21.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-11-20 15:10:21.000000000","tz":60},"subject":"Update zeroconf constraint to new release 0.24.0","message":"Update zeroconf constraint to new release 0.24.0\n\nChange-Id: I2eeb951f900a9d6916e57371d5fd7b8368cfe327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/732515745d0cba86dc36a52cbce3d948577e06e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/732515745d0cba86dc36a52cbce3d948577e06e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
