)]}'
{"id":"openstack%2Frequirements~695791","triplet_id":"openstack%2Frequirements~master~Ibb2e4d0c859e6cd1421c8f87a37d3c8d439ffe27","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ibb2e4d0c859e6cd1421c8f87a37d3c8d439ffe27","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-23 06:28:28.000000000","updated":"2019-11-25 22:07:56.000000000","submitted":"2019-11-25 22:06:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695791-1574719586689-de65efff","meta_rev_id":"e73c9d35f5326d44957e4d38cc1d8d7e82bc0141","_number":695791,"virtual_id_number":695791,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 22:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-25 17:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-25 17:23:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 14:33:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-25 17:23:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-25 22:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"562267afdd4f2d37e6f0c2b7988648b4fdd1a70c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-23 06:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"800b1565f885618a7e71f111e69f7e472f947edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 07:50:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c15fd57b1284ce48a843c8fe7531b0e : SUCCESS in 3m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a03df899204943fa9e9e2ec4b3ae80fb : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e89d77d8cb4d7db2cb53fee759a679 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/633cc5aef77d44308b3e56dc24a45933 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f499aa85bc43e0971a8066d2ac9437 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fce4590a01ae47b798612eab60c3ee61 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13f54425925748c991390aa3c7cedcc6 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63ffa77e5f334fcba9044abc93b1edf4 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3115beb3dbf84540906d9c36e4f54861 : FAILURE in 11m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2bd6ff5bafe465aa92b2581d88be3ef : FAILURE in 29m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e91f7fbbfe7648c6958d68a5e8b2083e : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5fd1e3afebd541659b2bd256066ac62b : FAILURE in 10m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f9eeb5ed37134f00bdd3814be071db46 : SUCCESS in 4m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/06a8bd06338d4d82a6469cf74c404081 : FAILURE in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bd0f79a5ede8415eb8e43b30e4ec2a7e : SUCCESS in 11m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/435dede0a53943b689c56b89b349bad3 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/890ef9fdb5364d5eaa6efe5d498d97b9 : SUCCESS in 13m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c0f91262bfaa41f6b01028410fdab8b3 : SUCCESS in 17m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a29cf246eb84316ad5bafd27f2d6a85 : SUCCESS in 13m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b6b7747aac1f461287b4a27327c4de21 : SUCCESS in 24m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3165add333124891a20ea654d0a72bec : SUCCESS in 20m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/272011d391944352bda6a5d657c7ff74 : SUCCESS in 10m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/88299c0ff61649a8a9bff9209d7aa6fe : FAILURE in 11m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e50d873eef2d4ace9dab380d99ff8953 : FAILURE in 11m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/835d4dbb676444bbbd9bedbd437aca3f : SUCCESS in 20m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bb1fff6eda2049beae5b2c091fc5148d : FAILURE in 4m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/766459f716cf424e86b22bb39d6d5ff8 : FAILURE in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a135087c0d074d5d98ac9d1e34acfbe2 : SUCCESS in 5m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/67c3f07a9cd0426987599321be9a8f3a : SUCCESS in 9m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/198074bcef7c4f0982f9773a6c2b96b3 : SUCCESS in 8m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0dfda0784c824108a0f66ea702c5dbfb : FAILURE in 1h 07m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b738ac0260a48b8b76c05283260f57c : FAILURE in 1h 08m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b805de281ffb485a89f260b80becd9bc : FAILURE in 1h 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cebf56ef0b8755de1dc0156fc94241647c4369ab","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-25 06:12:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b258de37a65ed0729b7cffa3fe955225bbd4ead7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:17:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/607444184d784ff68352e78eca70a3c8 : SUCCESS in 8m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b5c5f8af2ec04e4bbe3616c3f51ebfa6 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814353292de2463bb5c8f86e79c9b321 : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64a2e91b4b3b4d98a7cf344e448267ca : SUCCESS in 11m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f1a6de726c74c74b57ee351063328c2 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/777ca74c407e44c6a6c68f2214ce4292 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/387f5a1211224e92a456a118b1661c42 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3dad054f9c3f43e2ace7577b56fba3cd : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/25aacba4e9e740ccb46b42b0003199a5 : FAILURE in 7m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/85032a31646f482cbcbc60e1f5796628 : FAILURE in 11m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b717959cfe3e4ab896d9e11d39a57400 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd44ca68cf494b53a4e4974aaf3d32ad : FAILURE in 12m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ea12fee8adc74ed2abc30f8e36f02d6f : SUCCESS in 5m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d807f8e1c427423cb803fbd0825f9f18 : FAILURE in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c9e0f309ad83405083eb3ef7113ebca0 : SUCCESS in 16m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dbedbf45bcda4ce8b7631d854522b321 : SUCCESS in 14m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92f28b6aac2c4c399adf7d43bb4f44c6 : SUCCESS in 23m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4a78dd36870c49ccaa8d3eb0a3a2f75b : SUCCESS in 13m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f90d2abd9e5340b0be6a698b812ef0a5 : SUCCESS in 13m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/71aa341f994b46028156e34928437056 : SUCCESS in 24m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8b804f9311b94f73bd59d0964810d1f7 : SUCCESS in 25m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a2c0c48899c74903afbcb5bba1ab6ee6 : SUCCESS in 14m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/90f6596b10b7484bb321160a80176394 : FAILURE in 11m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aaf27ff2e5684ccb8558231f199ee2be : FAILURE in 12m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6ba2ad98bc44bfe8d229c3ae6a82a59 : SUCCESS in 22m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/04079dd89f344e3baa46d5194cf3a9a2 : FAILURE in 5m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/879b97eb961a430893176aec6f1b7b46 : FAILURE in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af578efbef08403b96cfb5da07d6d1da : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cd24b7e38c5941048d234b48a9521368 : SUCCESS in 14m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f20b7a48525446f799ec20b131bc055f : SUCCESS in 16m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c6cc97489414dbb999c5ab7e1f05e00 : FAILURE in 1h 20m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9243d102340548b4abb273d6a163959c : FAILURE in 1h 06m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e5a1b3475dd04bd6b04fd084f0d53cf7 : FAILURE in 54m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e3445f3d9a98d1621f8ae656d4c02e71650771ef","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 14:33:45.000000000","message":"Patch Set 2: Code-Review-1\n\nWe need to bump openstack/taskflow to pick up https://review.opendev.org/689611 before we can bump networkx to 2.4.","accounts_in_message":[],"_revision_number":2},{"id":"4f8df654b17d3f9f4d02d0baed5a69ee7b67ff88","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 14:36:40.000000000","message":"Patch Set 2:\n\nIt looks like the release for openstack/taskflow is here: https://review.opendev.org/#/c/694844/","accounts_in_message":[],"_revision_number":2},{"id":"2517f3735bdc3998efd4069738a552bb8f8ebb15","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 15:28:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4465fe057e91db346d955fa0b2afc9f5916b790","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 15:28:59.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"c1339ce0ef579d11dbcae74bff9d0d3e1a15913e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:09:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6642b3f5e34b38b1422ecc6e1a102e : SUCCESS in 3m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4a670f62a46e45098b5490cbcefda5ab : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67df2318c5144768d0c296477000ab3 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f92efd62e4e14c4cb40ade10c850c4bc : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28b99c9f78634955b19a1e4bf0c76944 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14dce1a2ed084120b5c8931e3f811d15 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/734e23dc921d49238a83ca9cb73d15cf : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85acf6349d8344b7ba09ab18175af892 : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fac0475cc68b43c5babb3f9ef47b6c56 : SUCCESS in 8m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b751f83e71a495192fa6ca7ca381f5d : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/63fc7d549f86445688fc3d049e9543f3 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f6c1f290047b40adb60320e28c199b5d : SUCCESS in 8m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3fb93583476843218a7129d753019b06 : SUCCESS in 4m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/afa3c8f0287d423dbc9198f817e2d7e3 : SUCCESS in 4m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/27576fe12ed145bda1438db0c3b2b70a : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b7ac7a02df745ca9476df0c044533c6 : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c93ef3f092d148fbba32be5c19825226 : SUCCESS in 7m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d67adc654cee47fa923ad25bb2c3842b : SUCCESS in 14m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e2269c5d1b224ba3b391dc8809b82660 : SUCCESS in 13m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0e2992f07afd4de5b12259e3bfb2e009 : SUCCESS in 23m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/04126d5699b44010841eaed66f588838 : SUCCESS in 24m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d1040b3c2758477bba118d1a432dbc1a : SUCCESS in 12m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b0753a39399c40d18e74e8769b3f0d73 : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/368cb19229aa4ce68693ed5d14f133f5 : SUCCESS in 13m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac56549d39cd49fd92ad9a1e43c14d00 : SUCCESS in 15m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/487f817c990246c7b4543ead1506f95c : SUCCESS in 4m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ca251f58842c4811a474075441e605d7 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/445fc0e9119740f98f0d5c6dbf2cb959 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cea4f3cabe9a424586b4f169995eab90 : SUCCESS in 14m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f592d23160af4bc89ed2c8115d1d6f43 : SUCCESS in 9m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e49070481e14b2c9c0bbfff22767b25 : SUCCESS in 1h 08m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/644b95bcb96c49f4b4858d6e820f6062 : SUCCESS in 1h 34m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1ac75ba076bb4ee8bede50e2e0b9f906 : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0c22ad3980563259fd7b212165192112d3214f91","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 17:23:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5e489853f8ba338cf0e54661e599cf97f157f166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:25:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94122fe387682465c62819759f6bd1afe70af56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 22:06:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f16afa3732141ea925faf8d17f92683 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de5133730ea4002820518062f000b23 : SUCCESS in 2m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6822088e196443a5857785bb13f019af : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ff99bec35b044469b6d63cf4f9255e0 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bb2771ffaaa4aa5909572d9b5a1ef0b : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/880b018a228e4dada786f0e0aaf869fb : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77baa46acdf44f988d5834359d3e492d : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2072c17b9c474db7905dcb14a0c4fc85 : SUCCESS in 7m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f431a6dc2d7c4af1b0e1e3d0408d41bf : SUCCESS in 7m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4616647951364c07a05bd448eabd140b : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eeeb4597cad443a2bf98adc0bec2e067 : SUCCESS in 9m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5a92442d89bf4cbfa4bd1b03ae6e8f68 : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d4894d33efe454fb13cf8cdd80a34b0 : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4c67f7bd4ddc404eb8bf5dc748dc4f8d : SUCCESS in 10m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d6e53d585d3b46f39a1e259b5d3ebcc1 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17766392b6ea4e4ba29e23355ca19c6a : SUCCESS in 6m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9677269d8f394074bbeba2a638e60f75 : SUCCESS in 13m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b68cdcbbd22f458daa7dd7bc85f5080a : SUCCESS in 14m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c8ab25207e0846fb9a1528cc500c4038 : SUCCESS in 23m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ab86302070024073a88bfbc4c280f6cd : SUCCESS in 24m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/266e9b2300d142d4b17b630052304a20 : SUCCESS in 13m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/daffef972f354059872fa0beafda6203 : SUCCESS in 12m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/053fc0d57e614d78be7bf5190d86d24d : SUCCESS in 14m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8b220520d0a4a8aa956320b1b106c51 : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e690be85bba47979d76d7dab5565a48 : SUCCESS in 5m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/430dedcc5acd4188950b1b8ee5ded527 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90d74bbb6d324203b75f7b3ba1211dc1 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32b1e85172b448c6a122b64352a91b96 : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c96e6da4f3664542baff6c8067e9e384 : SUCCESS in 9m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/266db6880a334bb9b3c3cafcc3392696 : SUCCESS in 1h 04m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dd7e889dc2b4410b9f4cb104fc0548ec : SUCCESS in 1h 34m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c54956249c364cbf9e75b245ca87b3d4 : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":3},{"id":"67c0556ca4e2ed7f149833c0eda0b0bec253c81b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 22:06:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e73c9d35f5326d44957e4d38cc1d8d7e82bc0141","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 22:07:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9c38fdcf274c12a5682c853752c751 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7978479d64f13c1ca01fc877a70a170b64a83234","revisions":{"894a9111d4198b9249cd7907eef56fd0cfbc9839":{"kind":"REWORK","_number":1,"created":"2019-11-23 06:28:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/695791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/695791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/695791/1"}}},"commit":{"parents":[{"commit":"81b8bf06674d0e4ab04980bc3011b59d713910f6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81b8bf06674d0e4ab04980bc3011b59d713910f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-23 06:28:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-23 06:28:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibb2e4d0c859e6cd1421c8f87a37d3c8d439ffe27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/894a9111d4198b9249cd7907eef56fd0cfbc9839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/894a9111d4198b9249cd7907eef56fd0cfbc9839"}]},"branch":"refs/heads/master"},"5fd3723fef2ce36b4aadb9df1fdc495118fc543a":{"kind":"REWORK","_number":2,"created":"2019-11-25 06:12:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/695791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/695791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/695791/2"}}},"commit":{"parents":[{"commit":"bf122399158e96ecf69de6bec3860ee2df4c9a5e","subject":"fix gunicorn cap (typo made it in)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf122399158e96ecf69de6bec3860ee2df4c9a5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 06:12:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 06:12:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibb2e4d0c859e6cd1421c8f87a37d3c8d439ffe27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5fd3723fef2ce36b4aadb9df1fdc495118fc543a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5fd3723fef2ce36b4aadb9df1fdc495118fc543a"}]},"branch":"refs/heads/master"},"7978479d64f13c1ca01fc877a70a170b64a83234":{"kind":"REWORK","_number":3,"created":"2019-11-25 15:28:55.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/695791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/695791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/695791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/695791/3"}}},"commit":{"parents":[{"commit":"bf122399158e96ecf69de6bec3860ee2df4c9a5e","subject":"fix gunicorn cap (typo made it in)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf122399158e96ecf69de6bec3860ee2df4c9a5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 06:12:16.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-25 15:19:06.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back...\n\nbotocore/boto3 (they hard lock on a version of python-dateutil)\nnetworkx (test issues)\ntinyrpc (test issues)\n\nChange-Id: Ibb2e4d0c859e6cd1421c8f87a37d3c8d439ffe27\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7978479d64f13c1ca01fc877a70a170b64a83234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7978479d64f13c1ca01fc877a70a170b64a83234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
