)]}'
{"id":"openstack%2Frequirements~695952","triplet_id":"openstack%2Frequirements~master~I5e2332dfd404a4b543d43e52c5a91c389862c672","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5e2332dfd404a4b543d43e52c5a91c389862c672","subject":"update constraint for oslo.cache to new release 1.38.0","status":"MERGED","created":"2019-11-25 17:15:28.000000000","updated":"2019-11-26 02:22:41.000000000","submitted":"2019-11-26 02:21:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695952-1574734896179-c9d8f7e1","meta_rev_id":"5700ef1a0cd24468fa166b1999f2b74f007335ae","_number":695952,"virtual_id_number":695952,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 02:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-25 19:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-25 21:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-25 21:43:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 19:42:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-25 21:43:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-26 02:21:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61dbebf97a0ce08ce6db1577e2b30f0eaec2c932","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-25 17:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e91f849b4c0e9e5a926f4db2a6aed25786d5d7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 19:24:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95376096c3546818ec7a9d25f746f42 : SUCCESS in 4m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b7523ed3206e44b8815ba21a9ba23338 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a84d859d0f945be8ac5ddbbcca57d48 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/496979c142cc4f59969dbffd44d5a24f : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02865f26fbfd40238fb783f61cc63460 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/442caf20888f4302a71773a4f20ae0af : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f3ed3c39d114dd78811eeff80dfe009 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bb957702abb4459bce64839b54e20ca : SUCCESS in 4m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ffb6724bc326437f967a9dff8c1782bf : SUCCESS in 6m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8db5dd87c6ba4b19ad7e5a494a6696f6 : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/805096e8e6d44ec19983c8574113e6bb : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9cbfea52af344d0ba761e980b4d59e4a : SUCCESS in 12m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cc46a7e62b014ae9baaeae4a5c79c9c0 : SUCCESS in 5m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f18a5fb48a4e4016a8a60a0cb423cdea : SUCCESS in 7m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a4387c5e19594a84885032cf34aa4dd6 : SUCCESS in 10m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9558d63f9a3f4a77b5f19154ab29cb57 : RETRY_LIMIT in 2m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8a289dedbd340acbb66f91b8a07d738 : RETRY_LIMIT in 2m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1ea1dbdd62eb4eed9130b80bd555a7d3 : SUCCESS in 12m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5a667a2f9ece4e04bf58653defe92607 : SUCCESS in 22m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d71670302a14476dbc46c9377d3a9c61 : SUCCESS in 22m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d4a9a827331042fb8c217982557b1e8d : SUCCESS in 20m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/68d3394cc30542ceb197a65e0ec66291 : SUCCESS in 16m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/130a1228b5834421823dfc1081a54670 : SUCCESS in 11m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fe29e01c1ab7404d939ba494764290b5 : SUCCESS in 16m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7b0ca2b40584afb9c5a864a30b57302 : SUCCESS in 14m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/30941c4cb1b74042ab0f46ee9e1e8e96 : SUCCESS in 5m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ce314dae931144baacf06f467ea20bf1 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/801591b0a0db4887a91d678c2a3587b2 : SUCCESS in 5m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/010266ea128248659dcf236ee3985867 : SUCCESS in 9m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7988484a29849c2820d258d4b09b597 : SUCCESS in 12m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fc68b01c6f84668a0fc28c0517694db : SUCCESS in 1h 13m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/070f5d32425b44aea81b8b40be726119 : SUCCESS in 1h 35m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a8115df672fa42378e13657b08c4962a : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8a68365681cbce72f40a1ba489e6240750bfad71","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-25 19:42:27.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nE: Failed to fetch http://mirror.mtl01.inap.opendev.org/ubuntu/dists/bionic-updates/main/binary-amd64/Packages.gz  File has unexpected size (1119184 !\u003d 1119717). Mirror sync in progress?","accounts_in_message":[],"_revision_number":1},{"id":"487b5cc349ef69c1dd404fc717e6114efbfb05e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 19:43:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a09f4fd23697767fd8e15f62dd11c4f7020444af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 21:23:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e990f634f1466d99dcdf6b2b0f7c64 : SUCCESS in 4m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/28851cf13800468f975d32d937e65dd0 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90fb6e0d3c014f22a8b894d0187d2a09 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6aa296805dfc432488b4bc1fe0f4ae3d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d22870c03f949ae998b0ef2f12cd448 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10b178b956df4e1fbc453c56320da0b1 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebd40f7d62ea4764a2cdd2a12d8dcdb6 : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/822d9db4906a43ea9c5c477d7bd2c451 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ae1f6d3e7700457f874d70b09c852cb6 : SUCCESS in 9m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b03de7f10a944e58143f8c56c869639 : SUCCESS in 7m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4a1df0104e2947d0a121d59d2b797c78 : SUCCESS in 8m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0c3e2646caec472582e693740e15bfc6 : SUCCESS in 10m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/433e92cbb68f4c8385a1b2a3aad6964e : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/71a5dffc1f4544039201a1e7e77a6424 : SUCCESS in 5m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/59aadd23933a4c509bfda9250e64896d : SUCCESS in 10m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bd0f1371d2c4627a37c9c23084fff1b : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2972e7239f84342bb95c5c0082ac973 : SUCCESS in 7m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f4b5209a67194cd0819d40a0014e2dfd : SUCCESS in 20m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/355e1ae0ba084671a04af2ba5c0f4936 : SUCCESS in 20m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/285196b009634130b175bbfb913ffe6d : SUCCESS in 21m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b6c2a35d39904c1bb3cf8e97ba4c1ca8 : SUCCESS in 20m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/62ea6d12041a4a13acf391cac69e3822 : SUCCESS in 14m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea8c302fc3044657a3b4e4fcda856ea8 : SUCCESS in 13m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0c75972a21a94a988237dcb34439cf4b : SUCCESS in 15m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80db69f1b55f4642923cb8f8fc09946a : SUCCESS in 18m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a3b1ad0c5a046c2a1b20e05deeb97ea : SUCCESS in 4m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5940cebcbb8145878a8c2b89564e11a0 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f48716eb986044f099d12b76dcdda496 : SUCCESS in 6m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f5bbd7409dad48de85874f88d534a939 : SUCCESS in 9m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba4cbe1497b448b5a088338bebb6bf57 : SUCCESS in 9m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/023052a79bf846438ec647cf5c5ca085 : SUCCESS in 1h 16m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a44463eae9946e7b32488a323c1e977 : SUCCESS in 1h 34m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/55a07362de614d19934937901cdd7729 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0485eac0c1c0856b0c832576fbad03332fe6fee7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 21:43:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b880b5a8f99d2356dabb507f87ed45c48cdc56eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 21:46:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edcaa17254b2acbd1571b06ab5de9dbf2609b2fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 02:21:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3085323072e4155abbde073d1af67c6 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1fa8d7068241d18ca35477a1f2aec7 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0babcce484b3437d877368d254b4bad8 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/804b43707f4f4095a0132e8d9203e284 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb37af67b07241c285d0fe989c9f37e8 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c79cd200b80042f28688c9f437992d4d : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39c907ef7f6a41a09345dc6f68db2672 : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5349d66ae358492a8adb04c1d9310295 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/357be9f163d1480fbac464ec93036bfd : SUCCESS in 5m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9191afd7c23546b68e00b9221434ee6e : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3b2081b1e57c43e98cf3dc994ab64e91 : SUCCESS in 9m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/36e88d601559446d9072b244808b498f : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c59e1ecbe0bd433e8e0a78aac84691c5 : SUCCESS in 4m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e807581343564d5e9eea5bbf6a7d6723 : SUCCESS in 9m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb07df37a64849ce8849b82743dc14f2 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/233f26114865434eba5c421aabbbfc22 : SUCCESS in 5m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fb49c7b6ccba4b5aa8b067dc64b2fff8 : SUCCESS in 14m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ecd38c5e3cdb4df896242f5bd0f2163a : SUCCESS in 11m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a91e526992ae4140a53997cef577fae3 : SUCCESS in 21m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/004926e4570b45e79fd9cc566f23893f : SUCCESS in 21m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7644ad4729e84ec283b6f6f274c8708d : SUCCESS in 12m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed2974377faf4e08b6e8c3475085cdac : SUCCESS in 15m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d815398faa94af88c7d2e6f0b54b830 : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51dce0d5a7e249e4ac1bf1189294afb6 : SUCCESS in 17m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/94589c444bca4e6d818a6093abdb7a25 : SUCCESS in 3m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a1a3319195654cd9a67c0c69523f6595 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ecd6b41c0974f1fa0c68a8c8433373f : SUCCESS in 5m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/49f42258eb4040229e2ee52ed1f89cd5 : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd560e97f5934c5db2a7da9da44dcbcb : SUCCESS in 8m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/397e6452a18347b2a44b5d1b9ec576e1 : SUCCESS in 1h 11m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2df4179f18024e59b0d034cb717936fe : SUCCESS in 1h 34m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9d5df59a1eaa43f080da0e9ffc8025c9 : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"690a0c024a89502627a7d5a20ef3c3956cc4042c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 02:21:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"754f3956e624ba94e155cce8ef6625734d1f1491","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 02:22:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59ae6edf8b040b498d4253cb0169c37 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24973b53400daea48bdb3cb023b31386f126e20b","revisions":{"24973b53400daea48bdb3cb023b31386f126e20b":{"kind":"REWORK","_number":1,"created":"2019-11-25 17:15:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/695952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/695952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/695952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/695952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/695952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/695952/1"}}},"commit":{"parents":[{"commit":"bf122399158e96ecf69de6bec3860ee2df4c9a5e","subject":"fix gunicorn cap (typo made it in)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf122399158e96ecf69de6bec3860ee2df4c9a5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 17:15:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 17:15:26.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.38.0","message":"update constraint for oslo.cache to new release 1.38.0\n\nChange-Id: I5e2332dfd404a4b543d43e52c5a91c389862c672\nmeta:version: 1.38.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I1ffe1ad55d39cc37728e1c30dbc8d4b98f6383f4\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24973b53400daea48bdb3cb023b31386f126e20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24973b53400daea48bdb3cb023b31386f126e20b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
