)]}'
{"id":"openstack%2Frequirements~695955","triplet_id":"openstack%2Frequirements~master~I1093be20ef5d71d18aac485f1010e382dc128ede","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1093be20ef5d71d18aac485f1010e382dc128ede","subject":"update constraint for oslo.log to new release 3.45.0","status":"MERGED","created":"2019-11-25 17:16:05.000000000","updated":"2019-11-25 23:37:43.000000000","submitted":"2019-11-25 23:36:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695955-1574724962568-a496186d","meta_rev_id":"783deaf3edf6bc03dc54401e53a6fcb382028e8c","_number":695955,"virtual_id_number":695955,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 23:36:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-25 19:38:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-25 19:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-25 19:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 19:38:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-25 19:43:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-25 23:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8253ae1846fdf5c942e6e0350b1ddeb2d34be757","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-25 17:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6592336e4fd301c69b247b3f16ebe04d6923aa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 19:36:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72251f466264bd290909cfb289ef47e : SUCCESS in 4m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/da8c785be27c4356aa888671e7aec4d6 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869120605b0945cfb3eac1cea2b8a036 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/512ca6d61a114c5bb3eeccf48a5829ac : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b0ff23eb115440eb9028beb508cbc79 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55232d3565dd4bc3b12132ea395c90a6 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad678678e39a4a5e98268e36441b3be1 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb547e10b8164a6684b1f817c4b4a8ae : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/84d776a62e474afe94aff98290706c9a : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/56a52c63736a40609697cb0791d94f39 : SUCCESS in 7m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f48ef1e5359c4552a69590695e048cee : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a9c2666a4595406eb201cc51d5553654 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1e99dc948cfd4fd3b2322c8d4529f896 : SUCCESS in 6m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3727aa204e04e6f9e55105538ed3651 : SUCCESS in 5m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/90bcaa06bf6147d896c5538283d5ce7e : SUCCESS in 10m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d21c3381d45e4ff5823f6920394ae8d2 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef563416fa824c28baf1f526f0212489 : SUCCESS in 8m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/93261a4de5dd4caeaf6ead3af21c32c2 : SUCCESS in 19m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/244924f96a92407d9acc9afd901de153 : SUCCESS in 14m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8ddc9fc69223410db70d308bb336f59a : SUCCESS in 24m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fe49a3367e548acab394a8335206686 : SUCCESS in 30m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/21a3c63a53734478b214e7cf11296138 : SUCCESS in 13m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/adbe9bb720744e4e99c5ba6b2fb8c2d9 : SUCCESS in 14m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2d2142b7a5d6406792c67073bada813f : SUCCESS in 20m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1d601f8604f444aa467ec6c0ba3ea13 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/961e11ee0fc9499c9100263feed21c8f : SUCCESS in 4m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2c4a3fd866444cff8500dc4ac705433c : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d60715a58bc645e0a501b8f73546d3a7 : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce10d9e0964441e594b27c83011b59f9 : SUCCESS in 9m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aac5f05140a5423fbe1b5c8bf13baff8 : SUCCESS in 9m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/39b7f149a64d4f49b9aa10208eb908c3 : SUCCESS in 1h 25m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bceea0a402854f37b8623235b5afbdad : SUCCESS in 1h 37m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6a93374bfccc4c48b71abfc158b8ab68 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e113f7325c31fec1808c83bd32643445c90a5bd7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-25 19:38:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d082d47b3f79fc05f2070fc601bca86196276e52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 19:43:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56a3de24576eb2e836f32d465aff840e24cc987a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 19:44:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e2bd56fecab7c3f8b39a53f2df39dd8c0ac4c92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 23:36:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"297ca7b1b07edd92c996fce8a73ce3a7256ebce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 23:36:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3669d7b61036489b8aa2fdcc465d6c53 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686c947a7a4b463683551dcf8a50ce68 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/221ebca04b9e4b9f86a6c338b4690469 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa84d805d2a4584bf3584a988d9c8ab : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84e44e6fc33d4d478131f33d20f5b814 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e997e23d73a481c823dc83b38fcb7dc : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f6a18a4b107428985a2a55b27021127 : SUCCESS in 6m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c91000dc65d24d5db88850e803b3dbb9 : SUCCESS in 9m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9184e85d0cbe40309205188fdae5514d : SUCCESS in 9m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ba7bcc664533410a84d0415d078b6dd4 : SUCCESS in 11m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e243a7fedf14438b36a28b8301201fe : SUCCESS in 11m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c5d01e6860b74c73b2e07d8c30b2d992 : SUCCESS in 6m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/90ead62987de4704af4d27fc79f51ca8 : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/527debeec67a453bb9a8f478b1a35107 : SUCCESS in 12m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/70c020041be944d9bf9e97bd602b58e7 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7f92c7a390742f79e9024145749978f : SUCCESS in 9m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ee56143f4b594041836940561bc0b2df : SUCCESS in 15m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e2c823a591b4d57a0aed0c065e82a4b : SUCCESS in 27m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4f831f919b1244dc9f51dcb83947d077 : SUCCESS in 24m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/36a2c5495d894b05b3e1127aa8ec6b13 : SUCCESS in 24m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e18d13f3dddb46038cdd5571cdfb3804 : SUCCESS in 13m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0623c1ec225f447b81667e085652d0e0 : SUCCESS in 24m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e781de6777d74e80b90fe15b4df8ea05 : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db34bb91d2634d79b56de11c39ea772f : SUCCESS in 18m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2e1c12593de64f5f828d191e3002aeb7 : SUCCESS in 7m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/788607c2b1314163819c5cf89b9c85e4 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f43755542a5847cea63d3af4a31c421d : SUCCESS in 7m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0aa97610066d4cc889d91f4e895fcbb6 : SUCCESS in 10m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b8600032d62b47cd94ecb18ed86bbe2f : SUCCESS in 10m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3957ff2b3f604c5d87a8ca95ac6deeb3 : SUCCESS in 1h 19m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/297d744df1be4b9184edcc3c0d737dd4 : SUCCESS in 1h 25m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/62e70960ae1f48538e57ea22eec8048e : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4cef3b7b7c564597ef87db6b25fb651695d06c22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 23:37:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13289a53c62b41519df2338fcc606e3f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"247038923753a8a0251554a992ed8ad64554e309","revisions":{"247038923753a8a0251554a992ed8ad64554e309":{"kind":"REWORK","_number":1,"created":"2019-11-25 17:16:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/695955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/695955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/695955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/695955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/695955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/695955/1"}}},"commit":{"parents":[{"commit":"bf122399158e96ecf69de6bec3860ee2df4c9a5e","subject":"fix gunicorn cap (typo made it in)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf122399158e96ecf69de6bec3860ee2df4c9a5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 17:16:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 17:16:04.000000000","tz":0},"subject":"update constraint for oslo.log to new release 3.45.0","message":"update constraint for oslo.log to new release 3.45.0\n\nChange-Id: I1093be20ef5d71d18aac485f1010e382dc128ede\nmeta:version: 3.45.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I1ffe1ad55d39cc37728e1c30dbc8d4b98f6383f4\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/247038923753a8a0251554a992ed8ad64554e309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/247038923753a8a0251554a992ed8ad64554e309"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
