)]}'
{"id":"openstack%2Frequirements~695957","triplet_id":"openstack%2Frequirements~master~I0177f1a97dcd7e2f9631c510560829e118823701","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0177f1a97dcd7e2f9631c510560829e118823701","subject":"update constraint for futurist to new release 1.10.0","status":"MERGED","created":"2019-11-25 17:16:45.000000000","updated":"2019-11-26 00:33:07.000000000","submitted":"2019-11-26 00:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695957-1574728313809-a496186d","meta_rev_id":"6b3a4daf3314efe4806d3db2727ccb7a4b992e68","_number":695957,"virtual_id_number":695957,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 00:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-25 19:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-25 21:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-25 21:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 19:45:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-25 21:00:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-26 00:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0eb6963bddea522e606f400652ccced7a826270f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-25 17:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dea88af9cd86ea4caf72bed85ee02857ac82e8c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-25 19:45:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7898cd1c6020b837e9c2fe68ce1d9fd93d1a20b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 19:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fe66c9c2d8414e99dabb258ba554f7 : SUCCESS in 3m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/19c83f8177ef48ff8b4ad11c3dcfc87c : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a6e458651c4a0886b94c9607e3c9b2 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59f90026357f42f0b9ad3b4a0cd8de2d : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d8717e8676f41f2b3973524bc39655b : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1883617a50e4d49b4f2bcda4d0ddc9d : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc8c18c98e634e84b2fda7c4118214b5 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8747cb7a5df4db6bcb50b220367db5f : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ffdbc66874c741be8b906a281c9a1a18 : SUCCESS in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5cf1dc330fb6427ea7878099bad2e3d8 : SUCCESS in 8m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f3af3ee0373b46768adf8994c28aa99d : SUCCESS in 12m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b439266e855b4b6e8f0a05208db0e6dc : SUCCESS in 11m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/31701f83874b479db5570431e09d6dbe : SUCCESS in 5m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cdb659e753864fb987bc7bbe060ba0bf : SUCCESS in 4m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a83c57cc51ad478891caa92a8b10ea0e : SUCCESS in 11m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5b44c50168449d398685ddf133947be : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ecf31c36c4849c2915e05e4ed292af8 : SUCCESS in 9m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7b30ccedf48a45abbd0325eb0e6013a2 : SUCCESS in 18m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/909848be3bc845bc82b7eae5722aac0e : SUCCESS in 15m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d41cef96756d49ec814b0f2a9924ec29 : SUCCESS in 23m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/93a8a2954aa04df5b73b61d322f5f082 : SUCCESS in 23m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7f3c57401ac94bd68bd82acb285c3ed4 : SUCCESS in 19m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/073667749dbd40b48247918e79245e49 : SUCCESS in 15m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/071d1e7e54374ec788f2eca70ba4be22 : SUCCESS in 17m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8680475de49f446b8cd3701cda310fc7 : SUCCESS in 18m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7b4d88fd68f14a4d9650340e59c6dc67 : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/61a51ae00d2b40b78469ead8d6ca25d7 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c62a348de8484fcfbe1dedd58949845c : SUCCESS in 5m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/943a4f6610e24a4387e1223acbdb9d3a : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd3d95451b9348d1bb7cf017296f6c55 : SUCCESS in 10m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb10655fd3a1453b9bb604b0becc998a : SUCCESS in 1h 21m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2b1a0696a7244f149dd3143f42c2cc8c : SUCCESS in 1h 41m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6eaa5b732c0a49bab30bf5aeeba00cf5 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"840eaf6d70bf9257e57cc20e0fb82eeb4c74dafb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-25 21:00:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aca335f8c11c298ec774b58e87d204dc2f92a968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 21:01:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48a394b83b014934f74a99a764572530014e727c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e0b49dc8b74f77825a2647a3bf7d67 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46259f28c82b4a91b1c349892207c96c : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4e0ab9655d94279b7dee15aeee442f4 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0820869757e3435099a3723e766643e9 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d09ca72c6a234ac48099ff642469726c : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89527ef6e3104a76a1735f1045ce5a9f : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42989f282fbf4107b03eb4fe7590e9bd : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7323f77abbca4d24b43730f2f63665cd : SUCCESS in 6m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/746b2a5366374194bef7a8034e598ff9 : SUCCESS in 6m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a9bec4db161e4796b879b3bdfc06812e : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a48591547f247efa86970792172dfb6 : SUCCESS in 9m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/af7948a7fec44d1cbf46f2e8da65126a : SUCCESS in 5m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/136fd977fc5642b6b26e32ab2feb8864 : SUCCESS in 6m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1bb188b28f014eca97a94a08c2a21552 : SUCCESS in 8m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3173ea572ef44ee3a3c960d0627af553 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6353d83875d94f3d8f499254bf15c9d2 : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/31474a907e6c476d85033b106b13b00c : SUCCESS in 22m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/153e9b430fec460db969062dc00578a8 : SUCCESS in 22m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7a23e5212a264f0b84cf592e950b2f10 : SUCCESS in 18m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a778b12dfc7d466594692f9fe704a9c0 : SUCCESS in 19m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d6a8ade03804fa78e0ecd80243e4152 : SUCCESS in 13m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/654036b0cd4f4fda9bbd27600789817e : SUCCESS in 13m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d8f9dd6743b44538998737ab07c3349d : SUCCESS in 29m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcedfcdd3ebb4b4d96952fa13aa8fc2e : SUCCESS in 17m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57e2ea935a9549df9529c80abe19cb91 : SUCCESS in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/70d6c0ddee474b35a85bfeb0159c1af6 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/111a9b9beff0457eac86d843807fbb3f : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5efb93ea05044447ba517a8d1e829908 : SUCCESS in 10m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b54275f1819440609d8737f2fb8cd399 : SUCCESS in 10m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe6e1e2a84f041bb81d2099d55be577d : SUCCESS in 1h 04m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a63f80ac99b4309867ad3ad0e410d86 : SUCCESS in 1h 23m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/479790e6f576439ba96bcb576b0499c5 : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"855a56ba08f8ace2f8f8f6c8fefdc9bf33723263","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:31:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cb77fbda0519ecf1c5c564307ce1a15c5fdb2fa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:33:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758e08a6a18a481cabf6879f7710e173 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e08549d6013732fe6acb8c6e4a55652b5196bdaf","revisions":{"e08549d6013732fe6acb8c6e4a55652b5196bdaf":{"kind":"REWORK","_number":1,"created":"2019-11-25 17:16:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/695957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/695957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/695957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/695957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/695957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/695957/1"}}},"commit":{"parents":[{"commit":"bf122399158e96ecf69de6bec3860ee2df4c9a5e","subject":"fix gunicorn cap (typo made it in)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf122399158e96ecf69de6bec3860ee2df4c9a5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 17:16:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-25 17:16:43.000000000","tz":0},"subject":"update constraint for futurist to new release 1.10.0","message":"update constraint for futurist to new release 1.10.0\n\nChange-Id: I0177f1a97dcd7e2f9631c510560829e118823701\nmeta:version: 1.10.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I1ffe1ad55d39cc37728e1c30dbc8d4b98f6383f4\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e08549d6013732fe6acb8c6e4a55652b5196bdaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e08549d6013732fe6acb8c6e4a55652b5196bdaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
