)]}'
{"id":"openstack%2Frequirements~696232","triplet_id":"openstack%2Frequirements~master~Ia005a6bdc48fdfbdaea0adf0f6138ba04725c560","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ia005a6bdc48fdfbdaea0adf0f6138ba04725c560","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-27 06:15:24.000000000","updated":"2019-11-27 17:15:49.000000000","submitted":"2019-11-27 17:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696232-1574874805083-fe519b9b","meta_rev_id":"d268e2b84c2d3936d077efa1dbfd122a69f597a9","_number":696232,"virtual_id_number":696232,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 17:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-27 07:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-11-27 12:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-27 07:50:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 07:50:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-27 12:52:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-27 17:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec5728ba795ac02373ad9d6233cb7ac5ab1ad209","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-27 06:15:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ca87fb45003eee5b518b97298a037168e22f28e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 07:50:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"456b5e98ffa3998ea9a8fd62fb84f3c342374cd6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 07:50:07.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"7e4cee14545c77bc25731e3e4cb66035927daf4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 07:50:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9273c377b9d32e5063ab0a217aed7c58f76a6d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 09:37:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3595984c673c4f9f94a10a3a2bf0007a : SUCCESS in 3m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e3c4c12564e347f59215aa5e276a4e81 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5264c20a36b740b7a24bcae3c19ca060 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f9cf83faf5540c086a5e82f0c2186ec : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08c4d3db87954fcabfda03da2d9c6daa : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5b7cd3c3d19429c90469060ba594779 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23eacc31039a45948fb02186f955fb08 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d17e997933c6414c993973b5aae37772 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/104d6a2b01a54302bf674083b6abaf90 : SUCCESS in 15m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e4d6795d4654a0089f7e76638bc9ed0 : SUCCESS in 12m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b71e7767e3764926b74274069cf275a9 : SUCCESS in 17m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97dcc4713166415f9ffe51178cdc90c3 : SUCCESS in 15m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a7a14fa514b54032935638a8df2529b7 : SUCCESS in 7m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c568d3ac77f94b5587da54c6e2058525 : FAILURE in 7m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/238fc3b8ff38418c95e379ae93954271 : SUCCESS in 9m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55d609849d8e4bda899eef67fdde5dbc : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11333ace44e94f6cb01acf28ee78f461 : SUCCESS in 9m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e9b6c9e060ae48ca8869a4bb0982abb7 : SUCCESS in 23m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09db7af656624a8b981c9fc7e9842bec : SUCCESS in 28m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/153b3c2714924033879576fce46b089e : SUCCESS in 27m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cacd309377a14385b945301133a957eb : SUCCESS in 27m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b4d8b7ba358441ee9d6bd6605496fe3f : SUCCESS in 18m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b95ad4558e3540d4b17b3fcd5a1ab1e1 : SUCCESS in 18m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9b94b035f08b4738ba1507b4d5321774 : SUCCESS in 12m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/afce3e1342ab4bac98f930d2d57aed13 : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3bbc052bc234b8392f225bbb20acb4d : SUCCESS in 5m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8794b59753ba4b0a82f836ff5f30cbeb : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfb5f429192945779909d95a800187f1 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/834cb408b26e475aabfdf6ebba3c15a9 : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/49fe636096b84e9aab72ab3222e94490 : SUCCESS in 8m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a68676529f8d468990ae4980e324861a : SUCCESS in 1h 14m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e9cfad465dc44c6abf6e271e9d48dfa : SUCCESS in 1h 27m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ad46d0c4cb9648c6837e29d224a5df96 : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d975d378bbaeb48bd8b9d18d725a5b42e5e751e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-27 12:52:46.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck\n\nKnown glance, stestr, py3 issue: \"ValueError: Not a text type application/octet-stream\"","accounts_in_message":[],"_revision_number":2},{"id":"48ffa155a678fe633fef5205c3433b85be74b8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:36:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062570ee49f548ffb4855a19655fd78d : SUCCESS in 3m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/945463352e9445de9b33e58b8ec7b63a : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc4031e7c0a4abf92f01f07b9d2e60a : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afeb5b2db3b44385bded41748311b8a8 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/205515d55c4443d1a56b1b4aa281504a : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bad7308602cf4af2b4bec9cc7281e89f : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ce865091c6b47cc8f3aa9282be67119 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/885bf19f3d9d48b3b7a09727449efeee : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/87eb15421ecd4301848edf9a34a12a76 : SUCCESS in 9m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eed071bbd9874579aca19c61ab30477d : SUCCESS in 7m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/62079ec4198d43daa71d845f9e6f3564 : SUCCESS in 13m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7e3c6b878bd94c5e82d3658384dceae3 : SUCCESS in 8m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d9e2d2cb9ad14d50b897852bc50bc8fc : SUCCESS in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/baf6acd858fe4d0b9c380ba7125b428b : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3d7d6d9bea634a4b878c6c4697804e3f : SUCCESS in 12m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c87b7ac223f54d908d50883cd32919fd : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b28fb31f7fe411ca8ff670b300bdf98 : SUCCESS in 6m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bd15b03c196c4b5d9d30c24f139e41db : SUCCESS in 15m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f141989c1f5e4028bb0293c061914dc9 : SUCCESS in 15m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4696a8bdbbe1415d920a71163cf8d180 : SUCCESS in 23m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/08858c63cee8496a8501499e6a9112a4 : SUCCESS in 23m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ab514ce29951426b8c7d4b7f14e29ea1 : SUCCESS in 14m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5ac1a77d589642f99fae828d11c429b7 : SUCCESS in 15m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/42309faed8d046c2b339a187421574f7 : SUCCESS in 12m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd219543559448ddb6f2bcb56ef6bf6d : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/de46b6c11f364bea8a14c8e123a5f47f : SUCCESS in 4m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/beb0c85baa95423e8a5b2700831b76ea : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c9161347d79495285b592759178edb4 : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/749a631ca92a420d832b0d6a4c37d73a : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0b8ee12fb46b41f39241d5a9282cb51b : SUCCESS in 12m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bb33ebb781b46ad9f0188fe72a2073e : SUCCESS in 1h 06m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6770c87223ac4e718a5fe91ef5f5373b : SUCCESS in 1h 20m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/711870a3a3d644dea6bf3409f386bad0 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":2},{"id":"551d8a86be77f05ae41fb1c0b4451d70265e662c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:36:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10d98d0aaed7a86b475332b3a8d620ebdb0ea06f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:13:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae68317dedc54c6c92e424fb0f8f116d : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa8d02d73c04cbb89b33a9329bf8f0c : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01cb51b80c2646eb892ca2c93089aec1 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1c047bde62d4af29658b996b0cb2c17 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fc0492356f94469a9093d0968811def : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f11fc3eaf7344bdbb5a0564b245552ff : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/051f172233d94ca09740e2bdcab7324b : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/faa823f82bfb4ed79ef282c21024cdca : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2307108fb0ca40229c987726cafa2174 : SUCCESS in 7m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b6d8193ab6f543048a915eae68f422b1 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc2797f25d6b4f94a51cf31a8169f65c : SUCCESS in 9m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/198199956c1e430c93fa4017b2d1f334 : SUCCESS in 4m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3bd48d34c7764dad833a723518aa8923 : SUCCESS in 4m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5da81500576a4803a3ac7754b52b0e7b : SUCCESS in 10m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b6810680cbc43999c65163181bed6ff : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39480227ecdc4bcab29977a35f1f83b0 : SUCCESS in 6m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9d30ff0feba440d1a0d5805c4da55668 : SUCCESS in 14m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef26d10a31844dadb52447df0f6bcdc8 : SUCCESS in 11m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ef20761b31434ae4a286c22ff4d5269f : SUCCESS in 22m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2703c541b98e491397de8703d2e1055d : SUCCESS in 19m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/90c8542ba12a4395b9832fc676858e92 : SUCCESS in 14m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a1719c1054084cf0a87f51ab152a1917 : SUCCESS in 16m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ff92634e33174faab16722a9916d72ac : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f38b1ef2b6e4bb780541b0dd71ae6dd : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f7ca655a5eeb426c99ecc8ac61fa3371 : SUCCESS in 5m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f6dd56686bc242b58bea80ef684d8880 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4bd16021cd5b47e687cc57a755db3db0 : SUCCESS in 5m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/68b6026eb2bd4afd8d3b9db65454a8c2 : SUCCESS in 9m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cac03efb2814c509abf3e55b6f34ac7 : SUCCESS in 13m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ac0b154862c4c5e97e803fbcb05217f : SUCCESS in 1h 13m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1828cfb4943e4817ad468e14389176c8 : SUCCESS in 1h 47m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6517dc222acb430787675152b8ce558e : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6377081d10e73745f2b847c93663b5e3530ec18d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:13:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d268e2b84c2d3936d077efa1dbfd122a69f597a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:15:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833aeb7c876b4a71919da05020c787b2 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ccdbf27f0c2031db5b355e44dc162d3b225bf2a","revisions":{"14eb72c024723ee2e98976063aa5904710a37567":{"kind":"REWORK","_number":1,"created":"2019-11-27 06:15:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/696232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/696232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/696232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/696232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/696232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/696232/1"}}},"commit":{"parents":[{"commit":"d8d2816c2c90d0833ee9cb043e332909679222cd","subject":"update constraint for aodhclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8d2816c2c90d0833ee9cb043e332909679222cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 06:15:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 06:15:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia005a6bdc48fdfbdaea0adf0f6138ba04725c560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14eb72c024723ee2e98976063aa5904710a37567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14eb72c024723ee2e98976063aa5904710a37567"}]},"branch":"refs/heads/master"},"9ccdbf27f0c2031db5b355e44dc162d3b225bf2a":{"kind":"REWORK","_number":2,"created":"2019-11-27 07:50:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/32/696232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/696232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/696232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/696232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/696232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/696232/2"}}},"commit":{"parents":[{"commit":"d8d2816c2c90d0833ee9cb043e332909679222cd","subject":"update constraint for aodhclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8d2816c2c90d0833ee9cb043e332909679222cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 06:15:22.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-27 07:49:59.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia005a6bdc48fdfbdaea0adf0f6138ba04725c560\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ccdbf27f0c2031db5b355e44dc162d3b225bf2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ccdbf27f0c2031db5b355e44dc162d3b225bf2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
