)]}'
{"id":"openstack%2Frequirements~696261","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I8ce7e7d81d74b2a9a9cddffa4e12b6ba73c26ca3","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I8ce7e7d81d74b2a9a9cddffa4e12b6ba73c26ca3","subject":"update constraint for python-zunclient to new release 3.5.1","status":"MERGED","created":"2019-11-27 09:42:53.000000000","updated":"2019-11-27 18:39:53.000000000","submitted":"2019-11-27 18:39:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696261-1574879993281-1353a74c","meta_rev_id":"c521064fdb1035f704b294d9f4eb0f0e524f9649","_number":696261,"virtual_id_number":696261,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 18:39:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-27 12:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-27 16:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-27 16:45:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 12:53:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-27 16:45:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-27 18:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f8ec1a796ba70b0a985f67559ad8b4f2d987ba0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-27 09:42:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b951b780359d4aade5b87d2c0b59b316a4298ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 11:46:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/34bad1ddc680457990c435b425c3ff59 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380c6b32dce145ac8e59bdd70d09ddb7 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae6a9abcb341451a83b9fdbc1f58ea46 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7120aefcc764113b48aef1bdb8d732c : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b09f66dfb15466eb2f684f1adaf2eee : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51e4b36ce8ab41b3a5a3f583e3b0433a : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5848588e608046b79cf20811df86b536 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f9d9da41a80a4c698b634ba71ced8a1d : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e600fbaeb6f64a15b65843dfc56f517e : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b81f778b60ce4ea89ec17b45320957f8 : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2a858ffe552f4c1480db6ed44375f765 : SUCCESS in 10m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/18163a7d6a0b49708f77c5668a8f8a13 : SUCCESS in 4m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2dfc5a9139d445dbd91c5cbe672386d : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/07d7643ea98d46329bdd95e8cff452ef : SUCCESS in 11m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fc91d59b1e0149af862381021c4033bc : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7eb3fd391f74b9ba4915d259142a3b3 : SUCCESS in 6m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d9246ad187584bd48f5e8c942d9dc1ae : SUCCESS in 14m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/25e18f6a48db42dc9acd00c6c214b548 : SUCCESS in 11m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1999e7b32f7945d49be99da21274e4c4 : SUCCESS in 23m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4a7778487a674208be448648a33e0ab6 : SUCCESS in 21m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5665d9139e87410fb1a2a3d51b12a6cf : SUCCESS in 13m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/132f0361a8ac4430a60d3ae94d74af51 : SUCCESS in 11m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d35cde812fe446a391ec3d9ad420399a : SUCCESS in 14m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36d9cfaf7d444de588e36901e9c0dada : SUCCESS in 18m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24d5f0cee2944c92b442e1a3125479a0 : SUCCESS in 4m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d8f4a9a5639449e947deed0b7865319 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/76fea46cc4894df3b756d394af5685c6 : SUCCESS in 8m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d2ee6dc67944da4beef1da6f2e2d688 : FAILURE in 11m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3bf4c114b90249a0a0e7c910175e5df3 : SUCCESS in 1h 57m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ed8cb66995e4d8ab63ff43c7f52ec78 : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b87f628cf67228c027aee9d904bdf0df9bacc516","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-27 12:53:00.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"91fa0dbc2c86658f3d5af70acba4edb12274efac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:35:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f08d6c7f995f408a920f6c8ab6f24885 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a91af879fe4405bd54c2fd670806eb : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0921a244396a4a409b4e908986d4f20f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0399e4a3ca474428b9a30fd4ecb57414 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c1cabcf61b44cccb066e654cee58465 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e13125f8dde49bdaf28fcb9db674461 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/492be9f075fa4e09a6f24494ef877378 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e2cfcb33805e4ba9a73d127483d4e683 : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa683eee2a014980bfc2c1e85d2eda28 : SUCCESS in 9m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/613fc1906a9a4fad84b5c0d57ac171cb : SUCCESS in 8m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8be9f30256b645c28a5687dabf004be0 : SUCCESS in 8m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0cebbffac89a49079b70e4cf7fca8321 : SUCCESS in 4m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e8decc9a119a47d3b4dfc35a631e2782 : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/62b9df0071644881beac5b8c7f98c643 : SUCCESS in 11m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57e82a180fa6416280d5e7d6ad373858 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d534b6cf24ab49df835f09c22956e17d : SUCCESS in 7m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5a30f85ca1c540f296047a772c86b845 : SUCCESS in 27m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/734d8a1044984ee88785a804d6abe993 : SUCCESS in 20m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4591d6fa113345dbab2f86dc6f81fd48 : SUCCESS in 23m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/edd34c4202fd484ebc460b9cca4c5829 : SUCCESS in 19m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f0f950bfa3804455a2ecc59a80980f46 : SUCCESS in 16m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cce4c063f2dc4e66be0819596c310ca7 : SUCCESS in 13m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e8d1daf58914230bb5ef636933dffb8 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8390dd088229479b85365b599edc7c50 : SUCCESS in 18m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35a09733557848dfa567b62ecc1fb10f : SUCCESS in 5m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/65c1cceb22b740f98338136758264517 : SUCCESS in 8m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e39244d8f9e443d3b004301192ce20a0 : SUCCESS in 8m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/553aa13ba1ac4b91aa0b0d51361f3afd : SUCCESS in 1h 06m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e7cdd0b4cf744998bd952f90b4057ee : SUCCESS in 1h 36m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5438d5def0df40c7b04c46fe4215319b : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5992e46e819c221aa9d0816cb02af73493fda59a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 16:45:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c2d80509696429fffd605cd2eaa069a15d7c3cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 16:46:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"190472190e8366f353f7b391c4ae102484a5a2d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:39:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7223a7c885074f2f99d901132a07d64b : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea09c77b99bf4f63b401985a08f69b9f : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c759d92e2bdf4c4eb9f9e1f72574a2dd : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3125adf8581e4bc895a57e9216fb8f72 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4879fbed36c04a1f8fdaf2412500369f : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8031078370ee47dd98d32b6ec3a43fb7 : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/78eea81148fb4fc4bd0482aa8bc163c3 : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d69b490c5aff4ac09a21e16f877289df : SUCCESS in 6m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f960e7c3e6b14bd88a128592ad7d9238 : SUCCESS in 10m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b9fb4b5d6df5481a8fc2e7a5a5a43665 : SUCCESS in 11m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d2361847ce2a4ad3b4d8930f462fa02b : SUCCESS in 5m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/389585b0375c48d7b47669a12eb9d156 : SUCCESS in 4m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3c8782ba0cea4fdca1c92333814997bc : SUCCESS in 12m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/52eee7e783764080a3c5c1f40509b4a6 : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48f0cf1230144138a6f60caa878b900c : SUCCESS in 8m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/460f51a73f614be7be7c9e913906ead6 : SUCCESS in 13m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c90d8310f634867bb0aa45fb074d615 : SUCCESS in 13m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/020398948e954a999b73484cbad975f5 : SUCCESS in 26m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f9597b3f6fcb4ea3bb5c563d23df6fa4 : SUCCESS in 25m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e9444b9f88e04d1cb218385d2e4c22c4 : SUCCESS in 14m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a60771857a69446db87a9c839c97e173 : SUCCESS in 12m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/36fcc51626c94dbb913c03da78925f0e : SUCCESS in 14m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37cbdfb4b98b4b34835047a32759a4be : SUCCESS in 22m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdd9e04a9f8b400e919d4719b60eb309 : SUCCESS in 7m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a7fca5c9bbf9406ebccf25e963acc52c : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cad493fdb1c74c21baec506c84b5bb91 : SUCCESS in 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c0d281638874bf9ba3b454549ebe6c5 : SUCCESS in 1h 15m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2f6e79fb64574cdf894b0bb67e1fed65 : SUCCESS in 1h 34m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af959c34762444f58ee7d41359778526 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d0c8f70a04f944b601c9fbe4634de3487c1e790a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:39:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae","revisions":{"a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae":{"kind":"REWORK","_number":1,"created":"2019-11-27 09:42:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/696261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/696261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/696261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/696261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/696261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/696261/1"}}},"commit":{"parents":[{"commit":"ce8acbb5608dc8f33dba26c528d825dbe7c21e19","subject":"Merge \"Update constraints for amqp to new release 2.5.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce8acbb5608dc8f33dba26c528d825dbe7c21e19"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 09:42:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 09:42:51.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 3.5.1","message":"update constraint for python-zunclient to new release 3.5.1\n\nChange-Id: I8ce7e7d81d74b2a9a9cddffa4e12b6ba73c26ca3\nmeta:version: 3.5.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Commit: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Change-Id: Ic3f7b0395fc16407b95ff4367c92bb982b6decf0\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
