)]}'
{"id":"openstack%2Frequirements~696297","triplet_id":"openstack%2Frequirements~master~I8c021b04448dd9acbd3b3b0bc7e976889a266f1c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8c021b04448dd9acbd3b3b0bc7e976889a266f1c","subject":"update constraint for osc-placement to new release 1.8.0","status":"MERGED","created":"2019-11-27 12:27:09.000000000","updated":"2019-11-27 18:27:53.000000000","submitted":"2019-11-27 18:26:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696297-1574879181878-81677534","meta_rev_id":"8b6ae8e751820ea21f47a724d51bd36bcd5b3b88","_number":696297,"virtual_id_number":696297,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 18:26:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-27 16:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-27 16:45:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 16:45:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-27 18:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30ff242f41465f5e1eaa62fe106a96543772504d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-27 12:27:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"115904f73bf854197ac1e66c5235630b6c12bcf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:09:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56f7766dc8d4b13bf0dfbdde810c202 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ef8f28b4911e479a8d92fb755b221dc7 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a902d64d2f3487abf1a06b38a993510 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e48afbd5e7a84c0dbb80be4275b5dd86 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea723af2b4d949a396440ee8e704e5a2 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c09bce8f6e3440eaa523d0194a386d2 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17f104fb2d944275a39fa31524080902 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4841235e25ee4a53a05b007d1d44e5ff : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a65f9de916124c90a17bd0c564d81605 : SUCCESS in 10m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d9d0b5707cd44248a909b2a27feb1d1 : SUCCESS in 9m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/234e2093801d4b9e80a835797f1aff51 : SUCCESS in 10m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7984998ee2b3445483c794ad75f0dce7 : SUCCESS in 8m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/266ba69504d94cb7b20effef4c7ff488 : SUCCESS in 4m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0126f36d79b34709b473c737c4f1bf40 : SUCCESS in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6c8c91fe3f6345ae916ad000342116de : SUCCESS in 11m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d07c8fcd7d5b4024bd21ac24dc7d0131 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e576233d17a4f40b9ecac8755cfe45d : SUCCESS in 7m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9a6538b7642748c88aad8e19aeb6ada3 : SUCCESS in 13m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9583a5abbb3e47dd954cc587e705fd3c : SUCCESS in 19m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9f5ba62164ad4b77aa83a8d55bec7176 : SUCCESS in 19m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b13613ac70c4c729b1c93e8fff3977f : SUCCESS in 20m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7d15a93f6bb2487a8b028574988902cf : SUCCESS in 13m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37234c4182ab436182e1d0241c7258f1 : SUCCESS in 14m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/48223e815ede4d04a9767baf70be2288 : SUCCESS in 13m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba22d44857bb45889aed0d53afdeeecc : SUCCESS in 21m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bb889ac9eb444453985d7b64c80d0000 : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7ccb85849d0742458db8b1b56e34ef44 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ac4f946a7234726b26f2d8f2f0bdb78 : SUCCESS in 8m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/532ef3ff4680427683602199af2c7c57 : SUCCESS in 11m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f28226a66751434ea69b63432313b268 : SUCCESS in 8m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa3e1058669c4eb898aa4aef9739b46f : SUCCESS in 1h 16m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f21b50c8fa2483aa0f876d82cc7805b : SUCCESS in 1h 33m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e26abe5fd93a4963ba5f1bfeade010ca : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"105ab27c6819921bb9c07f9d69e28344ab211643","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 16:45:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93ca49cd22c7d35e1fb8aade0ae306fcfaad24d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 16:45:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a767fe8b1ff1f3219b4c45f0a8325ae0eee4301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:26:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d446114d7ca41e1aa0e619d4e756e1c : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda1994e1f5441cc87efda3027e9d3e3 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b632121a6584214a4abd3ebafb7656e : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18c4d904b3714a2cb313b12129c55699 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b97cef91ca5743b69a182e9f48b2f581 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a03cba135c1407e91edaa75ec29be28 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e54ceb6735634b119cc50eb1f1288ec6 : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c79d6d727507419cb9fd6b98b4e410ca : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db8eace525234199bb230f46fbc191cb : SUCCESS in 6m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/347062c564f14b3b9b3ee99ded116d8f : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0bcf08e1a5174d19bc29eccd34455365 : SUCCESS in 9m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1dd47077f7a9404bb682c442a2ced308 : SUCCESS in 6m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cfea482fb17b4429a98286f1c84b6b91 : SUCCESS in 6m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9aebf6a146d344d98301058d7d04fd6c : SUCCESS in 13m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4433aef75644b859203fd724a98f934 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7344211fa18341468135c05ce43d49b4 : SUCCESS in 8m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e4ce542c5ce94199a27451fbc8afef2d : SUCCESS in 15m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ba3bb34a4484844b16be0a3afc8f7ea : SUCCESS in 15m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/19f1b1f545154fc8bf0bf6eb1e45a180 : SUCCESS in 24m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b35139cf9694804b38b3253bf98bcf4 : SUCCESS in 23m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/81e63ea683884c38bf06d59466028c6a : SUCCESS in 24m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dfd9008d523242e895c7b09a390c6b66 : SUCCESS in 16m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2c117a9bfc6742fdb86fe5cc2543a870 : SUCCESS in 20m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/faba39d8a9f746739de3123f7d7931e5 : SUCCESS in 21m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3524ac0a0c65419892ca6e25e40d8255 : SUCCESS in 4m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d9d6ea4d858648f9b57157bbf72259ad : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4258719af3d74874b19605fc314d8e39 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6c877e8aeb0e4463a25b80e7b562677d : SUCCESS in 11m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc17741e015a42049c84b6071015930a : SUCCESS in 10m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5148c368d59c4924a6ce21ade8a58e83 : SUCCESS in 1h 01m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b3ebb4db05e464a88aeef9b8d4f49b6 : SUCCESS in 1h 26m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f001aa9ddab54714903fcd63662d6345 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7aa97476b803cb17850818dee40a4a62783ccac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:26:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf1860b23807a13e76dc2da3ee36538f7ed804e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:27:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32e5f596ac14e9e9baf777ca0a5301a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e9f87301b85799a3e156c0439e70a3eea0aa64c","revisions":{"8e9f87301b85799a3e156c0439e70a3eea0aa64c":{"kind":"REWORK","_number":1,"created":"2019-11-27 12:27:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/696297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/696297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/696297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/696297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/696297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/696297/1"}}},"commit":{"parents":[{"commit":"d8d2816c2c90d0833ee9cb043e332909679222cd","subject":"update constraint for aodhclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8d2816c2c90d0833ee9cb043e332909679222cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 12:27:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 12:27:06.000000000","tz":0},"subject":"update constraint for osc-placement to new release 1.8.0","message":"update constraint for osc-placement to new release 1.8.0\n\nChange-Id: I8c021b04448dd9acbd3b3b0bc7e976889a266f1c\nmeta:version: 1.8.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I1020bfa5b1b1c82c25cdc15c14d0d63bbabe035b\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+1: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nmeta:release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e9f87301b85799a3e156c0439e70a3eea0aa64c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e9f87301b85799a3e156c0439e70a3eea0aa64c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
