)]}'
{"id":"openstack%2Frequirements~696399","triplet_id":"openstack%2Frequirements~master~I8d34c609df176b78cf7c37d0e58edaeee6476476","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8d34c609df176b78cf7c37d0e58edaeee6476476","subject":"update constraint for tripleo-common to new release 12.0.0","status":"MERGED","created":"2019-11-27 19:44:44.000000000","updated":"2019-11-28 01:46:11.000000000","submitted":"2019-11-28 01:44:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696399-1574905488620-a496186d","meta_rev_id":"427f3aa100a63fed430c321363b9a92d77ef3bac","_number":696399,"virtual_id_number":696399,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 01:44:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-27 23:55:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-27 23:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 23:55:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-28 01:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f95ec0c4c1ea2bceb863c4fd8af44cfa47f922a5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-27 19:44:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"964ed541d8778e77e25173b026be64113e8d63cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 22:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37351694af0d441b913c21725c496434 : SUCCESS in 3m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2949b34a8bbc416d9f6cb555aa4ab334 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e5794c77b10490697d6317aafc745b2 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8b9889856c544ffa0171618edc93d66 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1187fd57d2e44a4584c699953b2bf243 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e0f25e67cf84343931c6fd3b7216066 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38bfa208b4994de7a86defec958d4417 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/864d654be4484a7b9be9f19aae50723c : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/eefc1f574ec4449ea01f65fbac6e55cd : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/83d2b50929324edea189485a9eb659a3 : SUCCESS in 6m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2e559b59873f48b08e9830ef0b883a3f : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a9d57fe695214f84b89f529f45727aa9 : SUCCESS in 9m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/74782193fabe44b0a05f87665e36a960 : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c8f6a30f6bd348079c8fde963cead0a6 : SUCCESS in 5m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c18636a198574cf299f23903311709a3 : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cebdfeb85bb34c44813fe865ed997644 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/265fff7e8f624a6ba7f36e9ac4ce99e5 : SUCCESS in 7m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2dff866d7eca4bb3b1e2cfa565065ad6 : SUCCESS in 14m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dc54331c087a4b89b34ef81686c4dad8 : SUCCESS in 15m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/beb3ba4f931b412eb4c65e016b5dc85e : SUCCESS in 24m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2eff835553c04c67ae2e55c4512b211d : SUCCESS in 22m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9da747c6a1eb4fc3bcfe528565b9fff7 : SUCCESS in 15m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f12e93a3afd54b4ea91060d581cb986a : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/64f12c7490be48ac9dc1d4134dfe13f7 : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40730b6b413d4f6caf19abbf1126dd56 : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8175a33c5b8545aab531847f4df7093a : SUCCESS in 4m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b2d8d78ad99246e5916b78ab3590d0db : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd96a8a383e74f14b22b90c4d100d1f0 : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/86b04ffe60354274acfb2d6189edcd68 : SUCCESS in 10m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/efa19312ff5d4752a92654025d181655 : SUCCESS in 10m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e56b525c67d84411830170f1ae58449a : SUCCESS in 1h 14m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/386621c12ac743058bf564204d7efd81 : SUCCESS in 2h 01m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eafc7dd457844bf0b48630a14818f4c0 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b1fe6f15a5bc5a57a6a5bb574095cce56babd65d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-27 23:55:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cff664e550364e6b511b148d9060292aac11fe5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 23:55:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b355bdab58e867908292600397ef14ca4e11cb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 01:44:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09ccaaa12504b829bd9b1fda8ddb662 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3845c41cb5742f7b128a1f12535889f : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14876f0ce0f440c096288b5accef4ac8 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/846cc93a64444eac8fd86601e66a252b : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ce6eb501b9344fd96283c47b3abdd15 : SUCCESS in 2m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fadd2d6a03142b8b11bc86e301a9e48 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d67fd6ecc5bd47259b665b5ac2e90271 : SUCCESS in 3m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/719ee8c27b444772b54c4da7ae45bd10 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2704421b2b8347ca88e6c838ca177be2 : SUCCESS in 6m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/dd28fdbc88254ef7a0ead20bf3eae2db : SUCCESS in 8m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb21717310724b489082cb52e428aa94 : SUCCESS in 7m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/22c700083063430d8d485a6b4cc3af5c : SUCCESS in 3m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e9ede18de52844c890e223a9b8df359f : SUCCESS in 4m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/47a805b9bff7462597125e294dc71f55 : SUCCESS in 8m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bab4a03c4c7549c1a3fceaaccd071369 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbd7029941214acb8f23ee57742bdcd1 : SUCCESS in 5m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c091e4ae421f4d938445322650269e41 : SUCCESS in 18m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/593ffaf5f85f4bc68b28dcb47e68f334 : SUCCESS in 13m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9290ee48752b47f5aa61b00312e63f96 : SUCCESS in 18m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2c20ccd7c2af448187a9d858e640b20e : SUCCESS in 20m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/433b622cae63490280ee2029a719443a : SUCCESS in 11m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8d971db6ff64dfa8fdce75b9c81a9c5 : SUCCESS in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/53b469c102d64bd999054c92f9ca8ab8 : SUCCESS in 12m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f0adf3fd0604c1e8b4d6fe6d1f2c23b : SUCCESS in 16m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/32b5bd7daa0c4db69e7ba8a4ebf0d79f : SUCCESS in 3m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/11947180fc5b47cdaba3b26217c6fd3b : SUCCESS in 3m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00f9366ec2da4974bb63e5f248d8728a : SUCCESS in 5m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/719303881cc3497588f47e9bbd46543b : SUCCESS in 8m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b866da2206d9400ab5f0399d6f3a16a9 : SUCCESS in 14m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54bee88432de45368882b81c387dcdf9 : SUCCESS in 1h 04m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/25615ece11c04171bd1d9c94b0e510c8 : SUCCESS in 1h 20m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/834b613427d44bb9b6af56be6c3146c3 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a592f3668257df30313099179be2bf6099280316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 01:44:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d94b816134f47d4497375b4c5ccdc18517bb2476","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 01:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78be1e613feb4e33a036d049ae46f3df : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66cbf322585dca60b5a2216ca57ae535daa14cfc","revisions":{"66cbf322585dca60b5a2216ca57ae535daa14cfc":{"kind":"REWORK","_number":1,"created":"2019-11-27 19:44:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/696399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/696399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/696399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/696399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/696399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/696399/1"}}},"commit":{"parents":[{"commit":"8fcfbf4857135ce17a7aad8d9e545df657dc479b","subject":"Merge \"update constraint for osc-placement to new release 1.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fcfbf4857135ce17a7aad8d9e545df657dc479b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 19:44:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-27 19:44:41.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.0.0","message":"update constraint for tripleo-common to new release 12.0.0\n\nChange-Id: I8d34c609df176b78cf7c37d0e58edaeee6476476\nmeta:version: 12.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Co-Authored-By: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Co-Authored-By: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I02e38bcf10b8eea56bb40511868d862f3db70bbc\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66cbf322585dca60b5a2216ca57ae535daa14cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66cbf322585dca60b5a2216ca57ae535daa14cfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
