)]}'
{"id":"openstack%2Frequirements~696747","triplet_id":"openstack%2Frequirements~master~I8ff4ec574eb5160806f93a869f65df1d2b2ada38","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8ff4ec574eb5160806f93a869f65df1d2b2ada38","subject":"update constraint for openstacksdk to new release 0.38.0","status":"MERGED","created":"2019-11-29 17:20:16.000000000","updated":"2019-11-29 21:44:11.000000000","submitted":"2019-11-29 21:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696747-1575063778606-98394890","meta_rev_id":"8313b3b7b9597214851cb4e7197045a3beb848ba","_number":696747,"virtual_id_number":696747,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-29 21:42:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-29 20:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-29 20:08:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-29 20:08:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-29 21:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71ff4b26edd37636119056fd9a019e9249c2b313","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-29 17:20:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6702a6211ebbb557ce105daba39b7a924dec6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 18:49:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ba36134926454485e13827d7f904df : SUCCESS in 3m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a5587401b3054b3095161650fd9b9104 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e87302cc69c41de8cf8453252286510 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b6050c3de8f470aa36792fe8c080639 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68c77723d0c74e1ea75339a4739605f1 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/077a62e75b8540f89808c0741fd3e92f : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3539b410f9b4292a36db77a85226aba : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/560d19ca51824ac592c0d10e3e51e535 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/045faecf05ea49cc9da4a5c64a5b7c07 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/778ecd5b303e4e1f9e47e95226f5a98a : SUCCESS in 7m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/48d1bc0b6a3c49c28645e7317ff34f76 : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c1116c7097d7499b8c03c098e9311d03 : SUCCESS in 12m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6fe90cc54181466784d83cc6e8fb3920 : SUCCESS in 5m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b1b9113a0b844b7db4b37c3b508f95ae : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c8840c1894cf4267bf960448df6cd5cf : SUCCESS in 10m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d2464c1a3b04b2cb6339c61c00c09d5 : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0aa7c3f55b2415d91e8f3c43f6a7716 : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7a9d156f97cf428a9be74850617c865c : SUCCESS in 18m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bba3487964954efea40c4e0321ac11d8 : SUCCESS in 13m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f1e069ce516f42a78953bd0acc0aa4d3 : SUCCESS in 23m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/60d6e3d61e26408ab6503264997f33d6 : SUCCESS in 22m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/139d2380822f4f2ea0db5d15309fc6db : SUCCESS in 13m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/974c9a24d8954c329fde2a780f3f2aa5 : SUCCESS in 12m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d7352f153024cbd9290008966ea9507 : SUCCESS in 13m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f572018a9a0496ea9b54ab181638831 : SUCCESS in 16m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fac6f4c694574adc90da66970bb9e4b6 : SUCCESS in 3m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a09aad764d7b45f48920e7b0b0bdfd7a : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9259c306a81546dcb399bef4f8cc4a28 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6bd6a44e962540b6bedbf3e6cd9b9be4 : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b278b9d7766490bbcd28c60e461d75b : SUCCESS in 9m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f7dd27e012b49ffbb7feaadb556dc40 : SUCCESS in 1h 21m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e50742a743df4cc1993ab6b66b6db81b : SUCCESS in 1h 19m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/76bccf33a4e74cd398a9ec81b3d96e63 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"119d13482c825e2937bd4df550f680f437ca7b1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-29 20:08:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e2af12e0deef64467afcd837098d9b487e125d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 20:08:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c4e01a9c23614228816816af56442596f8c3dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 21:42:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1868d3e2b8b4fbdbc0d76ffb1b95c40 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94b50f4647884cbda3f883171a2607e4 : SUCCESS in 2m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4573bce3553424690ee3fa78d27f860 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f8abeabb5d45ec8b68d58dcbc47af8 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e08d306809a94a28afe6d71005a51ea4 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cafa3b8af50f4cafb8f4319633758505 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c71c750cf8a41c691d5b1d4d8e26ecd : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fd3e8c2f5b27488f830e93b1af413d12 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/989bd179616547cda3a93fb4533277b6 : SUCCESS in 7m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ecbf2095c5834b948176fa4828550259 : SUCCESS in 10m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d5d0386c0fb447beb4bbce45721d8182 : SUCCESS in 7m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e034316b3ef24453b868a937f3e0ec1e : SUCCESS in 4m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dcccb745c6cb4bc18465819f7da5f7ce : SUCCESS in 4m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/56f51f26703f40d19ce8b83da01c7b9f : SUCCESS in 8m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c58d9902cc104be4948f585ee4693010 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90e1c0537363496c954397db2fdf7032 : SUCCESS in 5m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cb453d8fab4b44ef88e31ce1987dbb40 : SUCCESS in 17m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8168ea1c7cc54a48933cd33f3c9b4a58 : SUCCESS in 12m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b262eba7150d45668f780a88cfe2e386 : SUCCESS in 21m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/251f0fb2fe9e49148592a2cb9b55be7b : SUCCESS in 22m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0325bf538dde47dab3c3334f302f0872 : SUCCESS in 12m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e014d889eb64858b03859caf2d84415 : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aa5dede07aa74ab5b8392e4ccf8bb7f5 : SUCCESS in 12m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed08906694ef499a8a8f632f5cc593c2 : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c8f730c11e334cf5bb69f90d7d267752 : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/15dcfed1bd514670820920fb8ea96def : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93550999179d477bad98982145a2a12f : SUCCESS in 5m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/371ecbe8009441a2a9bf12b8672765d9 : SUCCESS in 8m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8094d18577c44b4588097a22834e1fa1 : SUCCESS in 9m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b567e02492714e009b0970a24174e128 : SUCCESS in 1h 09m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a08fba7f065d4b3ba595bb3765c71f03 : SUCCESS in 1h 28m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e12d18d15f794f9d92a8e6d8e88f32ea : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"969c7f8f5a4455c99dd504fd4993a9870cbbc2e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 21:42:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9994052026b4b9d253b078ec7a151b29099af7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 21:44:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67329df250b24b868289a18c14e16879 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2","revisions":{"5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2":{"kind":"REWORK","_number":1,"created":"2019-11-29 17:20:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/696747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/696747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/696747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/696747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/696747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/696747/1"}}},"commit":{"parents":[{"commit":"8881ba47923b7091496589619d3bc075df3f50fc","subject":"Merge \"update constraint for os-net-config to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8881ba47923b7091496589619d3bc075df3f50fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-29 17:20:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-29 17:20:11.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.38.0","message":"update constraint for openstacksdk to new release 0.38.0\n\nChange-Id: I8ff4ec574eb5160806f93a869f65df1d2b2ada38\nmeta:version: 0.38.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: I9bd287974c52fd32102fd1c1c62f662f478f0d11\nmeta:release:Code-Review+1: Adam Harwell \u003cflux.adam@gmail.com\u003e\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta:release:Code-Review+1: Erik Olof Gunnar Andersson \u003ceandersson@blizzard.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
