)]}'
{"id":"openstack%2Frequirements~696769","triplet_id":"openstack%2Frequirements~master~Iac9156fdc0385806efef6b7c3209ac20aebf8b3b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Iac9156fdc0385806efef6b7c3209ac20aebf8b3b","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-11-30 06:10:54.000000000","updated":"2019-11-30 20:30:42.000000000","submitted":"2019-11-30 20:29:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696769-1575145777538-eb3ba9e4","meta_rev_id":"145c2bd3345962f69f62eec2ee7783e75f316c94","_number":696769,"virtual_id_number":696769,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-30 20:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-11-30 18:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-11-30 18:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-30 18:45:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-30 20:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6eddde0dcd2cf0521a44e4cbfad0acdde6e1073","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-30 06:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb8dab5b2e88b06b8dbd85b3e0920fb1fc3b54a1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-30 06:27:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89c5067b6725cf8d24991ec079428b07e6200afb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-30 06:27:32.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"a84252486fd42e6551e01eb32a194912d1979aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 08:27:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833d02a4ef784200a0173ab55d1d7917 : SUCCESS in 4m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1f1827b31cb943caaf8983c0b4868e67 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70850f37826745cf840cbbfe8ef15f46 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7721fbaa187f450e81b985a6ae809c1b : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d810cd42abca4dafb0537af873dfd700 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8296a6576b4443c0a320b18d7fd689eb : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95ae70b3b528494baf632207793c0e6c : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2885a224efb44e63ab68fc87b44d3175 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d93cb3d303b1492d8ac7dcf7b537051b : SUCCESS in 7m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/252cc614224c49e6a78bc9c811e630e5 : SUCCESS in 10m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8a37470ba36a4605b04b334f1ef8f983 : SUCCESS in 10m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/95ca126873914a27adaceefea21f15a1 : SUCCESS in 8m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/852a89f463a1443a98bf7a9f8d7438a7 : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/55f8184d4f3e4fe894a0dc12e2508baf : SUCCESS in 5m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7cb4ec8746a645a09fe101811ee12063 : SUCCESS in 9m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfc268f9f9394a44ae9dfa2ee4ebb721 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5022957b82504466b4206040d7eb14b1 : SUCCESS in 6m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/15dbf9c354c84cd093a862aaab70aebc : SUCCESS in 14m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/72c268ad2d204b579c5f203ab01f83a0 : SUCCESS in 13m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/85928dfa383d45dab6068253022fa26f : SUCCESS in 21m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c5175b24c5cd49a597d904bfb1fad430 : SUCCESS in 22m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/93e691ddd14746c0bb89c81966315f49 : SUCCESS in 12m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68681ea9f3f14622897f5b70cd6b28db : SUCCESS in 16m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7528a51c2a5b4ce889ae4da43facdb6c : SUCCESS in 12m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ffbc86265bc41aa84f8bc1b914045ba : SUCCESS in 17m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/03039afe6ab34d90aa59c61e27161767 : SUCCESS in 4m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4f17144e4bb94ad7963274ed0a96fd2d : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5243397a1a741b98b8f3499cc058963 : SUCCESS in 6m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0240eda3da614a8690044a78a4f65532 : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e3e68b431894b1fa4771a2ad0be7456 : SUCCESS in 9m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e5a54ad1f794ad2b4e4dfeaf5347a50 : SUCCESS in 1h 37m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/05bbb970bcc44e81b96838f8951eec3b : SUCCESS in 1h 52m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/be4073b4ae0e4eb3991abb3aed0bc742 : SUCCESS in 1h 30m 18s","accounts_in_message":[],"_revision_number":2},{"id":"9a0e4e4f89b19ec88a347fb264680c12553b71d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-30 18:45:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a08674a44132101a6682ab530815c2783f63e710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 18:46:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14611be7b241a0aafb03ec3e035bc75ec5967648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 20:29:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8216731add7b42fd9db0cf48ba2aefa3 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d441f7b0804f45aeb42b5816d25325 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a4d046f7224480084e353809d4e830c : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f95ef44fb6a748299bf9be58a2dc08dd : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc5f9fc426cc4490bcbd0396cdb98342 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/302ca91d843a4f9780725969765c13c6 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae4766464a0645a7b60e4e16b3e7a0ce : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dcf6365d9e7c40c6ad46eff2c29cf872 : SUCCESS in 7m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/282b68e9477f4172a7ed46d26965efa1 : SUCCESS in 6m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8f9cd7ffecfc49a88bd4a29b0a889a2e : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59092d3ea5c342e3bb48085ca149155b : SUCCESS in 8m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5536ce078af84ce39a4552b5d7fa59a4 : SUCCESS in 4m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/164bee280bca40d48993704305b34263 : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4fa374391dc3476684d686e1ebcb081c : SUCCESS in 8m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9abee2c43ae84d049732b4568d93cbbd : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/879842a3cd2b48e191f98fadef1cf9c7 : SUCCESS in 7m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a38bec610f894a8284667f293bca2731 : SUCCESS in 18m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dfd499021ce34a45bc5ad7d36f0aaef4 : SUCCESS in 17m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4ec6cf76ef88476aae7a93c289849b4b : SUCCESS in 18m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec6f3e703482497487cc1e610d493bbd : SUCCESS in 19m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c5c1e278a14547c4b558d31f059a8fa3 : SUCCESS in 12m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a20e02d2135413dbf090a303c025754 : SUCCESS in 11m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/40badeeed36245e68a4b9974947e9c19 : SUCCESS in 13m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79ca46d6cc244b69ba73a80d0b7b923a : SUCCESS in 13m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/08fdefe98f2d47709bb1eefedc12ef1a : SUCCESS in 3m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e08d6bb649304cb7a98777b7842e3680 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48ca6ccc02dc4b359b4f143ae78819a2 : SUCCESS in 5m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/751e9774049c4bcbad7c2f5f50befa46 : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66665d3636a54ebdb3b5d3944bf112ba : SUCCESS in 9m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99b207ee85574f7c8949dddaff9ffd45 : SUCCESS in 1h 03m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e1860e27678e497f84104139ca186faf : SUCCESS in 1h 27m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f3662b8c2d584b0e8af21577e7d1b444 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":2},{"id":"738fd812f35827c19ee5c7b2a9dfc9962b026c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 20:29:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"145c2bd3345962f69f62eec2ee7783e75f316c94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 20:30:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8974876cd71a4eb7b2b5a4e160f46243 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc1fb3c45a24f352041ac636750adde99163acb9","revisions":{"5acccbb854fcc8d870937ea040f9de81b8ed94b0":{"kind":"REWORK","_number":1,"created":"2019-11-30 06:10:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/696769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/696769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/696769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/696769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/696769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/696769/1"}}},"commit":{"parents":[{"commit":"5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2","subject":"update constraint for openstacksdk to new release 0.38.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-30 06:10:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-30 06:10:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iac9156fdc0385806efef6b7c3209ac20aebf8b3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5acccbb854fcc8d870937ea040f9de81b8ed94b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5acccbb854fcc8d870937ea040f9de81b8ed94b0"}]},"branch":"refs/heads/master"},"cc1fb3c45a24f352041ac636750adde99163acb9":{"kind":"REWORK","_number":2,"created":"2019-11-30 06:27:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/69/696769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/696769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/696769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/696769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/696769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/696769/2"}}},"commit":{"parents":[{"commit":"5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2","subject":"update constraint for openstacksdk to new release 0.38.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c1eb9f9b81ac5eb59b33903e742d303eb57dbb2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-30 06:10:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-11-30 06:27:17.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iac9156fdc0385806efef6b7c3209ac20aebf8b3b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc1fb3c45a24f352041ac636750adde99163acb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc1fb3c45a24f352041ac636750adde99163acb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
