)]}'
{"id":"openstack%2Frequirements~696785","triplet_id":"openstack%2Frequirements~master~Iea823c851a5f1eadcf0eee10a58e3bed1fb98bfc","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Iea823c851a5f1eadcf0eee10a58e3bed1fb98bfc","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-12-01 06:18:17.000000000","updated":"2019-12-01 18:38:21.000000000","submitted":"2019-12-01 18:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696785-1575225427487-22ec06f9","meta_rev_id":"fdab5441ba20564bc55a6891de232a191408331c","_number":696785,"virtual_id_number":696785,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-01 18:37:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-01 08:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-01 08:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-01 08:36:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-01 18:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ed42a961121026d72561fc60ad9d88a3d719c91","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-01 06:18:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fde5a3b4aa77d8f6b79f7e7115b80ce19c74b64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 08:02:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472a7d1aee0a4adbac2c54d9785596a9 : SUCCESS in 3m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9fc4d8f1e8da4b6aa116eddcbda30e2e : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4767338f72694493baa3f8b81171f20a : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfb9dfd244254c77a289cc2c1eaed3ca : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8caedd18fa884e32ad3a00fcff3329fb : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bec1bea75ac4450a95482583c6216f7 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0fb8f8dcc06e44448210c4e4d600f75a : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ee628440f9e4867b27c04fd9506be31 : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d2876415ce894d1787d27b1835b3fe61 : FAILURE in 7m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9c3d39612ee4586ac4d751d30c8cdfe : FAILURE in 8m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e03d088374314d2c998ce7e29905aae9 : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bad62790d69b4f14b345323a7d7b8d54 : SUCCESS in 7m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e45750e7140846959fce64f47a3194e9 : SUCCESS in 4m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa426ec05e07467e9b7699d099fd9f6b : SUCCESS in 6m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/135fad13e18549e191fbff3fc328df24 : SUCCESS in 11m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a150a41613ca4be6a0f9d5e343756c20 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92837c4538144fe3900df3f358c3fd5a : SUCCESS in 7m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f32be68a10c644b694452adae95992f1 : SUCCESS in 13m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9698de74eef84080a57f74bb6c8ab37a : SUCCESS in 19m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f495f46bc3f94cedb6651e8a72207edc : SUCCESS in 25m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/25d5569299bd4b568fac50e55c1ad825 : SUCCESS in 24m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ca2386d3f2ab44dc87944e471b65df43 : SUCCESS in 14m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8c0436d7afba420dab9defd4e6863e5b : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8a80e3860ebc4f2bae1a6c69454e5204 : SUCCESS in 13m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1f3919dc028441ab43792b55a73e3b8 : SUCCESS in 21m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f42b19dce8e34eaf95c975b1b2d823cf : SUCCESS in 3m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8f5e623d7ded412882b2e19c51e8ed87 : SUCCESS in 3m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f3641a1f7e7464f8e9dee6bbfb1ab8e : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6dd0e6d943744c2ab09d0fcee8cf5886 : SUCCESS in 13m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a180d5dcad447af8788600d46a280ee : SUCCESS in 9m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74b0d89fd3be49f4957ffbad9c9da4f8 : SUCCESS in 1h 13m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0932864d35254010b8dfe6737b41e24b : SUCCESS in 1h 30m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/87ad504d88a24e328f19b79522564886 : FAILURE in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0f910e15b4589c5b4327ef1487f11c15976b9156","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-01 08:36:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b66318b8be5d87cf65f40784a3472efa783100af","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-01 08:36:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"840f6f11546d4ea576f84d2be160f98cbcddce2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 10:01:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db3ad90034c4d55b1a49b96c22c6439 : SUCCESS in 3m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1345a319e60e4a0f999f86edc6d1d3ed : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fd84c62afe4f42ad6a87659cfc2825 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0328434a016145268804647105d373e1 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27aa6f2d31464e3b8dba5597e00ba950 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c47accdb132e42d9a425e80b39146fff : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b616280e51b5469482dfaff659c002a4 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01e0bfedf3d546648e07c2826bd68495 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/36ffb4a8de3e4d45a159d24850ab06cb : SUCCESS in 8m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e482c4577514e68aa89ce6a9336eaed : SUCCESS in 7m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8eec4501b3034b6797ce8c088820b1dc : SUCCESS in 8m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ec9a7731ee7d4fac91fe8aef380619ad : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bb31a5deb02d444881c6e63c65594cac : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b57c0580a344a4899ecbda454fb939b : SUCCESS in 4m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c2a0f6b87e534e0eaa010fee4bd924c1 : SUCCESS in 8m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6fc3742c0bb4e8eb482d653733676e9 : SUCCESS in 7m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/694b28e93cbc4080aa427dd07c6e6bc3 : SUCCESS in 6m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/876a00768bda4aad94905b13d558e44d : SUCCESS in 17m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44701eb018444952aa3e79635a6de6f6 : SUCCESS in 48m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/741fa3064f92492f970e1ac047917b81 : SUCCESS in 21m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/555d54c012ec4f4eb6077c41994f9582 : SUCCESS in 18m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8e720d7a73254153957f7a7e251ef56a : SUCCESS in 12m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1e46ae2b24e24ccd8f71aedcf7af36c6 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b2eb09e145284099ad7e201b3b87066e : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48560f4345534399a8884ad6ce831865 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/66d7c91453364f04820e34ade9405b55 : SUCCESS in 3m 44s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4a36b5e74426466d8a4fe08696727c26 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bcc5e37e81644be80f0ca963c278056 : SUCCESS in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96e7f38bfefb4cd199f487e332a0bb66 : SUCCESS in 9m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/54509c6358d748e0bc70e15b2a05dbff : SUCCESS in 8m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c63a7ad05f0441ba2bc98c5fcff4969 : FAILURE in 57m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb4949a4985d4ae2b4cf29d5e4fb94f0 : SUCCESS in 1h 18m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ee7283e38ace44c792496c6c374c1581 : FAILURE in 1h 07m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6e7364b21a39304440888cfe0dcff356fbc4377e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-01 15:30:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"714c071a55253d206af1a0aef14a030116371c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 17:09:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8377dd56f0cb478b9bf88ef4fba7ffa1 : SUCCESS in 3m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/09acbaa17d814ae9a326291aaeab09d8 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f03380919747e08e0b4543e50e003e : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18f01461c17145ec826d290af08ad5f0 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ff9a3857cd644718fa8bf883e18dcb0 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75e190054d154c1997718ee216f34016 : SUCCESS in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4c0e27a383c4e11b3dc33715f9a2a02 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46ca1e4557954f3d9f936edb12d2f092 : SUCCESS in 2m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d1564095ee784ff0911ccef8f5f2c8be : SUCCESS in 7m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/05ef2fe0933d4653a06e9ee61a5ce10b : SUCCESS in 6m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c1afa6a7b54546bdbdf90cb3325773dc : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8bec64b0b7d0461a9c474c108e084ae9 : SUCCESS in 7m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1eb1173d14b4456eac25f1880f107a8e : SUCCESS in 4m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/051ef67811d746c3a1a4d1c90cb5b0d1 : SUCCESS in 3m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7df06564968b4d21b995908c51a5c0f4 : SUCCESS in 10m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8bafab87feb242cf8566ffa69af9793f : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/458c8bf08e244859a4bdabfca24d7696 : SUCCESS in 5m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7c5d5d42df534daa904dee7dd96d87a2 : SUCCESS in 21m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/73429d731bba4196b0753a19bf0253c4 : SUCCESS in 20m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6c634cbe6d9b43ecac238935cde338fe : SUCCESS in 31m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1eddc9cc175c47dea1b16bdfb8c6303c : SUCCESS in 31m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a257a89759a7488b9655714c3a4ace5b : SUCCESS in 11m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ecf80b3413e6474ba451c252f647484f : SUCCESS in 11m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6642ad42104a44638907788ca94a59e4 : SUCCESS in 13m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e5b9d6f91924932801abc8884dbdd21 : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5801aa0ae5ae4780bad067b77617de62 : SUCCESS in 3m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/642221942119457e813ae9955068e975 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef28a2fde4e94ee28b5664d76c104dde : SUCCESS in 4m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6eebe74d039146f5b4b48838a608989f : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0e7483f2b8be4bae8419f1f487df3e68 : SUCCESS in 8m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceb9bb8a62f5423eb58b95cabee3c209 : SUCCESS in 59m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4ad249b23ce4497b8663c33824f7b99 : SUCCESS in 1h 31m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9442443e73424553bd8510ff45d5b73c : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a040cee5cd8bc539f63e37f1580230c8861ec802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 17:09:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"41a372fb5bdba32a6d023c88f613bdaee0377e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 18:37:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3070fa077f5cb82c71cc7d0bddd69fba7894b221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 18:37:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ea6e05c89e476ebcf3abc882b2d1b8 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de3ae197d8b402a95e1b1996a3d470e : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47dd2623ecfd49a6b617c539d4ca5a64 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b88588052f3c44d9a63941613466e24c : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b80a24ab8304b1ab330a142cee4e415 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7499f9af0584108b2cbbd0f2d74d5e6 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c65b4a03c3694471a15b232a4ebbb726 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9861e39311fa42169b601df07e892655 : SUCCESS in 9m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b42d298b757477a88756dee5ee17f2e : SUCCESS in 8m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ea5694fcfd704f709c9a6881a5bbc4c5 : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/75b684bd08e9408297f230e24e8702e1 : SUCCESS in 8m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/014f061915a647768b83c5d1b1a40669 : SUCCESS in 4m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4c24c2fdeae456db7fadfe7982b52b1 : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ce533722f4134957a11a7fb453395120 : SUCCESS in 13m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3942557f719649849ecc4522e088a929 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f9adc5e639e46e689916d425e3ac3ad : SUCCESS in 10m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/756f215e9f00478983c9ac73985d0303 : SUCCESS in 24m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b06e5777eee45e88cf351e6af04ce3b : SUCCESS in 20m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/69c2d9890e6c481ba0c8ac5fa7df99a8 : SUCCESS in 19m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/57138a8db5214f18b08f063104aefc98 : SUCCESS in 19m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/531e9db7729e4409b4fbc3644157f713 : SUCCESS in 12m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/66694ed3b5d84eecb31a1b6949555019 : SUCCESS in 15m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/78b9bbd487044e469f6ac92c0beafc16 : SUCCESS in 13m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dda5a42476514ece99a15b8bcac44f00 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f5dd6a8f8af14487846c766ebf7c514f : SUCCESS in 4m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fd8cea1a001c441faa50ffbc4026ce2e : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55573ec5958d4da9b82853414da9244e : SUCCESS in 5m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3561384635694c0783f9064542d09251 : SUCCESS in 9m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a55533f1d4654be7a5c83964ebfa1cc6 : SUCCESS in 12m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea0491a10e38469e95f67f97b6b39194 : SUCCESS in 1h 05m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f9e9bbe42ad24b82b9d8370bec3372e8 : SUCCESS in 1h 19m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4ea69f6c6e1a47a9933d453db250571a : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":2},{"id":"32af1e80c9e45b381c1a8ba695fbeea5cdbdc8d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 18:38:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7dab90d3e148919f956d9a539a3cf6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ae758a91ded41aa5b88590de1abd5791e154f07","revisions":{"f01ed62a69b9a9d7d2378a7d98d2b24ad3c3e260":{"kind":"REWORK","_number":1,"created":"2019-12-01 06:18:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/696785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/696785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/696785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/696785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/696785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/696785/1"}}},"commit":{"parents":[{"commit":"cc1fb3c45a24f352041ac636750adde99163acb9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc1fb3c45a24f352041ac636750adde99163acb9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-01 06:18:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-01 06:18:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iea823c851a5f1eadcf0eee10a58e3bed1fb98bfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01ed62a69b9a9d7d2378a7d98d2b24ad3c3e260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01ed62a69b9a9d7d2378a7d98d2b24ad3c3e260"}]},"branch":"refs/heads/master"},"3ae758a91ded41aa5b88590de1abd5791e154f07":{"kind":"REWORK","_number":2,"created":"2019-12-01 08:36:05.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/85/696785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/696785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/696785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/696785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/696785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/696785/2"}}},"commit":{"parents":[{"commit":"cc1fb3c45a24f352041ac636750adde99163acb9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc1fb3c45a24f352041ac636750adde99163acb9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-01 06:18:11.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-01 08:35:35.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iea823c851a5f1eadcf0eee10a58e3bed1fb98bfc\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ae758a91ded41aa5b88590de1abd5791e154f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ae758a91ded41aa5b88590de1abd5791e154f07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
