)]}'
{"id":"openstack%2Frequirements~697003","triplet_id":"openstack%2Frequirements~master~I2a49dea72c49a2b5ed5fbe5e8e1a5e2e4be4ae92","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I2a49dea72c49a2b5ed5fbe5e8e1a5e2e4be4ae92","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-12-03 06:18:33.000000000","updated":"2019-12-04 03:45:47.000000000","submitted":"2019-12-04 03:44:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697003-1575431081257-656d7ee0","meta_rev_id":"3aafa891aa360a3a6fa2e0a67be7f77afcb656a2","_number":697003,"virtual_id_number":697003,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 03:44:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-03 06:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-03 06:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 06:40:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-04 03:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ccb92cf8276ebb80706e15feb0c852b57103169","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-03 06:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00c3b41beb3d3c670095ef5d87292756057e5e09","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-03 06:39:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6736c5034febfcfd252eb0164d52104b74584603","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-03 06:40:14.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"fd448fbbc97d5245974557c8b3564530e3779eed","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-03 06:40:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13d00296a1ff12895165ba5ff95a07c7d82a41a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 08:23:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9141ba76163140bbb8244b5ad60daa90 : SUCCESS in 5m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ec7913fcbdcc4de2b07023159dba7ad1 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54e8650998a40678f8a080334b6732e : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85ecb3911bea47258df8dc7ddf667dfc : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ff95951b8c246bc98056013a7194e9d : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ce2eeb1ad3a435ea1f86c59a1cb54ba : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9566ded99094faf9c7253cbb6587ced : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/601b8b6cd51b496cab5ae21db20baa3d : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/125fb6b57490455fa47941a95a2d8583 : SUCCESS in 8m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c1ae528dbabc4915a51b7d2a125750d9 : SUCCESS in 7m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a4a14a2eb9c4a928de6a93cffdf29f3 : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d77c163743d4a31b4341fd2a7870b20 : FAILURE in 11m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2ed55474636d479184bc961e22b76cc9 : SUCCESS in 5m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5982d5b41ce44b29b96f79455914cae5 : SUCCESS in 5m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8ad44cda204642d2a97a90e8610420cd : SUCCESS in 10m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/614f0cbbaca14330b2e321705452fe89 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/144e6f82336f4e2892fecc56f4611fb4 : SUCCESS in 8m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/02735fb63b004145b5917765b869261a : SUCCESS in 14m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a35aebac174549e490c7eb759aa4c191 : SUCCESS in 22m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4847636665d04a9e82a3903a4e7756ef : SUCCESS in 20m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/59956549b5eb4174a11cc8b096b13979 : SUCCESS in 24m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5ea90097ad4443218188a0c058101b11 : SUCCESS in 16m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5f6a9ef5440d4a5d85723b68d7567fe5 : SUCCESS in 13m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3b3f3da932df41e6aea0f526ad6f9d2a : SUCCESS in 15m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea060faeca794be0b80c380f39efbf0c : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/020ddb34f77e4b738bb7f69352729e06 : SUCCESS in 4m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/39da90dd85344ce0b965da8963239366 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54f339cf348f489498fe7dccefbed712 : SUCCESS in 6m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/95749c2b6a6a4540a143ce64b6e3e1ca : SUCCESS in 9m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc51fd994a2f47879cea68a1f388b9bd : SUCCESS in 14m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed80ea01eb3b481abb35921462ac1a23 : SUCCESS in 1h 07m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/41f5c2c0ee294dda82ab33e1102c7049 : SUCCESS in 1h 41m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5bb33f6f23114189866067a049496115 : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a9bf23c9067b921a2a7a66f2ed736b4c743f77dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-03 16:21:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d6ff66f02a78d4ddda58beb0abc2509d74df767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:04:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b0f089627f440bae04a7fb30b70f4e : SUCCESS in 3m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/84ee2857c9aa42de91302961c8f5d14b : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96ae9c687664b2c91263b2668c0d019 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b837cf31c0749caadca3fbd593b7f46 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a9f41f63a9b45398dbf611036c0f366 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35b9c60f5cd54c7196f5ee7a0eadfa4d : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a540df68931d4485bd1724285c318b9e : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c23ccd5642346a3ba5fd88faeff697b : SUCCESS in 4m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c3a36b81f1dc475a9dd256bfcaf62357 : SUCCESS in 8m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/808ddbf0a4d74c2cb27637ef992f632e : SUCCESS in 7m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5678fe26e8d84f4da0470a4b7da839b1 : SUCCESS in 11m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f12c06d6033c457a9bbfb6913d451cb8 : SUCCESS in 9m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3688761e367241b281685110cf810d7e : SUCCESS in 4m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8c36dc21bea74baa99a20ade2176f9d2 : SUCCESS in 4m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9a4f341b17644665b3c28ee8e114fd4b : SUCCESS in 13m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d4ce1aa4a5cd4a46a4a11e5f9d36f8a1 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66d2725a494043738db1b25464a36440 : SUCCESS in 7m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f7c0759b2bd7445f9426cf3732feef71 : SUCCESS in 14m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fe6dedf6f3624374b4c4cc6d82946e63 : SUCCESS in 23m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d0b070f0f54844d099f08706c291139a : SUCCESS in 27m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8e93bdf06d0745f29039cd95b13f4076 : SUCCESS in 23m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/319dc8da45c74292a90a75e512968d0e : SUCCESS in 15m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4720de980c37448585dcaee437956bb0 : SUCCESS in 16m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1d31ce0793f64f51a7f680aed74e942a : SUCCESS in 14m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/680014c2654b481792964de53daafbcf : SUCCESS in 22m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a1f7a83c3bda4a9f8a44b727ad4d8f68 : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/279744b17d6e41fd99eecaa2a2494cd9 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52b6be3edefd4c23a651fbfef65525ef : SUCCESS in 6m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b0ecd0794ad4ec08a46d6c20d3be18a : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8cbb0ed5f26540a1ba2062a538be7d05 : SUCCESS in 8m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7dc159d0093d4e79a27f20a10c88776d : SUCCESS in 1h 09m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0e358b37bd544e8b11d9269fa9350c9 : SUCCESS in 1h 18m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/25f7acae78494494bbc0b811b3e6381e : FAILURE in 37m 51s","accounts_in_message":[],"_revision_number":2},{"id":"17f1bf854ee637fd59a4c673c21889912f806c6e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-03 18:48:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"386f2df875c042c35d3c4fa600a6ea56bcdc0832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 21:17:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea09979b3a1421a8ad6beccc4f091d2 : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2f360e8c299a4b4bbdd7058686df5a3a : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e4cd9cb23441868898e49dd0a77b8d : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b92e9d4d16834090a4a2c34805ef2144 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd0995c43f34b6c8f4cda048c180b5c : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48e653e7973643698418dad1fe5aa7a7 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3a78aa2712449a1a16ef697c3371995 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e4799b338b741a3adcb9c4220ef2184 : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ef79c9d86d5740e9a1a6ecac88ca4f2f : SUCCESS in 8m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f633d1315254c7694907bc9c4c5084a : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e7006ed6a8dc492b93c5a510ea5e2172 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4fc79fcbaa714481b6aa766c8fe402e1 : SUCCESS in 11m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/992b4debcfae4cd2a1ee28c611bac56e : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b19c55e88a5745d4b567c78e96467a75 : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1c1870f730004762aa40bdcff5739066 : SUCCESS in 11m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/856e28013b814244a36c8f15f1ee1339 : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f43277abe77a46c99f7ba8a9b8618244 : SUCCESS in 6m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/592752e9c2f240c8b2c71ca62cc00a99 : SUCCESS in 13m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/343103c701e34ed3b6dd56373ff194ba : SUCCESS in 12m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c7b3e69ca71349639c0c85918d49d047 : SUCCESS in 23m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99ab6f1ad6364a6799d1be915620827f : SUCCESS in 21m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1cdaa43c61574b0485d5bda29d35ea38 : SUCCESS in 13m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/047013f515fd42199304411861dff324 : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/23e1229e4dcb4691840045dc1a7c2b71 : SUCCESS in 13m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b69746a73e8c41508d7ab68be640fb98 : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa9aebd0c20a4eb5bb606818c65a3e98 : SUCCESS in 4m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/90c87721ed2f407ab7ba225ad00bf3f5 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eccab7e7496e48bc83a0a2c9706cf95e : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/35c35052813b48d4adffea8eb36794f5 : SUCCESS in 9m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ae2bc0bdd21473cbe5d6b109727b589 : SUCCESS in 9m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/21e0c5cb0c7a4d8888001a651f603e72 : SUCCESS in 1h 18m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/560d8ec9ba6d404f9aedcb0a78fe3ee7 : SUCCESS in 2h 06m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ef9b9eee95df42df90420eb3d3bda5f9 : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":2},{"id":"520ebd26abdd726db26dc19f0534452d2d758d35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 21:17:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c24f4caa73c5d9d4f42d4517075131670750029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 23:38:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62052aef2634cb695abc62fa5e14028 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770319be10334de3b616a84cc9eb81dd : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a96dad3c24d74b37ad1e4c375efddc5c : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c479333195f408a80d1b3bf81448191 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e86d19572fe94c10ac279b52b5e5fb4f : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/323b735f10f04b73a0cdd3c20186d939 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3581e52bbfc64190956ab3aeeeb84730 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5afef84746da4e5f8ab08ab264853fcc : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2849beaaf0cf4fceb31971b05fe717df : SUCCESS in 7m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7db75d4fabcc447aaf66673c824a11ea : SUCCESS in 10m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ee242cf4a4147a6845246982920efef : SUCCESS in 9m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3000b172fcc941378fadd33c939e502b : SUCCESS in 5m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4764f7e5cc7743b886875e0589c5a285 : SUCCESS in 5m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b3ee619bf5d84e7cb4506cac0bc02d42 : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90abd4b428e247328130b5ec01097e2d : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/675b436663194c489e9ada8b057660cd : SUCCESS in 6m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/33e08d670e8e4628960799cfadf0f6c3 : SUCCESS in 17m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/62b4e3c9c6274cd493165cad79ffe0ca : SUCCESS in 13m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bf3bcf67270f4b66abcbed40058ed476 : SUCCESS in 20m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52e608110d9a4dcc857395db4380af8d : SUCCESS in 18m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6b80118a975a442b80c1c6cf31eb3f94 : SUCCESS in 14m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/991c14879fb8484c8001e3a67e6b348b : SUCCESS in 13m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/92322578a6134d5086ccfdd00357743f : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c7bd3367e5a4bc19eb1d6b34261a130 : SUCCESS in 15m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5206beebc9e0426d918192e66307add4 : SUCCESS in 4m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/eb6763e78a4f463e9cac61a915f23c49 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bb19bee77f74e688c10060f62c229f4 : SUCCESS in 7m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/990ba04eda644655a1548aaaf9912518 : SUCCESS in 11m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f3e5224102e403e861d185c9933fb13 : SUCCESS in 10m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/287a7f5ce8dd4f63931d69ac61991e5c : FAILURE in 1h 40m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f7740c9cf9a4f94bf49654c6faa52f6 : SUCCESS in 1h 33m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af1b26d85669453a8a50a00d90413cb9 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":2},{"id":"fb45afd3a91565deabd6edf0e6909179912d54fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-03 23:49:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03c2db45179a0edc63b5078a1d7cc5a4f61e03db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 01:54:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fddbd4c5772e475685e8b1954f50d75b : SUCCESS in 4m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/22cc78e3205f4e8a8a8bcfd0e9a933fe : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad6589408624ce38936ad39d072ea1d : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b856a6ec5927459ba9bbb5cd9621b27f : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b830754f2be14f68a451dbf50a10db7c : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/500559497d294822a8d35e6b7532210a : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69032171fb7041e39561303d95f7f1af : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80b2dd2385a746048decd5993211374d : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3969332cb5ec408aaa8a7bc84f191fa2 : SUCCESS in 8m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/998b7e2f059a4884ae7afc7a49ba4002 : SUCCESS in 11m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/513701574e9a487abd5807708709121a : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d82e387b0f243f39baf44d6518fb88a : SUCCESS in 12m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/52a022e17c2f45b1bddc57b4e8847e84 : SUCCESS in 10m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ee47e64bc3b4d07a7dcaeb827322190 : SUCCESS in 4m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cfc1dea8071e492096212c24324451c1 : SUCCESS in 9m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e70501d7883f4ef3b906d7c2b6124d02 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37f6bc4105194601bc0fec9ea9dcc224 : SUCCESS in 11m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1a61659412db4cab9ec632426445b876 : SUCCESS in 46m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5437ef63b6df4a168d70da1b9293502a : SUCCESS in 15m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0137f0d26ae940c3beb604f00201fa42 : SUCCESS in 25m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/389013bbe87c4777804ab61ff2d801a7 : SUCCESS in 25m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d24fbe48464a4cd28ffc3732f1e8a8d1 : SUCCESS in 13m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dfa5aebccba24cb9b1010513cd600ca5 : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e81541e146bd4ea3a9d9d6ff49059c10 : SUCCESS in 14m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f9e870b6a674eb28c7d3a39c12dc7b7 : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cce5c86ab01241df86aeaec95257ff5d : SUCCESS in 4m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3e34231908e843a28e1b0a71358fdac2 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f8baccfbf49414a90f5f6fd3cd01474 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f61511da9fcb40a0b329ae568e957b89 : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73338c1ffe604c24986eee321b4683b5 : SUCCESS in 10m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb3e45dcbe80411999d7a448bf2b4010 : SUCCESS in 1h 16m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c6140767b3144d6ea09e580c4bf506a7 : SUCCESS in 1h 50m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f3e7305808034ab0a58c7dd89a355c7c : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":2},{"id":"343e4c6b472e47408ab143afb1fb0948be1b4169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 01:55:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee887ad0e6a173714e212822ffc88d66cb5bb905","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:44:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"169172cbcd7b106ab9c4030e38bf802959268103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:44:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7381469f3614137969434f60fc896c6 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0f1a237a84424ab850854bc07e8416 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a635ded691154679b832c437677567a3 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfe9cb0d6f714d12a72f794a911e0556 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6cfd581de674e19875f511e2c74d1fb : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2fd8e5c63da44799b6cc9c93d12a5c9 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3515a4cc3f1d4e5b91d7a34eb72801eb : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fea06aadc7b7403a860088b9d48fe3b0 : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0641e517db14c208a38ecb90a8a4c35 : SUCCESS in 7m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f296438d062a498a8d0e752db7f41368 : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/395db18079c7428ea75657f828f6e5fc : SUCCESS in 9m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3ee67a14b0eb4d64a8f104443d10f15a : SUCCESS in 4m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a0ee3b19732d44819d38953ab408d56f : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/136643da463a444aba07ee9b07cc9a66 : SUCCESS in 10m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/638c165511774d63b653b4aae3ac980d : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8d527e5f33b4c5488d0b840526a48b0 : SUCCESS in 6m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6db9a5b57d39430c9a0c344c6f304929 : SUCCESS in 14m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1543afee5bdc43fbb65323857e63a550 : SUCCESS in 13m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a80614d8ecfd488b84a2d7ca4fb7caf7 : SUCCESS in 22m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2024bd44216346b0996d46961d4eb852 : SUCCESS in 21m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e97ed3aecee44c5a806cb8f6e5de8f55 : SUCCESS in 12m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/89779d04d1bd4021b83f3b75f46ada4c : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1fd827143103458b9a42168106080878 : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99d23d81e9c241da8178646ea1f3f340 : SUCCESS in 18m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/589ff47e9e144d37a4cb3eef1257771c : SUCCESS in 4m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/95ae62f2714141f688e88e78a88624f8 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f8409aa58794215813364da7959d6f0 : SUCCESS in 5m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f88d2e8aea64170abac690aea338b32 : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e11726bcb492401da2e6d19f83e426b0 : SUCCESS in 9m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb49a0cf4d3b4ef1875a000fa484b32f : SUCCESS in 1h 06m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a203bc1cfd4a42d0b0bd0e8eddaf52fd : SUCCESS in 1h 42m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/14e99a98bea048208419a90aa93983c3 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3aafa891aa360a3a6fa2e0a67be7f77afcb656a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:45:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b406223aa95143c5a5afdd30687c5563 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd633415859d7ce00166e5bac0061a9d3923729a","revisions":{"3366f4305e14c8e4cd8d744ca88fc62ad24170ce":{"kind":"REWORK","_number":1,"created":"2019-12-03 06:18:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/697003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/697003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/697003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/697003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/697003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/697003/1"}}},"commit":{"parents":[{"commit":"3ae758a91ded41aa5b88590de1abd5791e154f07","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ae758a91ded41aa5b88590de1abd5791e154f07"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-03 06:18:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-03 06:18:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2a49dea72c49a2b5ed5fbe5e8e1a5e2e4be4ae92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3366f4305e14c8e4cd8d744ca88fc62ad24170ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3366f4305e14c8e4cd8d744ca88fc62ad24170ce"}]},"branch":"refs/heads/master"},"cd633415859d7ce00166e5bac0061a9d3923729a":{"kind":"REWORK","_number":2,"created":"2019-12-03 06:39:57.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/03/697003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/697003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/697003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/697003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/697003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/697003/2"}}},"commit":{"parents":[{"commit":"3ae758a91ded41aa5b88590de1abd5791e154f07","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ae758a91ded41aa5b88590de1abd5791e154f07"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-03 06:18:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-03 06:39:47.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2a49dea72c49a2b5ed5fbe5e8e1a5e2e4be4ae92\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd633415859d7ce00166e5bac0061a9d3923729a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd633415859d7ce00166e5bac0061a9d3923729a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
