)]}'
{"id":"openstack%2Frequirements~697370","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I8e95dc48b8ba2a7e16f1d34eb5ad0e87a3e96f4f","project":"openstack/requirements","branch":"stable/train","topic":"neutron","hashtags":[],"change_id":"I8e95dc48b8ba2a7e16f1d34eb5ad0e87a3e96f4f","subject":"Add neutron to stable upper-constraints","status":"MERGED","created":"2019-12-04 21:25:21.000000000","updated":"2020-01-09 20:16:45.000000000","submitted":"2020-01-09 20:16:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697370-1578601005894-0e1c537b","meta_rev_id":"a19bf6336d5cedb176abe7b553ef84a3e631764f","_number":697370,"virtual_id_number":697370,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 20:16:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-20 01:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-17 14:09:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-09 17:44:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 14:09:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-09 17:44:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-09 20:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02f30c212e96326e02e0a82034aa041320cd56f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-04 21:25:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cfad674d3339b60b2cda366c47405f5d1c719d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:05:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/194694094c714bad940b48dd870403e0 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd3cea5a8de44df8ed693624cc93a01 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cfd4509501c4508b769a0f119efcaca : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7651190d7644197b3d30c52d7abc119 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/165a8754ad5c4e20bd70c69de24a0a4e : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6844cfdfca0d4c53831b5a2efdfc1830 : FAILURE in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2aa91349673d4848a44130feaf412971 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/48715849c17d41a2aacd01d4b62dd86f : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2e2a43754f747a5ac1b89c4d380be00 : SUCCESS in 7m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c96a9ec9f2e34cecbcc92d74d27e10bb : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b91abccd63c4245bd8240976d51f192 : SUCCESS in 8m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5b92c7f8af99460c8716bb899629dfca : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/61d1eb53fd7c4bbfa09e6a3c128786b6 : SUCCESS in 4m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1876eeaa136446b7ad7c88a869562eb4 : SUCCESS in 10m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8639c20881b4418c997c70b1f70a197c : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/721d44f0028d4037924a0fe706b060a3 : SUCCESS in 8m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/da98f3d94ab7443d8d3aecf8c03bce11 : SUCCESS in 13m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c7fa9f3792bd4a4e9dbd7550e05d5454 : SUCCESS in 18m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/069d4fcf9f7f45f390ce0d0a8b12d902 : SUCCESS in 22m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1a45472ce7fd4bfc971335427181d2ba : SUCCESS in 23m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7b19c2aefd764935b0151a70d980c77f : SUCCESS in 12m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f23634fedb64249a95fb96b448b6e6b : SUCCESS in 16m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f35346c22cc44387bbde84ff640cf345 : SUCCESS in 12m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44e4cb41b51b470a9748d02db409a0f4 : SUCCESS in 18m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23d4eb46e35348e098bed657675bae9e : SUCCESS in 5m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed8c3e062286473f9eb7c68e53bc8b16 : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cca0665751e4427a9fbf562bbad11b80 : SUCCESS in 10m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9413c9ef786141e8a6d97a7007abb1c6 : SUCCESS in 1h 11m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d01c194032634c3d83c5329d64443c05 : SUCCESS in 1h 26m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b4b63913c544bd987c325199dc752bd : FAILURE in 1h 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f73e8b54e78dc1b8d708b133dbe1aeaf1ba89918","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-06 10:22:22.000000000","message":"Patch Set 1: Code-Review+1\n\nmake sense","accounts_in_message":[],"_revision_number":1},{"id":"d84edac89e9250df6b5511b7117e57375dfef001","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 19:12:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e916d11a519dd19861b54d52ed48bce59b43121e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 20:44:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/39bbc949be164b3aaa7c4fcd63e0175f : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f491bf073e04a2e880c202be9553a80 : FAILURE in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a39fdd70afa7436699c7b1d90667a886 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b1071b2ade64b499f212c95adb6255c : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/281d6977c1ff4831b0e329bedd609c54 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d19a065c3db4c58b0c5b2dd1ea93a91 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a660d6f17d64d61a63235b15188964f : SUCCESS in 6m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/efbdc33a4c7641eb898b55cbcd57ffcd : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/961d1ad85b5a44e3a247420a188a0f30 : SUCCESS in 7m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ac0a34e4aed244fc86122159d7d7bebb : SUCCESS in 10m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d14e947023564ebd9c47ecfc1d80c54a : SUCCESS in 8m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f5c2d0257c7c414dab9d5c7847afbdc2 : SUCCESS in 6m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9fa315315a8b4466822103a668b60986 : SUCCESS in 4m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e919fc696f8142f9ab59ebfa6e99f7c5 : SUCCESS in 10m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b5798a6d257748b7be9a199b11cffbdf : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1a6ffa473b94afd807a2c7afd90c633 : SUCCESS in 7m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eeb86c44bfbb48e5bb5d681e44a4fa08 : SUCCESS in 12m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/14ac8916d13d464f9c84248918f9077c : SUCCESS in 21m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b23df09803544e1da0b4a8766f74a951 : FAILURE in 4m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a6b5b950c08645acaab9bb111382dca5 : FAILURE in 4m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/97b6bd18632541efaa55acb25de43e17 : SUCCESS in 24m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/26a0eb92f5624478a918fb5dda88cdb1 : SUCCESS in 13m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4276df6f21b04869aaa90d7fbbd5c49b : SUCCESS in 16m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94590859981542e6bb8427dc39a9c014 : SUCCESS in 18m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ed368d101424d24b7165e8d254ae32f : SUCCESS in 5m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b0cacfaa22b04f7e987eeab2e713bac3 : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/451521d756f8432db013027c75a01d9f : SUCCESS in 8m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5aaa053d08dd43839701931fa82bdb8c : SUCCESS in 1h 02m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a027a6fc8b9644d88b7c5c66dca9ec2f : SUCCESS in 1h 31m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/36d5658cc99d43eabcf29a14378128ed : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":2},{"id":"59b647d3f68dc5f106b8f900b34b8c3f4ec9e267","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 21:08:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3ba293efe1b215ff27f74c6b42fef58bf337a637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 22:35:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/11848d71b240433a858191af55fff1c9 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa73571c74a14149956ea5908d71c424 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6433293fa5ce437cb47b867899d050b1 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e526b8033dca43cfbf43584ea91b3b9c : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14396cd9b7d44708a7afcae96a663ed2 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/474255eda1074a008382c101120f44bd : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/532176be1ddd443b8268507e3eb2c136 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/28e40c15295243699948675f2c146084 : SUCCESS in 6m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28b726166a9a4ad9a2d91f7f61972c2d : SUCCESS in 6m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c1a5b25e1ed6469cbfb7ef736e6c5f20 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/982f03d4fe594af38a467482e747eb83 : SUCCESS in 8m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/020ee78bfd3142f194376f032c0d9649 : SUCCESS in 4m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/461025f6c6b045a7a36c8e952167f328 : SUCCESS in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/028a7d938a714efb8a1674c7bbe5d5f0 : SUCCESS in 9m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fa76428752dd416b92eb3e36483f4159 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/31403746f7b745d4a7708533c468ffc2 : SUCCESS in 6m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b82c42a21bb4434c850a0f23d9cafc77 : SUCCESS in 18m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/addb7ae65f8d4d74ac55f02780d385e4 : SUCCESS in 19m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e73e9491c3ad47ac9bf79e555b3488cc : FAILURE in 4m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/01bc9759cd0c44af96440ecd1926012a : FAILURE in 4m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a103a49198044afe8f1cb16497c4867e : SUCCESS in 12m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dde14241bea24c0ba7fa19b8bae339fb : SUCCESS in 11m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b12fa1a03c1148bd8e285de2e8eb083a : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9ef22ac5cf54f7aa152f1c8dc24fd4d : SUCCESS in 17m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/444b29ddc7894cb29ff99368f949a11d : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9697b90a74ab4fa893113f0a86c6c054 : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d0ab598ca8645babe9e3bf583e2d75d : SUCCESS in 9m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/515b97b469214f7c85a874d7fc04f520 : SUCCESS in 1h 03m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/619fbb76987840e8a837eb681be0eb82 : SUCCESS in 1h 24m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a26a3102a73a4dbfaab05078ed3ffc15 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":3},{"id":"459eb445d73f368bbc3ce91417bfa38c51910966","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-17 14:09:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"45925c8c9e0da3cf71c94366afd42fa26ee9edba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-17 14:09:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d3633e5785a5b4278ce735bf54c98f086abe9e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:56:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/95c4e227d1d242de9d97cc02e9493708 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b77bae53ef3241f58f5deb5d4d12e220 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e84fd952e47a4a50a2df2e0ae44e18ef : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30f2b1a992a34e47800136f38ee55d6a : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c67091ec78794cc1992fc7db5393c8aa : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c80d6cb6d2ec4a1abc03adc0c83fe704 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/141ffeae399444af88c06c28a94fdcd7 : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/43886873472849f39e31df9efcb12d5c : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d051b65b003c4a87bcf027e1337e51f5 : SUCCESS in 8m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1aacaffc72f44dc79fc524c5a9aa3d3b : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3951c5a248354fb9ae7e317be025e984 : SUCCESS in 11m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4d2a98877cc64be29ab8496db26b81ae : SUCCESS in 6m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/969b1312b2de42c08cee796004b907e5 : SUCCESS in 5m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8ab517012608474e9b65d11597cbe7c2 : SUCCESS in 10m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/afc4ff8253914431a50766de2d174bd2 : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66507dbe35b44019af95cae1955b6311 : SUCCESS in 8m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fe92bff968f04617855d643f9ff536fb : SUCCESS in 17m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8698f06927b4a0ab5d596866e089956 : SUCCESS in 15m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d3b3ba27c74a4253896f9366fcf0cfb8 : FAILURE in 5m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bdccee81226d4131b4145dea5c49bb8e : FAILURE in 4m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f5942b55d66d452b9ead63c4f4d459fe : SUCCESS in 13m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/407121ef929342078ef3bf5ff9402735 : SUCCESS in 13m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/115e906d084546d29a9b8e1df000425a : SUCCESS in 15m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d16d94f5ec684a2b9f2d6ef3561c7cb2 : SUCCESS in 21m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c411c6c5fd5e40a5acbc60dc2e0e7e2a : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1e8e2e30a916441b9f6a8d04fb855209 : SUCCESS in 12m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73611c427ca54d7f8896a8e394a7457c : SUCCESS in 10m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/833145592952404fa4bb6d72c85b8627 : SUCCESS in 1h 16m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8308a4a6dc9e47bc9a377553efda182f : SUCCESS in 1h 31m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9e71ffef250b435a880cac7a13337f57 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":3},{"id":"64e178c9e0efa81d6ef3d21ce28712340536bf8b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 01:00:49.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c7d3cc207efa3e95ba5be68bc37487b28a37a0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 02:42:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4ca847ec57f642a38c1c2ea2fc4d5854 : POST_FAILURE in 6m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/370e02b7494346b997068d649009e3cc : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d9258f32aed47e1b18b0453ec74fa35 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f0cbd0ab494984b09d683db3e80553 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76d9f8b2f3804132ba9423bcbc4111b3 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8a7fa0cf6cd4739bf866a892137bf2c : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8be6dfee22244e86976afe2677a561b7 : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d7b67ad12faf4bccba54813ab47bdf09 : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73a065d8ea464cbf9d12001ccbed52d1 : SUCCESS in 11m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/703ab70a9eac44208585bd7f90d677df : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4aa5b7a1dbc4450982e75860aae5c9f7 : SUCCESS in 9m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3fc27172f37a4145b22e7f27ea4d9315 : SUCCESS in 4m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a922a26e2f1b4d93a5de42baecf33ed8 : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9299b874c9cd4c0a8565ce4e1a0c7c43 : SUCCESS in 11m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cfb2e82b58cf44609826d4507c6f8b87 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e86851a399e42d98bb45a4899c41597 : SUCCESS in 7m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6bfe5877178c4f83a4edb4cd87f61ccc : SUCCESS in 18m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f5aecafa36e434791f148bd41970a73 : SUCCESS in 13m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d883cc91d84c45718c2fca66eb4581d8 : SUCCESS in 29m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8cd7f6df192d41f9a22a60677e6a8259 : SUCCESS in 21m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e545a68cfb5f443d984d4a052a53fc47 : SUCCESS in 17m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0d96b4159a67441cbd67c349e611df2d : SUCCESS in 13m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ea2b889e39f94e459c75fda056ff4f6f : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/055cae9a36424d4faed793e6e872b1e6 : SUCCESS in 18m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2888aecbe2234bb987c569f596739721 : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/40bd8466d92142e8abb5626ffd827860 : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2bfae367755241c8ab34423382a51d45 : SUCCESS in 11m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b185a1976404f2db3cb3986ff97a45f : SUCCESS in 1h 06m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/caff1fed72c14d1ea40f7acb46c88643 : SUCCESS in 1h 35m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/24bfc2a62cc048fdaa0bbea15c208dd4 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":3},{"id":"26e1ef0a84e36dbb541a829ebde8fd7d55fca22a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 17:44:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6e1d8cc98c1b595f5a1dbe87136330ae89f7358e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:45:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7a5cd4ad8507c1a3fbcc47b85bad29d76a718a24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:16:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1f315d97b312fbf5fc46dc804fe14ac24cc53512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:16:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20e36b91cd74d93b593a38d7488a2da : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/191dc27083874908b6d697912f2bcfab : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f34c18ef21b41018f2744e3480d5b3f : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47fe716c0e464b519135dbb808a4b0fc : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b330273c080e43138f999f13182032ba : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/900139de8a9b4e0b8d61ab04bab13bb0 : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/da6daa1f7cfb450f9389dbce00c00ddf : SUCCESS in 8m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6d83d4dbff042d28466c73eadcd4f59 : SUCCESS in 8m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/98409f3210754b7c831fdfa989c662dd : SUCCESS in 11m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/27f461f27bf7409b9c3089e000374de1 : SUCCESS in 9m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8b9722ab93234de781a1c9e279a6b78c : SUCCESS in 4m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75c04b96bad944cc8480e0bc67b9792a : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/44e7f4be5a55437487dea29c16e96a35 : SUCCESS in 11m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7be6c40239de4823936d543ba6f54492 : SUCCESS in 13m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99d44595793d46029edd5ab91c765c2c : SUCCESS in 6m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bb9b01a7b280404f90645a56f2aa18f4 : SUCCESS in 21m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1f96adc867947b199877ef7e59314f4 : SUCCESS in 21m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/10b6f55b60774f9bbf0b9d759d2f54e7 : SUCCESS in 27m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3a3fd35569da46c7b64daba5d6b820bf : SUCCESS in 25m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a10b47a2a82a4d79893c238b024d5a43 : SUCCESS in 18m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7499ac98a5c745fc98aaaa7230b0d7e2 : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1fb8de22dd5b4eeb9f89c4a9f616f22d : SUCCESS in 16m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c1752d95dca46e791b02cc85cd1162b : SUCCESS in 21m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fd79d566a434a308ce0c6de4a2b02ac : SUCCESS in 8m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a20448e64404650a9442cfc3400ce30 : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/59053248cf2944328b3060dc5316e7a7 : SUCCESS in 11m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81d7156b2bcd4223b91f31e75c0d746d : SUCCESS in 1h 11m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0cea86fc1b1441fba1858e661eac2827 : SUCCESS in 1h 32m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fbc1c9541bfc434da73130d971199ef5 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1d0f6fc806ec197beee808674c57a7aa6b281e81","revisions":{"dcde4e8ae33dca992fc56fe6d62411f54d612729":{"kind":"REWORK","_number":1,"created":"2019-12-04 21:25:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/70/697370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/697370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/697370/1"}}},"commit":{"parents":[{"commit":"a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae","subject":"update constraint for python-zunclient to new release 3.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-04 21:10:51.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-04 21:24:52.000000000","tz":-360},"subject":"Add neutron to upper-constraints","message":"Add neutron to upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nChange-Id: I8e95dc48b8ba2a7e16f1d34eb5ad0e87a3e96f4f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit ba81a34bfc7b1c84895d7d4dd5ea5236871c0d86)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcde4e8ae33dca992fc56fe6d62411f54d612729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcde4e8ae33dca992fc56fe6d62411f54d612729"}]},"branch":"refs/heads/stable/train"},"e2d20dc6b2a384611b9e9a7e7d0f05d66e25dbaf":{"kind":"REWORK","_number":2,"created":"2019-12-16 19:12:19.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/70/697370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/697370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/697370/2"}}},"commit":{"parents":[{"commit":"a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae","subject":"update constraint for python-zunclient to new release 3.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-04 21:10:51.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-16 19:11:07.000000000","tz":-360},"subject":"Add neutron to stable upper-constraints","message":"Add neutron to stable upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nChange-Id: I8e95dc48b8ba2a7e16f1d34eb5ad0e87a3e96f4f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2d20dc6b2a384611b9e9a7e7d0f05d66e25dbaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2d20dc6b2a384611b9e9a7e7d0f05d66e25dbaf"}]},"branch":"refs/heads/stable/train"},"1d0f6fc806ec197beee808674c57a7aa6b281e81":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-16 21:08:25.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/70/697370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/697370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/697370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/697370/3"}}},"commit":{"parents":[{"commit":"bfd4804c4763ef7b2990871eb5458288d54fc07f","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd4804c4763ef7b2990871eb5458288d54fc07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-04 21:10:51.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-16 21:08:25.000000000","tz":0},"subject":"Add neutron to stable upper-constraints","message":"Add neutron to stable upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nChange-Id: I8e95dc48b8ba2a7e16f1d34eb5ad0e87a3e96f4f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d0f6fc806ec197beee808674c57a7aa6b281e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d0f6fc806ec197beee808674c57a7aa6b281e81"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
