)]}'
{"id":"openstack%2Frequirements~697377","triplet_id":"openstack%2Frequirements~master~I6c9f545e09779431120cee942318013ed91a581a","project":"openstack/requirements","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I6c9f545e09779431120cee942318013ed91a581a","subject":"Remove most python-2.7 jobs","status":"MERGED","created":"2019-12-04 22:05:12.000000000","updated":"2019-12-05 06:00:39.000000000","submitted":"2019-12-05 05:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697377-1575525529472-cafc307f","meta_rev_id":"ba3903fee76100ebbf86bf06b1824763c13bace7","_number":697377,"virtual_id_number":697377,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-05 05:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-04 22:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-04 22:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-05 04:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-12-04 22:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-12-05 04:59:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-12-05 04:32:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 22:07:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-04 22:27:21.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-12-04 22:28:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-05 04:32:37.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-12-05 04:59:05.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-12-05 05:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6e37b3f4e0a5de8b80344b8260e481c27fed855","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-04 22:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f7034a3640b4d0a3bc423d84bc7c537617ceb0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-04 22:06:07.000000000","message":"Topic changed from remove-most-py27-jobs to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"3ce69891660ac26edb549335a525afec2b28453d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 22:07:13.000000000","message":"Patch Set 1: Code-Review+1\n\nCan we merge this faster plzkthx","accounts_in_message":[],"_revision_number":1},{"id":"47ae1a54c5141ab57a06312dde8eba077a02f856","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-04 22:27:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\n+1 for nova.","accounts_in_message":[],"_revision_number":1},{"id":"43d06200700bb72ef6aa593a24e5a7afb1fe9ca8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 22:28:35.000000000","message":"Patch Set 1:\n\n+1 octavia, we dropped our testing already","accounts_in_message":[],"_revision_number":1},{"id":"5c5edd40836237adc0f6b9bcc34cb61fc98e3923","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-04 22:28:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"18cc684b0b71ff14ff90946730046eaae7f99656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:43:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3a18ac9d14419cad3e00b2ea5d4897 : SUCCESS in 5m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cf8d686c5d2d4ad096a6855852f35504 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b522ffd7591b443fb596f8504e8a01fc : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c1acaca47b1452ba2edf0c3e84868bb : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aad2dc0ec81844f7b98e46afe05b2528 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c7eb097dfc541ad83f3cd3556ecb348 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1ea16b6b57c42799b2b81363f5c0140 : SUCCESS in 4m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0d0b064854e44036a38883ee421507f7 : SUCCESS in 8m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4364910390f249548b912581010fe29d : SUCCESS in 4m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b4cd718531b4ba98c50fb8fa4f9781f : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4768c4b4b174d64968f90e4389fd77a : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3f6d183a73324fe891aa1f6e8b42bdd6 : SUCCESS in 13m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c13fd6d1c3d465993812426160ed456 : SUCCESS in 23m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a2d0155ee712440380d31089869019a7 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c0dbeec2920a4d40be13a034141dbbbc : SUCCESS in 16m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4732f5aa327f43989f1e116fbf070e38 : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3fc2a1f30fc84d01a35c8888675297c5 : SUCCESS in 4m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/33a32a3fb9cc470ba138c87fd540ada7 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3817471726f4a829424053e8320b175 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3db3d5a259ba4c949ad1be5e5ccf7025 : SUCCESS in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e804f6bd39d3453d9ceb5be8b3711e70 : SUCCESS in 10m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38f2228af6cf46a6a572d2ea3dd2165d : SUCCESS in 1h 29m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9cdc4a3058f244adafbbe8cb0c58462c : SUCCESS in 1h 22m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/14cfcfb1cd7544e0814164d92ddca813 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"45e578b9a3253d279a7ac0c71a0d7330297b65f3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-12-05 04:32:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI think I would have preferred we drop them to check only and make them non-voting.  That was er get a little coverage until py27 is actually dropped","accounts_in_message":[],"_revision_number":1},{"id":"e0d562293293c0873a19cd478bdbd66b9c4f3cdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 04:32:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d025c315939fb2a5714d1b084dacd1b89315d8b2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-05 04:59:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9682f439732aa2d93900956c0e02ee9a2d6ad190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 05:58:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d20202caa643318d25b30dd2de8e24 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab39ecc0a2244ea1bc359cc9714b95a1 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2137a9eb8324d12bc1ba60ae7d8651d : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfabce40599c4ad4a0e6023e87e56d6f : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7a4cfe6e7e548d687640e34b0fb8aa2 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3678281983945ed82248e1846e81550 : SUCCESS in 2m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/355da3704ff040deac5243288218d194 : SUCCESS in 8m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7429bbe892b14d6887852f3bd2727d5d : SUCCESS in 4m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80408df33c6c4619a904612c6191b497 : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5daa3ee2972d4db1ad8bc6c74b91a337 : SUCCESS in 6m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be3fa7d2bfb545e0bbbbf2d375b2af27 : SUCCESS in 28m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d88d5f572b1342dd807ce311ba7213ab : SUCCESS in 18m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/071d80adac324de68cabb56566f45d77 : SUCCESS in 14m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/386012bdc16f4bfda89c385d3982e584 : SUCCESS in 4m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e3c19ac954cb4684aa2ec95830e6f073 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/739504bcd8a14053906a3a5d5f45f499 : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ddace0a4db5a4060a1bdb17ecce88c0b : SUCCESS in 8m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/33c5f7ff3b884906ab3d8dc0c9beddfb : SUCCESS in 10m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f989a13498e344988ba0b506da03863f : SUCCESS in 1h 04m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f4f0bd706244062a8b9956a3d03766d : SUCCESS in 1h 23m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e35f224163194fe9b0f92154a6e4af0a : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"80ed1b73240e42a031620e80c7cba211806becde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 05:58:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba3903fee76100ebbf86bf06b1824763c13bace7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:00:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d5441943b1d47449235b0407284aaf6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcda45553a415ba82360a779db6568554fa30537","revisions":{"bcda45553a415ba82360a779db6568554fa30537":{"kind":"REWORK","_number":1,"created":"2019-12-04 22:05:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/77/697377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/697377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/697377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/697377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/697377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/697377/1"}}},"commit":{"parents":[{"commit":"cd633415859d7ce00166e5bac0061a9d3923729a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd633415859d7ce00166e5bac0061a9d3923729a"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-04 22:00:39.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-04 22:05:01.000000000","tz":-360},"subject":"Remove most python-2.7 jobs","message":"Remove most python-2.7 jobs\n\nremove all jobs depending on projects that are set to remove support\nfor python-2.7 during the Ussuri cycle.  This will keep us from being\nsurprised when a project pulls support for py27 and breaking our gate.\n\nOriginally this was set to occur later in the cycle but as we depend\non the projects for cross jobs it needs to be done earlier.\n\nChange-Id: I6c9f545e09779431120cee942318013ed91a581a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcda45553a415ba82360a779db6568554fa30537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcda45553a415ba82360a779db6568554fa30537"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
