)]}'
{"id":"openstack%2Frequirements~697533","triplet_id":"openstack%2Frequirements~master~I0bc4d08ab9efca9f4796e9f90a3a08791f0a2008","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0bc4d08ab9efca9f4796e9f90a3a08791f0a2008","subject":"update constraint for oslo.concurrency to new release 3.31.0","status":"MERGED","created":"2019-12-05 17:49:00.000000000","updated":"2019-12-10 03:22:57.000000000","submitted":"2019-12-10 03:21:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697533-1575948110668-56cf989d","meta_rev_id":"89baf651951bb753628e62dc27b474200c6de46a","_number":697533,"virtual_id_number":697533,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 03:21:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-10 01:36:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-10 01:36:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 01:36:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-10 03:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7eac326f511f16cb0d8ccffb7f7449b38f98902","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-05 17:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"552a7b317634e1f711e0bee4054ce1f48265bd21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 20:24:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9900657e874f639ffcef610be998b1 : SUCCESS in 3m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/35d9f4b0ef5940e0931a0faf5e2509ba : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9877f5960c2b45e78fddba50bb710cbe : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c642a3d61a249fda8e92fb9f66b0ad3 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aef3e85aa684ee2b169ba46283813c5 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3463e690bbb3463393979a8e4828f042 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b6afa826e0d4a40acf6ed5e078f9647 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9370c6b2c67f41d699dc56af6ea0283e : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7f1d63ff35ac41b58df88c076c9c6272 : SUCCESS in 9m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e65c389b534c42fe8d727902ab90cdf3 : SUCCESS in 7m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/560d9847bcbd4e98addadd651f84f2ed : SUCCESS in 11m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5a27b9f755434f72ba5fa86b618eae23 : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/35c87eeadc444f7191723e1e2cd33128 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c5e603de56340358e56fa78c16069fd : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/76ce6aace5374ce28174ddbb29c42828 : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99c5ec5cf4f14b19b79d8c707b3b9ef5 : SUCCESS in 29m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37b26dd87e5748b096449659da5b2724 : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c67308390a25401685df0e3a4f6f09b3 : SUCCESS in 15m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33fa811df844463ca00b68e4b1024118 : SUCCESS in 23m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/be14c3816ee645cb8ba28441f0ef844c : SUCCESS in 4m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0535ca652d204130afe3f1ab5eed8948 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19b120c21c64406caf1152884d6f33e6 : SUCCESS in 7m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b5447fc0cc9d438e8c5447de15c648b0 : SUCCESS in 10m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/93ba0d749a78449cac48b4bd97f433eb : SUCCESS in 9m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fac12dccb5d94c1ca0c9924a19b59c1f : SUCCESS in 1h 14m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b602a48af7948e0bba6b9b36a5ad832 : SUCCESS in 1h 52m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ea5612b080954f3e9701b296c0e7734b : FAILURE in 47m 19s","accounts_in_message":[],"_revision_number":1},{"id":"50515e66205a98172938130bec3b166d9e92aa89","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-05 20:52:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b9717170f245120d3900aa573512d5aa0aa4b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 22:18:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90dda73ff0604990a702ab50acb831e2 : SUCCESS in 4m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0c7438660b4842f59edb85e816f82dfa : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c81877169a4567b83ea4f5aa2308b5 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de5b7b4f839f47ee92353e8db08f6cad : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6d81c9176bb44e59adedd30efba1cc3 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2eca2ceb963f4f7ea8eee9cd3b9c3f6b : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f052c08acac44f8b770d26b85418ff8 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f83d488a2a0447aae80743c12a14fae : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a9735f660d1f41dcbeceb1edf7a6c8d8 : SUCCESS in 14m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f39b4eb32b14456874dd36727987e8d : SUCCESS in 14m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4179ebae155f4dc9a2dc77f602ef011a : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57c902c7adc543bdb23b09b3c46d0340 : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fc1013801aba4115a956a4ced968578d : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7f15f1916e84419863b9d4907b3effa : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65a67fd7e3454edf9cde005f9275ae6d : SUCCESS in 18m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/121b40a9eb90431f8a9672e37ac7a7a6 : SUCCESS in 19m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1513393698eb469cb05db0d2a8d137ca : SUCCESS in 12m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fa471010641e4bb2b21e9cb9a39d0f98 : SUCCESS in 20m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/350b87bed31c473992853f44a2e76b59 : SUCCESS in 14m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5028f428834b4bc4ad711a13893c683d : SUCCESS in 3m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/abae39aeb8c34960b7ad51ecf1b4c7e4 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02ed1cc92e17413387f978b5b1dbcafd : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bd1433dbd8ec41ecbb2b9f55e50cde99 : SUCCESS in 8m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ee1260f00b3443fb2a14541958e4f2c : SUCCESS in 9m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d26bbe37d0446f582a1c5c319e8e290 : SUCCESS in 1h 17m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/53aa740cee40428598776f5c3e4f85df : SUCCESS in 1h 22m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/65a2ba5e12994b3db6adbdb5146032a5 : FAILURE in 1h 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a93bb9e225396aecd85dcaa26427695531b5bc3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-05 22:29:32.000000000","message":"Patch Set 1:\n\nrecheck\n\nhmmm","accounts_in_message":[],"_revision_number":1},{"id":"9c1796bde5cc46670a7cf14ac6a25705a6021aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 23:52:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2512d8b1f21498caa08fbaa960a9d69 : SUCCESS in 4m 58s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/daae654c98ff4f6381922aa23d8c51bb : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a0699cd59d465e9117b8c7571d1e84 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bbb5f694cf34c99a68cfeeae556e683 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85536d553be84529a3dd1c0ae486cc7c : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6761d99c9bfb4d68aad7663f5f3d91a4 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1df27a82ac0a42919375b92d2f6aa639 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8758ccbbf47432b992a950734dcecde : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0e80aad8e08d441fb3d45cda7c7ebab3 : SUCCESS in 10m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2ac6d57bb0b4261ab83e956c3d4541d : SUCCESS in 6m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/453e5d4b5c5c443585d6fe70fd1b01ce : SUCCESS in 11m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc47d0041fee46ceb68b0fb34367185c : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63173e6083b44c758141b4a0b4a48912 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00f5ac92bc764513b571039eaf924161 : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6bb56a0f96b45518b2439757618ff8a : SUCCESS in 15m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c60bcc4b412c457d9cca160466f45b80 : SUCCESS in 27m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3a166220470d46d0bd2cba55421a71a7 : SUCCESS in 15m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/35ad666766394e47b63e3298d2b953f4 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/431c273ef98f40ba9f670746be8894ca : SUCCESS in 21m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4670a304daa41cfa364010a1c48a191 : SUCCESS in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3686302e7721421897f2c2f5bf0a8639 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad42901d1ee24778ae309be550e922d0 : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/deb8d0e9a315436e98fbf02abb8848e9 : SUCCESS in 10m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/08d36b85f3ea4b6c814c6aa90a10a2c5 : SUCCESS in 8m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c7b2cfaba7a45dabb419fc5bd2b0240 : FAILURE in 1h 16m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0266d8b7e5e6475db1700bf59a362166 : SUCCESS in 1h 21m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/77bfacc1d9be4401825dad0eb7e8603e : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2c1f56f36f06dca172532ca3ce07ba588a03b9b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-06 00:17:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9006176692818b4a0403f5e06154e35c1f9b009f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:40:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83b5aedfffd4a9585aeee3665409f35 : SUCCESS in 3m 59s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2d62aa2815df447b8c0623fba2d24be1 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbe99d99a5344ff949c9a3fc8a51991 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02cc39d46de64017b4a6556bb463cdc9 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7e5e08d10541a29a11ea5fa70c3b8c : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25996d2b257a460495694f65befb9e21 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d50d5cda45884589a2bfbda23c657671 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db03eea930144b82a9763c2ea93b01c8 : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5f22f9b80c364b67b10535ea64ebb7e2 : SUCCESS in 8m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e7c22bfb7744a0a939e213ca7d45532 : SUCCESS in 8m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2423dd51e0bb4595ae442160c5cb8bda : SUCCESS in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e099f14887fb43d3a553a65a7e410315 : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7293007af38a42d6b4a0a3ade09bbd3b : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26d4a0ce1b61416fab465656a3f0f4ae : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d275279d79f4b5fa987665591c2ff4d : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/54717807bbc742688cde96590f39d9e2 : SUCCESS in 24m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1dfc163fc0c34d5d9e6768d9d45a542f : SUCCESS in 12m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc25a5deb0dc4302be76abd878b8bb40 : SUCCESS in 12m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d0692e3011a442ab90e536eb15d299f : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f50aa299c0404645aaff17724535e228 : SUCCESS in 4m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8a0b8e78a66e45578e178a0e06aeb69e : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5e9d160e09b4e01ae24a3b0e0fd7197 : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d3eb16c9c56c4ca7b7c916e9eeaec229 : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1455a95a8c9d4cef97c82d07b4b849fd : SUCCESS in 9m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/217f15c62f8e48499e00bf3bbe190378 : SUCCESS in 1h 04m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e96efcc4c3f34eb5a125234330e644b3 : SUCCESS in 1h 20m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1cfd77e36bae41bfba2b2a85ef356bea : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"28613e5465e899647912d49358808b48b4840516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:52:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75b119b53117509f568a2fc5920109c115142177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 08:42:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5720fcbe561749ca8fc5149c06428844 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7c45e86abb4183b12e57c14add21a9 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc78cc678cee453ea042153fd67f695f : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e3934ae44524ed58e830cfad8a6995d : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cbda9d3a027427d8502492817117127 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/127ee4a7cdf640df873c51414dc42b14 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47f510ab303e4870929926d2a0ab2361 : SUCCESS in 3m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3e148f51f5034cd585228a828489ee09 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13ae109b9150467aa0aecdf871863d3e : SUCCESS in 7m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/22ff8b35a77c48b69f0bfe6b29d1ba4a : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0cee2b941f1246df86c9723deef04af4 : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92a9fb6f65584f74978389049ad076e1 : SUCCESS in 8m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/905a8f6a8bee4d32b23d5baea6a25bf0 : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1081a780bd094b73a76c4d4ac897f518 : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/676e62de41ff4f06aa8a5ceeef250cc2 : SUCCESS in 22m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0783b118f8c44c959a54d4c8b8da44ff : SUCCESS in 18m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/22a306458850425aa3178e3141fbc102 : SUCCESS in 21m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b87510b1d3c493d9a73a648a346ea86 : SUCCESS in 18m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0d3205cc94e845909e54ec8bdd560ad1 : SUCCESS in 5m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ff9ded77f46e47c1a008256d1aec6e35 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6757f3d6f44548f2bef0db3be8d624fd : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/41bf0f0f0cdb4ad99c08ed40dc6ccf45 : SUCCESS in 9m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b49ece8493ef424baa695f6b6b247b2c : SUCCESS in 8m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a13aeee987594c1f8c95d67c7d9df43e : SUCCESS in 1h 12m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b45c05943e04416b872ffdffa9e470a : SUCCESS in 1h 24m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d298f3693d1d41a4958112b1874d7139 : FAILURE in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5e14ddfe4127573df5b1a047718b954bc3bc61ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-09 22:58:03.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"91e80a5cdcc8263b72171b13527d1b51e5b17086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 00:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d448747e9224797b4483f414399dac7 : SUCCESS in 3m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1f3d4a9cbc914bad97350d207843d9b1 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf25e5ed74a4c6aa2421699d0a8d4d3 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a65b497ff0b64f9f99a2f3446317ad2d : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ec716a7de94e3aa113863f9fd97a79 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a50b935fc674cd386dac6244f1bbc6d : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff4ad6acca3c441d8744bbfd726f3ef0 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3693d32ba0424afc8278bb6f98d9c98c : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/26b60ac2202a4f928ded616ec492d464 : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c772d31aa37648469d658b26e333a4b3 : SUCCESS in 11m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1eb0bcddf4a2469c9164844749708485 : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/463308cb841547d89f8e448c7a76d648 : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1cfe6b75294e4ac3abab5e2a827e083a : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86fedcf0b92a4b77a3321416e7f1c806 : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db4c2cfbca574eeabc10d7a5726f1e38 : SUCCESS in 24m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb10ed2632c54d10ae6db5c214aa7068 : SUCCESS in 19m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83fdf6b0a40749f8b4752881130056ca : SUCCESS in 14m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b60bd601df1b4ddeb6394e441fdd2246 : SUCCESS in 12m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d01eb80b98749dc8b61627f31c2c537 : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8b18b72745584fa0abc2f534184df34e : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5b7beaf27dad4eb1a40ae886e065cab5 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a330320a8f8401e8422d1e167326014 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d2f08b6c9fa64f1b8b9de8ba72540053 : SUCCESS in 9m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6b90861947a444dfa9ef8b428257ed8e : SUCCESS in 9m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8dcf28b7adc47c390ba340518929483 : SUCCESS in 1h 08m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/202619fdbb9b450da250aa7bc8d08496 : SUCCESS in 1h 29m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/839f82d46b2e49f18ef9f8e0cebe232b : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"76ab2bc7f84c3fd60d2d1adb531130c89f376d46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 01:36:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ae4315afe8c1816aa848bed53aaa6f36fcb907c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:36:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb2920cb43afd5ec1cc3fea9e88aee524fb4cdd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:21:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6350f9bbd460acc656434ae9d4de13b4f7b927ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:21:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287f1a4351a64b5384d036372956df90 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e98b5ddfff4c6fb6b8ee984518ce29 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3acad80769494e79bbe9fb7e70cfe127 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73b7c952e6ac4d17849c3d892a18450c : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc277953841744ca840bd9d10abbc289 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c2063bee51a48f281d0cdaf06f3af5c : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d21dc9b2810e46c698c6d55a0a11a16c : SUCCESS in 5m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b9ae2c63227948aa99bcaefad3208e2b : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2cb56b56c394445ba6569e59f41ab9fc : SUCCESS in 6m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a91c22c06e184f0cad0bcc44fdb7591f : SUCCESS in 11m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b440d5e3cf3a4ea4b7aabe12c4f9306b : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba6139a20b954689880a6d22c038ba28 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78a687bd25ca413b9558eb17ca98bac0 : SUCCESS in 6m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6d797726d3294598b0a7c2d92e91797a : SUCCESS in 18m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b271ad8563824be881de8a98a90aefd1 : SUCCESS in 24m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b486371a8844f778ae653714f303bf5 : SUCCESS in 12m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f39d566b5ee74667a7be3b041ff2c132 : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5b3da77f98c4ac3817dadd1f9ff176a : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af9bbbca0c484797b3034e0556cfe609 : SUCCESS in 5m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/50f5ed681c5346348edc163f36d6249a : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a01ab34f668948ccb11649805f8b70f4 : SUCCESS in 5m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/494eafc4ef0f492c86b1e051d60dbd8f : SUCCESS in 8m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a625338359c6463f9a077024ab6732c3 : SUCCESS in 13m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd62630a8d814ae1a1d55fd38daf9b55 : SUCCESS in 1h 08m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/944dd51ee76446589c9e7169b04aa0bf : SUCCESS in 1h 31m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fc23265034804735981e89bef04bf76c : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":1},{"id":"049fca993dfd85ed523161c39201e9e72b6f02c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:22:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110d45b4cca54b3590334dc13814ee26 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f20050b6b94399afd5913f13803a6bb9ecae2090","revisions":{"f20050b6b94399afd5913f13803a6bb9ecae2090":{"kind":"REWORK","_number":1,"created":"2019-12-05 17:49:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/697533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/697533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/697533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/697533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/697533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/697533/1"}}},"commit":{"parents":[{"commit":"fd3536a93a1d27260a10d37f1f8d99a3c3404060","subject":"update constraint for os-traits to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd3536a93a1d27260a10d37f1f8d99a3c3404060"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-05 17:48:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-05 17:48:57.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 3.31.0","message":"update constraint for oslo.concurrency to new release 3.31.0\n\nChange-Id: I0bc4d08ab9efca9f4796e9f90a3a08791f0a2008\nmeta:version: 3.31.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Idcad8ff2daf63d601aebbeeb087e5ac10ae73c2a\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f20050b6b94399afd5913f13803a6bb9ecae2090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f20050b6b94399afd5913f13803a6bb9ecae2090"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
