)]}'
{"id":"openstack%2Frequirements~697534","triplet_id":"openstack%2Frequirements~master~I116a269b254f33e4e83a35c0dda52c60ae880a55","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I116a269b254f33e4e83a35c0dda52c60ae880a55","subject":"update constraint for oslo.utils to new release 3.42.1","status":"MERGED","created":"2019-12-05 17:53:06.000000000","updated":"2019-12-10 03:23:51.000000000","submitted":"2019-12-10 03:21:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697534-1575948111631-7edb6769","meta_rev_id":"7c39b5f5192579f5dd03859f9815be309fe6bcc1","_number":697534,"virtual_id_number":697534,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 03:21:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-10 01:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-10 01:36:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 01:36:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-10 03:21:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84694c2f419148ceac4c8dd99d71df331232f2a3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-05 17:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef384d51af62ab0a6e7079ad7892b2c33e02cd1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 19:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f10bacafdcc4410881d713901bb2d5e : SUCCESS in 5m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/47a8f3e39fc14903999b2bc26efe3635 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82945183f50845adbe8bad02708aec7b : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6dcb054e1ca4bcf9df4dc9ac6a30b49 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3955a27025fd472db00abee5cdce760e : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/172d71ef7c6347f7852936de795a5c04 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2eab7f13edcb4dec8ae938db97f918da : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/993d090f8d314c60a71b3f7275c29e28 : SUCCESS in 3m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e4750da90781469c8d1d9d41c4debf66 : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/adb3d4b62b444ab284630c10d5f608fc : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4021fd3ed53c434097c8c57af616e584 : SUCCESS in 12m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/937bf8f4c02e4b65b64fdc0aeb8a1c5b : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1845938f15d04bdda7860a3431d5d8f5 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cbde2bd415245dea852e2251f022036 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8349672fdaad47c38a7aacc844cd7354 : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/feedc60eba1e409e9a9634424c1b3618 : SUCCESS in 24m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/06db1e1fa68e474999a83ba5d6b9147a : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/96939b7f49d941188a2d9d2ba4258c9c : SUCCESS in 12m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0601eebf556e43e9a94a717e4033ec45 : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/385e52428d6b4ddea9c05b09c7be732b : SUCCESS in 7m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0e4b8a4af2d44db7b01ad04bfc21c8f1 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ffc32a9aaf674d65bbf5a77e2f318791 : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f16b3f59d364fbdb3dbe7127e58a6bb : SUCCESS in 10m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0f6a9f2a23c94aa79c73183d9dc65bf6 : SUCCESS in 10m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c98c26b2fd14f80aafac37367f7f47a : SUCCESS in 1h 18m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80326ee3ba4f4061bf478e9003744ee6 : SUCCESS in 1h 19m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/24e7605a57fb43ce9b1bf2189d65d6b7 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6101118ccb8d2594488b8b80a26ea2b4a66cd862","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-05 20:52:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4de5a825da8ba6c393ae088b63725cdd4bf1519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 23:06:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5de96730fdedfafc1679bd8c8750dad65d98d13d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 03:20:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735853afbd23426292f861693025cb2c : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857d2b0d57df4006a1be713e53ce4120 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ad22c869598423c8c33097e762942e0 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c3c68ff79e34eb6a6ca20b45468e661 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/019f114bc0d1435d832855aaf7469dbd : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b54ef2bd141d4a36b099f45fcaa0031b : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1db39a3ced714497a7e7bc70f64a6215 : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0cfa71e61046495a8ffa1a5fa99c6d12 : SUCCESS in 8m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7013ef730e8f4019a456f137266402b6 : SUCCESS in 7m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/902b12c77d7847e0b78eaebc1ac0208c : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/614a762458da462d9e3e434b1c81cbb5 : SUCCESS in 6m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1148d776c0a442c3b3ab2791eb7a2c70 : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/defe51017d2341189e645a8b23e1677b : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d68bc4ce8b6b44d2b0a7a905fc74d76a : SUCCESS in 18m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc8fb59f7da7494db17df9b257cee069 : SUCCESS in 26m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6db48af826ae4252be6860d8499efe18 : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ff4a24651c4742ba983bd80c68d9ac7a : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/148e9878b6f2445a99258a8dd159ccea : SUCCESS in 19m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca471666aafd4ccf95b8db9753fcab34 : SUCCESS in 6m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3724211689ed44808f0d456556b57fcc : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec77c816a4a14a44b4a6e6e37dc36712 : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/975a603a196f4a8ca6d1a03096c10f96 : SUCCESS in 10m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c5d14f28f6eb4840ab7e88cba168e6a1 : SUCCESS in 10m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d706ae339e8473cbc95c3ec6a701608 : SUCCESS in 1h 03m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27c5ee27411848f681676506c6c58a1a : SUCCESS in 1h 20m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fd5cf3cfbdf34d93a4ce14d28e06f18a : FAILURE in 1h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"dea347f43bb68011fbcd6fbe64740cdf3f1f8162","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-06 03:22:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23d2516ba0c37ef170bb3afe2e7cfc8139273552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:53:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ebd48127f44284a1141e0d1680f3a8 : SUCCESS in 4m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/007c98e04bb14305894170a78fc4c787 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fcbfd189b2f4b42beb187847ab31e87 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3de2a080931940cf867164e697a2770c : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49a4a215727246c7a964221dcc986c5c : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0024bd631a224f34b24b4793383d1f39 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c143b4c376194a9e95b1b117871d0569 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a8f4d6509bb46b697ee91b7ad1b4946 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/db292749795a4f8eb92f9064e04c2be7 : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc830baf88ae49a691069a9d840379a4 : SUCCESS in 7m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4238973713494b3ab3fee3dbe035179d : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb352548b620449fa31514457420d455 : SUCCESS in 4m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/756cd05afbaa4fbca0a325fcf530003f : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8da492302f0849ba8fc1614358d1080a : SUCCESS in 6m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ba0fa05c5f7342f3b720035ac3c0460b : SUCCESS in 13m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b4bca42e2d494e3b814fdc2f40021873 : SUCCESS in 24m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76b01175cbd84f07a7467a4160901232 : SUCCESS in 12m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/37214c41a59c40b580904161c0de6784 : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50f9abbab10742d7a84a69b976f501e1 : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8a317f2a5ae4a6ca4e01adef8cffb59 : SUCCESS in 3m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/873a2e4a45be486190426ace2696c1d6 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12ff1fa34f53499594f297e52ac02c4b : SUCCESS in 5m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/27bcaf4f9e95468da849e02f0df9a54f : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bfb0b40940c0444dbac1c2cb24c773bd : SUCCESS in 9m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88c4ed0b4f8e4283b75121395f0bb9f8 : SUCCESS in 1h 14m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/96640a6f68354c53a52b07c91dbee99b : SUCCESS in 1h 26m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/60d2814e449247bfa4fe8d0aa48e1fb3 : FAILURE in 1h 08m 46s","accounts_in_message":[],"_revision_number":1},{"id":"24007d1c79bcc4441d323595fd8ddaaa228f5ab1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-06 04:57:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"732838f8e83cf21586c30df23d59ae2cc212c123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 06:21:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10c7f1d225f4d0fb8d0bcc8abe82275 : SUCCESS in 4m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/26aeddad61f34c5f97c5a8ac348816b1 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e021266a117449508a061b13af87fe6c : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5588bbf1f5db40f38f8bc7a64688ea61 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2584f751de5648c3aed2a9dffddcee5a : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79a912f17a4f4d33b66aaf8417da569a : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ada5c1985b4459c9acaabd01ac7c740 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4609f2f77ff74710a0ffbb8a17442ed2 : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1cfa8e4da1a4457e84024af12c24efdf : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1fd14c86a32f4d15961e695c2b14731d : SUCCESS in 6m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2500774dc2af4427a73563817bc3cf69 : SUCCESS in 7m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2bee2b9ab22a4c32991b5bf63b08b2e6 : SUCCESS in 4m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a7465c175a984f73a7bc1d681ec1a78d : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/deef2f81d224427ca707265b88ef6fe0 : SUCCESS in 6m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5fcc8f6fdd5e46ada2c4fafa90875d48 : SUCCESS in 18m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed2049ca267b45ec9d0a72a4eb3fdad7 : SUCCESS in 21m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b279621ae58f4e928d601f49a4694b5e : SUCCESS in 13m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/56a73444b0f54527981122056c632b9a : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/146a8ca38d7b43ea8455c61a0dd51ad3 : SUCCESS in 18m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b32e227739e3402b950985d28c895bc9 : SUCCESS in 3m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b96f5e37484340e29dc6882dd9082b0a : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f3347c104a54a55919c82fc9c63177c : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8208930be8584f9994ac2401fd70e254 : SUCCESS in 9m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c72a9ea111f94669ac8ecd1135fb500c : SUCCESS in 9m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f64a4d181d294139906c8bfc7eabac6f : SUCCESS in 1h 06m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/72242bc672b6450fbc58b26996851787 : SUCCESS in 1h 22m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97b26197c5ee407f9892924bec78e7ff : FAILURE in 1h 02m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fd58387d13c205d92442a603731c810ae41fbe97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-09 22:58:04.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe29424aa4fd3a00fb535773bec8e771b4586970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 00:35:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf2c3297edc4668bf3c72c614fd9588 : SUCCESS in 3m 53s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a32d0e27ab1f407cb7088bf198a76696 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80549ee5ceb446149ab95d1c687a4333 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9deca2e946d741d78b1693050fb8c4d4 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b5becfb50c7450da9804b07f997cda8 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/801afe58b0734613ae2bf520cbd0bfc5 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89a63394e52d4c708c39810e2af24c35 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/152d2f114b6d4f56b3d42a874a96c9e6 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/32bec4e193884bc88f595de079de4d56 : SUCCESS in 9m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1eb6d016b51b4d1d89cc4e48417da750 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6e8e1c862ad4413da327a3e412066c98 : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/98e89e91ffee470f8c7c715301688989 : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4ac893636dc44e9ae22ebbd73f6750b : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9613536e8216466a87ab3c905d1d28fe : SUCCESS in 6m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c64952f057284540b2af9ed6293f4da7 : SUCCESS in 18m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66a5041d2ae54c1f96c1a9c87d02e949 : SUCCESS in 22m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/45b56fc6dbaa410ca539a6000d433f22 : SUCCESS in 11m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/575b3ae801fb4833827a86b6b9155f53 : SUCCESS in 13m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/995491f9e8374beaba3a19d0f16d8f48 : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b0ca0f5f451c40df8c5f4be1f3be0648 : SUCCESS in 5m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ca16f31ce38343d7aa257ce8a382bc64 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16972e4877d3476cb42b88ce7f6eba6d : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2b1b548b54944ce5b3c2c29325cdec72 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/099cd9f5048f4bf78cacac4d5b45410c : SUCCESS in 10m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3daf1e899f94275b6d4a46a95e3229a : SUCCESS in 1h 04m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/57a30ef156fe4844a01cf8f2c9294c46 : SUCCESS in 1h 28m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a04d24b59dfb476eb0e66c525277d18f : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"121a2b970e8d098f25d791373c51fe0369341edc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 01:36:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c198711543edc2692b3cfa0298f1b99f09e57b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:37:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dd10a889afc03b3c736a9980eb2ff97efebadcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:21:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"35e87ecddf8d2e54b79a56865b24c551b3d59586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:21:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc2b11531be458eb5b47815d8f0dda6 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f60b521303475bb46862e8893748a0 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebc42bc5264e4ae28ec7b9b6a55f3c5a : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46ef066e37ea4920bb64e3617e045f48 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87f6be90c46146bfbb3d80a702e93d1a : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77fb9f552772404781af55561933e045 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfd7e4bd2dfa4a9fa6917f540da1d991 : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/64972aeeaa5d4a8fb91fcbffc17153d9 : SUCCESS in 10m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3dd0055facec4e40b7ca921bab9e0191 : SUCCESS in 6m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f492a3ca3234d7395f65bd97fbfe2ea : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dd2eaf6b8d554cdf894e10c1d1f6522f : SUCCESS in 7m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f8cd46ee3d0c4269b9bf6152329f2d7a : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d0950116e9c4f37954062f27a228ba6 : SUCCESS in 6m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/396601e604204091a5a15c40c317efc8 : SUCCESS in 21m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f223ae6b3c644599b2863311d31bbae : SUCCESS in 22m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f39ba8a28a94cb8940d96460be6ec5b : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9c3045820d534f9a8c64157632573fe5 : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/609b5dbb00c244c0b0a7e5029f9a7857 : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/83451cfc57f44bf582c160e17a3ed8e8 : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d6a64b40a4b14180ae70fea01409db55 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a53d5c0aece94e68a9ee38a5f7de69d8 : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/50db1904caf148a2b23181c3bed0e8d5 : SUCCESS in 14m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12f1bc196ac748088194fcefd8d0f323 : SUCCESS in 15m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd39540b5b1c4f109e7e94eb7d4b8f1c : SUCCESS in 1h 04m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/530283acdee34d67b72910a16c11f381 : SUCCESS in 1h 20m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e8627d86611d46febc2330de2714ce84 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2fe92f50b0cf8ff5a45684a1a0e7ec1dc33d3d60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:23:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b429cac31bfc4301a23fb36ad04184a9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85ce11a1b9717caa8336f9aafd0ed0a88a4e2bf6","revisions":{"85ce11a1b9717caa8336f9aafd0ed0a88a4e2bf6":{"kind":"REWORK","_number":1,"created":"2019-12-05 17:53:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/697534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/697534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/697534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/697534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/697534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/697534/1"}}},"commit":{"parents":[{"commit":"fd3536a93a1d27260a10d37f1f8d99a3c3404060","subject":"update constraint for os-traits to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd3536a93a1d27260a10d37f1f8d99a3c3404060"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-05 17:53:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-05 17:53:04.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.42.1","message":"update constraint for oslo.utils to new release 3.42.1\n\nChange-Id: I116a269b254f33e4e83a35c0dda52c60ae880a55\nmeta:version: 3.42.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Idcad8ff2daf63d601aebbeeb087e5ac10ae73c2a\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85ce11a1b9717caa8336f9aafd0ed0a88a4e2bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85ce11a1b9717caa8336f9aafd0ed0a88a4e2bf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
