)]}'
{"id":"openstack%2Frequirements~697952","triplet_id":"openstack%2Frequirements~master~Id83056ecc864b687ec7b7695daa1903742741a57","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Id83056ecc864b687ec7b7695daa1903742741a57","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-12-09 06:16:45.000000000","updated":"2019-12-09 22:58:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4ebb56390ad3a9cfcd5d36faab506f556c4ca7ab","_number":697952,"virtual_id_number":697952,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-09 08:20:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 08:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82da4e7ab42546aee12dab40b9d1adf9624a0f7f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-09 06:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ea5cd518659abc1cccc843990c75b19db632baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:20:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e93f4ba57c045e084be3cb1865f74ae : SUCCESS in 6m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c0fd5d28c0224331895fcde57e59bb24 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e68ecd3cf84f809157a3dadb031594 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdb6208eeb754390973e57853af1df4d : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/506dedad353543eb8c46a70d7ebd0531 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70dcf46eb2854e47b4c3fd08f5df6925 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e038a2c26cb44943b31e6c41b7e0a43e : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd9829449b254c64960c782056908a63 : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/598cfb287d8c439eba924309892d1396 : FAILURE in 9m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a3741637225c4d3c901b2a7ef4a88e43 : FAILURE in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/33485e643ef944328fa52ca892e457c3 : SUCCESS in 11m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8904db6712fd46a683f645ef731cf3ab : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57e47a68330740e0b9e8f025c02a8b3d : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8df413cb54146899b0c7aac85f08291 : SUCCESS in 6m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d8397b82f53f4f739057d36ec1f0d69a : SUCCESS in 13m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9169396e14264221b6736b780869ed6a : SUCCESS in 25m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/42dbf09b9eb54d339457505d19a7c2e7 : SUCCESS in 18m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3d0ca8e6777f4eda866aea65b0ce7a42 : SUCCESS in 13m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04cc609d60dd4bfd9c9e85296e4559d2 : SUCCESS in 16m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/458400e162c14673a3f4ebdc9b4f5cd3 : SUCCESS in 5m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/084010b2f98e495e87b72d0b26d30731 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d91c7fb7fb2643aa965a397431f2120e : SUCCESS in 9m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7df461ef2e3244989c6006acf78763bc : SUCCESS in 13m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dbaebe7de9c84c56a45733bd43e4b1c4 : SUCCESS in 16m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/12fb1ed215934c3f81a86a8f3ec07d74 : SUCCESS in 1h 07m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0ad5e8198874c9695a8250578c3973d : SUCCESS in 1h 29m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b02c247d0ccc4a0684b3acc84e565ea0 : FAILURE in 47m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b77281e4e4f67fc908394579f3ed620e04268228","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-09 22:58:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b35bb132ddc075828a6cfbd7c78d2262f5a009e","revisions":{"5b35bb132ddc075828a6cfbd7c78d2262f5a009e":{"kind":"REWORK","_number":1,"created":"2019-12-09 06:16:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/697952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/697952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/697952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/697952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/697952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/697952/1"}}},"commit":{"parents":[{"commit":"be7531995ecf3a144d38904a20ea4b0f5da425f9","subject":"Merge \"update constraint for octavia-lib to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be7531995ecf3a144d38904a20ea4b0f5da425f9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-09 06:16:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-09 06:16:40.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id83056ecc864b687ec7b7695daa1903742741a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b35bb132ddc075828a6cfbd7c78d2262f5a009e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b35bb132ddc075828a6cfbd7c78d2262f5a009e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
