)]}'
{"id":"openstack%2Frequirements~697966","triplet_id":"openstack%2Frequirements~master~Ifbebbc34d23737bda5b6c840d480d6629eeffaa1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifbebbc34d23737bda5b6c840d480d6629eeffaa1","subject":"update constraint for openstacksdk to new release 0.39.0","status":"MERGED","created":"2019-12-09 10:11:05.000000000","updated":"2019-12-12 02:29:34.000000000","submitted":"2019-12-12 02:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697966-1576117665754-7edb6769","meta_rev_id":"9de5f622a2926991a38421e74a994d3c1a7ea361","_number":697966,"virtual_id_number":697966,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-12-11 22:38:18.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 02:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-10 01:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-10 01:36:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 01:36:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-11 22:38:18.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-12 02:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffd02a93f454078fd006dd24cd56f13378280733","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-09 10:11:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"511d87fc54e4b8b88948419376a018f077b115cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:52:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3b59b0f8c44922a4d43a073c450591 : SUCCESS in 3m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7f7d17c921284f1a9308b91aaf527fac : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b31fe7a0ddbe4ef381c00611b9eed9bc : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dee4c71d71b34748a50df173ce31d458 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ef2bca37a14b1cbad092110055b292 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49ee5ed3798c4c49809c4612f48651c2 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cf977eed56e41499382897bd0b4976c : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/149ff73cbfb0427685ef924db251e7a1 : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/85ff373b21e142a3bfe7ffb328dcd02c : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a40104145ab4d0b8bb946c756910830 : FAILURE in 3m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f7389f573b04d539bb9b3bc27444c08 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd2ba3832c9e440c8661b43606e8bf04 : SUCCESS in 4m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f364a6192f14a8396918c763687c49e : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fef6f82b488143b291bf1b61cabbad56 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7a56476dbc7a456eb52a6919d3e7eab4 : SUCCESS in 14m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8661947a088d48d492ee6cc1957aff64 : SUCCESS in 22m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9fe69e04dcbf405f83cdca5618ce29e4 : SUCCESS in 11m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e8129795134475a8cbe8d042a801e2b : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6c8470b020145fe94dc0cc5da0b441a : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c51e347667a4f919590e26539ba0338 : SUCCESS in 4m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a34993360889431e8e204cb883132ae7 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14c58f7c3e1d4c738be742648543aa2e : SUCCESS in 5m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/306027fdb8e042769e73382dbe6f407b : SUCCESS in 10m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a7b81e0981b548f2b8b5f2c9ae4661ad : SUCCESS in 9m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/39e07e268d7b41ec8c41d24b42631ca6 : SUCCESS in 1h 15m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f01b25475ed4a48a4ca67add6796a78 : SUCCESS in 1h 35m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/082eedc1dddd416a817ea5e6c8f38282 : FAILURE in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"021199d6746a1705f3c352f88db721b0c3ab9f3a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-09 22:58:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7239bdd2efe7597438ac44d6d7b9eee0f6a2f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 00:55:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b40dd633dbd4caabf667656d717c930 : SUCCESS in 3m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4b7c6a70f96047dbaac90399220cb99e : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11bc1cd8ac3049c8bef33f7e608fdaf9 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/391409f708ee4a6f872ca5f0b9b39654 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/161a9ec8afd048efbcef7a228d6f628b : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/333a7a72cbef42b0bbfa6be5f2da906e : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/715f10400a9b443091efa7f3352033c7 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c0f3889937e4c8c986b4289997dfee1 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3385c47f4c224ba990fd5e87d6f40fd4 : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65bdcebf9e7b4d50b09074d808cde4f0 : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b759c302a724d80a81ca6f81fe5bb67 : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/270c7e98c2744679b59013fa5d45285c : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/564374ff7a6b45ed9108a9f6a2c20993 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9309395a3154581a763be419a9df50d : SUCCESS in 7m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/798e25b94d22437590be065697aba314 : SUCCESS in 18m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/758f599a35e741448779e78156232e00 : SUCCESS in 23m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5ef822421f1e4cf085f143d7779e703b : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/567e828b08bf4317885f4e33fcf4065c : SUCCESS in 12m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/407f82337a9e44a0946e9e2981bd006d : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d714cf1b9e44d1a9d28a88f79af2b65 : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/963930ed4a624832b1d26fcd7d5c7d23 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb3b11b8eec5402b9b51ee97dc17f474 : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d664422d0f94774a24534389e4899ad : SUCCESS in 8m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/972a31a3dbd04be782a6749c91aa4a2d : SUCCESS in 8m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7255ebdbca464dd68b17a2b04453cce4 : SUCCESS in 1h 06m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d461577c21f84678b6d59eff54fa827b : SUCCESS in 1h 39m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d2baf96bf17f4987b04b0adfacf5b1b0 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"83aea2f41ceed401cb1320c120f25a9d188b7086","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 01:36:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08046821fd845974b254ee2eca469b79f5a4ac38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:37:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f33bc95985f919a2ca5eae78a40d38c37fb3c17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 03:31:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb116bc9a194100956fa3451322ce67 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3c69a374c645ca8a7a441d5f6e0ba9 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a415310241584e60b197f89da96022eb : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3889e04bcc334baa943d711c2a3df4de : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/745e02d96cb84d33b6443bee1715fa87 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8dda7539de4e419b9c190fab92ae20ec : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e88934dbb744a4f933213973c74adad : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ed89a83ec1f54e6c91a6a88bb52e4aed : FAILURE in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93045e39fd25479ba6f2bfd525da7230 : SUCCESS in 7m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a1b0dc531f6248bba2634b82d12d6e3d : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd866aacf82c4ed4aad934490045879a : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4468fa2b6a8f4e958d87c368980a7f85 : SUCCESS in 7m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49f4c7221a50477aa40ecbcb6b58ed9e : SUCCESS in 6m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2245fd24ca08456ab17a990eb9d01aee : SUCCESS in 18m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8fc53d91567e4d13b974c6930b253bfc : SUCCESS in 20m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/939e603c4c4346fab8f73345da0523ed : SUCCESS in 13m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/904fa089466e42ad9a6d880b4adf6b86 : SUCCESS in 15m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/530dbdd1d4fc46bca44619c39e789c37 : SUCCESS in 17m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/13170e7a1fbd445ea996592b230b6fba : SUCCESS in 3m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3cb6457acd3149358499cba61d10ec85 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55613be7499f482eb3c0103fa7ba291b : SUCCESS in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eaa87528fd864236acdad29cd7f578b4 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cbd86b52ae37413a9934d6e9d9d5954e : SUCCESS in 8m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2298ba3aef6443b3a4ed682cad58f8f7 : SUCCESS in 1h 10m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/51485560798b465b931a10343a115264 : SUCCESS in 1h 16m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/72f5582f82514f6facab69c8f1bc56ca : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"54b11d9b8a440dc4d5c830811817680203005811","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-10 14:28:24.000000000","message":"Patch Set 1:\n\nrecheck ERROR: No matching distribution found for os-net-config\u003d\u003d\u003d12.0.0","accounts_in_message":[],"_revision_number":1},{"id":"4d6107bef850861e33c62a9b704e5b1798f9e74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:36:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d9ddf96f294de89b6473eb68ac2c79 : SUCCESS in 3m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8abbbc1e15204f38b28cebbb48397a6b : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e2faaac2dc4747a199d456acc20af4 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6082e821019416498a591c819eb7d4d : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e45e068fb1ab4c868c43e19e2c0ac1e2 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7934916772614719b13d57c6da43d2de : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc4d9a3733a44833a946bd3a63283259 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d334b29f89a348c69020edb3da473156 : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/19faed0b718e4cd1a0463038f113faa3 : SUCCESS in 13m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8cc6586c242f48dbb5e46f6f0288c611 : SUCCESS in 8m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/45cd922e4297435187bc5d52723243c0 : SUCCESS in 11m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/03ee8af176984568bf509ec8bf79cf98 : SUCCESS in 4m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/452b3876f0174eee89618c5084716aed : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a110319310847e69dda865f4dc41745 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e0e18bdf054541f3a0e7618e0090def4 : SUCCESS in 13m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a17ed632311475f99df345624c36497 : SUCCESS in 25m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/522cc16193794c73b35d440938babf07 : SUCCESS in 13m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0a0e8348780f457ea441b91e3b9a0fd3 : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6bb6998b1544dc5b095d78c57ffec91 : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/714136c11f8a4596ae39f8184a8cef5d : SUCCESS in 4m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1ad91216416e4a3398cd24e2a6334a84 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ca0971b3e04467cb5d7401ffe6b06ee : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/601f0ccbdfe54759bce796c6bd7129fe : SUCCESS in 9m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ab9a7236e654fbca1639348b3348e92 : SUCCESS in 8m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b56e9be3f4e8483696167565cad4b150 : SUCCESS in 1h 19m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8de40a24b7704967aa3ef518864c2004 : SUCCESS in 1h 31m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8714a3711f6b4b73ab85ed22f81930c2 : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3e2c976f9c20166f735dba244d854770760471a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:37:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17d5f2ae909efd7adfc3aad50a8b1a737553a384","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 18:41:48.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1dc06d2fdee2885aaacc3df170a8dba6ad824a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 19:25:36.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"047dff686c47cd87ca410cdc3e0ebe112fc37508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 21:43:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babdf5ca808c49c9b08fb9518ba3a9cd : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71136921fff94baaa199cdf20ddf7d93 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92da54a48cc44e2a83cd9fd5386d6207 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/088cce81b24c4a30925e3d5e6f140f6c : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18c9cd0358f24f8191428206ccb83ee7 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaa6744dc56845fd8e5c5b794c0be15b : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f55e6041fc084f98994cd71f5642c8c9 : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b7df32762ea94dcb9b5a1f5f2385b1ee : SUCCESS in 7m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac932bb0d89b44ae8198402b9fb33141 : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8c8a241020ac433bb45ed87540480b3e : FAILURE in 12m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/39d29f36f8664d25b57a3c9ff5a35582 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/97b6438e35b44b5e8042d9937a1a1320 : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a9e696c326542dc985a42684886358f : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ddfd3c41abf5414c847b7c5bf44bf03e : SUCCESS in 13m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/806396b596384eb9a76ffb83db0e6d22 : SUCCESS in 21m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d4a2b97ad8e43209a8b42cfd37c92e5 : SUCCESS in 14m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/71620aad182b4a019c44d2add72503f7 : SUCCESS in 16m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ecd3376cddc4267bad1cb6f937a3f51 : SUCCESS in 18m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d6753a609f2345f2b9e6524514406d17 : SUCCESS in 7m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e30dcb32f9374f48a2d117024589cb8e : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e84b4e0c44e477b876d37c10fb77731 : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c9a3d04a8d824579a80a32a07bac1e40 : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/da7f6f3bcb764a40a0f1b77dbbe0e838 : SUCCESS in 14m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44cccf1f40294d9e9368baa17148d596 : SUCCESS in 1h 06m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a7c318059ceb41718dae31af1ad779ff : SUCCESS in 1h 40m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ddc5242c4dec4f5cb82cfeecb9807ca7 : FAILURE in 1h 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3f50af2b847ef2c4132d759dbbacbc05d6c5c9d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 21:51:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08df08391ee91bcf47cbec38822b0363209b1c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 23:45:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943afdf600164c1c9d684e05034419e4 : SUCCESS in 5m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/83cf68f3ed5c4b40a2b49227782dbd82 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d2365113ed4e588a27408ab26e90e4 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/546d4ec314c844bbb481126a80322e9f : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e5b43fa978b44ad818a803459ffef52 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/106a8a32b0034be78e959aa369c66c14 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03b67f89a3144441921bc0e0cafb978f : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fef198f5f41f4b16bd8e52cb084aa283 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/57618b10fc1f4d849e1d68dc6e5a10c4 : SUCCESS in 7m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b86877e88404cd5aa04c7b2087f993a : SUCCESS in 8m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a11dd821affe4f2299c1fd30e5fc9729 : FAILURE in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44f3f2c0f4754dec83efca196e2f862c : SUCCESS in 4m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/42f3bf32a7ad47139fc4605dbcb9612f : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f83995df18f44c8bb390fc00b0a0334 : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d61e2ce41c694ef5ac39788068b6be96 : SUCCESS in 21m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e442809bd09f4704aceaf8937fa3ab80 : SUCCESS in 22m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f2fa09c010cb4118b568154dba098366 : SUCCESS in 13m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/af841f509a404cfe8385d7adf4dec98c : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9374ac3d36ef4c718d6c55390a4a5ed1 : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/294a8473595a43a6aeb185f7ad116ac4 : SUCCESS in 4m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6cc2c92bed0b4699a7b93a46f086bbaf : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a994d441f2f41f3ba4265a30848a1bd : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5636b1d765a346428feb024913c38474 : SUCCESS in 10m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b297a40f555423386fb9e68310a2f65 : SUCCESS in 11m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b6aebf1225d42dd9061b0cbf454e397 : SUCCESS in 1h 07m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8df32dadeec0438086b3f72930510c2a : SUCCESS in 1h 27m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e51ad0cede6f4df597c33ef46203c42d : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cfe0095ace579d8a537ebc0de88c583dd6e12989","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 23:47:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"370baa1157b1f0b5b3fb5a6e93eba2d995da43dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 01:20:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876e28df2a8a472082c8814a50fd8ecb : SUCCESS in 4m 23s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4ca318dd36534fcbb241732bdc144531 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d837024139064bc2b4a213fb11d2e89d : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8818bc4e0e4a4c55b8204d89abd0c541 : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7100da99a524b60b1bce7c46eb94035 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5263c01ba5c7426d9457c52242522e63 : SUCCESS in 10m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32ee6c03f5224476898d8cf77b7e2961 : SUCCESS in 9m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b724e68fb9e48f3ab31d2bef6046dc7 : SUCCESS in 5m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/af62cbc9580d4abe955c6558ca27e159 : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e751ef1a204f494ba7f4fe2239b6b2d8 : SUCCESS in 6m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9135f8a88d5a4fec9dceb89a89f0eb26 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b57d258a01784693abccf82cbe482df5 : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e02f4b4963614492b4a3fc2e2f575aeb : SUCCESS in 7m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3488c5b25454456e8d2d24ef96a31f99 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e782972742304a7cb299f048fe5b8e61 : SUCCESS in 18m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/98e41f413a4042b09825b81823d4aeed : SUCCESS in 24m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a49a2a6ef6824c3a840ef1c0c21b7844 : SUCCESS in 13m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/39f2aabbb9df4ef0bf91fd9e58c651d2 : SUCCESS in 10m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d851456ed6047fc8bd35aee6a22ffc7 : SUCCESS in 20m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e4cb6b41e3b64d73b2594726f1d17522 : SUCCESS in 6m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dcd12775013040409acc32c8540ee9b1 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81dccf1359c04accbd482e8dcc01e257 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/523783c8db3e49aba2726e18f967e7a0 : SUCCESS in 8m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e4a2c2a3f8a145118c4670cd9119467f : SUCCESS in 12m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6678f6f70f384fcca87360894eb95b38 : SUCCESS in 1h 05m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9bab542477b0413b99d36ca7ea9040e7 : SUCCESS in 1h 27m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3a4a34530284d189004cc39ef317b21 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":1},{"id":"423e35605fb290c78a1d0dcbc29198ded3084323","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 01:21:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"623b0e5a2936b1d88ee8e0f7c9d3218b1ca93230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 04:52:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d0b47918e64405a676328ab77c45ef : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd4424451fa48b98f77017880c38194 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5eed856d28584b52943539bf41ec8c56 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dfb2f1155344e21b64cedc3fa484996 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32469dfd79974f82811a54c4b6363ee9 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e555a9a1cc07429ba8ac89715aa6049d : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72f9def227b44c07968ea364617de34e : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a4f061a27ee0434faaa1ffa5a3d0b6f6 : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/62bf9d0b0f574e3a82a90414c17e6830 : SUCCESS in 6m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f953579968f34e8580258af843dce98f : SUCCESS in 7m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/25f0ac271c044b1cab69f2f660c8f9a7 : SUCCESS in 3m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/59f4fa679a644b1ba3b0b96d023db677 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fecac4c18405491e9f4a8bbc0296afb4 : SUCCESS in 5m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f10669828d534ca98dde369b50fa280b : SUCCESS in 16m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18ce6a011e20464a813105c436072355 : SUCCESS in 19m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a1b21bf2324743d89021e7f6d3eef76c : SUCCESS in 11m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0974bec461df4d40ad670339c0c8a95b : SUCCESS in 16m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aeb6f578a9094405ab9bd940586f6d1d : SUCCESS in 19m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5620ba597a434bbb87d130382dc6dd48 : SUCCESS in 3m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/79d814df07ea44728dc27106a4162eb5 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c5114cf22cd46459391effcb90168be : SUCCESS in 5m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b2156624a0c846e496f5fd6b65ffbde8 : SUCCESS in 9m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5613cb1d361e4a069dee57dbe3eb36f1 : SUCCESS in 11m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50af3cc2437b4501b4d96702e81c9d4e : SUCCESS in 1h 06m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d96546b89e0b457d9ea2c45fa7c733ee : SUCCESS in 1h 25m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6356772ca6548c5bdf212c9201c48bb : FAILURE in 1h 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"385ecb802a745dd6e380386bb302bf08befb3121","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 04:54:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"994c46793620a91c37fa468a10f12a009b866894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 06:21:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b0650f16974e41b603a85f05679bde : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9f315529906248df8a7aef2c817f255f : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b540c70ca44574abb63f97453d5233 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dca85e9767b41d3b62fc19f35f8c4a7 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e57a28e14c5e4c69b3a362a787f56eff : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f9c66f6c7fc49cebde35f11bf265f23 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c14f93a084254c0581c9617effdade50 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b092f2727fdb42bb869513a6907e95bb : SUCCESS in 2m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c62b282658934c63b2f9be0b7dbd3935 : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f0473bf160c49e28c22f090b858c5f2 : SUCCESS in 5m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2486201c81d4f62b526f672478e63d9 : SUCCESS in 7m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd5ffe07771e4780b26b34ecee6848d1 : SUCCESS in 3m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/acc839ecb6634b1bb0104c39522bf499 : SUCCESS in 8m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd0f50adc9a94421b88a4d2280487785 : SUCCESS in 5m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/76731f3cacc641f8a1e9b610353ae268 : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7990f95e580a4a79afb86523f2fd9b08 : SUCCESS in 19m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c839bb722ff459ea73673fc9b77a202 : SUCCESS in 11m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8d104e13ade14a618b63b76ea4747ca4 : SUCCESS in 11m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b844358445d455c93c388aff6fce538 : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1c1e7d02e3194506ad9bfa865d2f2d25 : SUCCESS in 3m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/65d5821ada824a1b996b199ca3a1de48 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d175f5ba73b4f82b8a0758501983aba : SUCCESS in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/25451de4f130498f862dcd388431a533 : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87c5f01a5c034764bd63fc94bd035f49 : SUCCESS in 9m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f44463365782449f943cbe582a19ef8b : SUCCESS in 1h 01m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cdf5ce8ce73a4b7b9953255d9447be7d : SUCCESS in 1h 25m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a92090c005994561b1397d558fec58b9 : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":1},{"id":"61919fcb3ff16d2bfa7a2109a73506f66be3688d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 06:22:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae4cce4c7278406616ad34af1267f3e1ad2727f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:07:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062b1a326f304549af726d0aa08bcd7c : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c40a3475604b40a54519e1d1350b0f : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1652a7e3f7a7430bba4301007c362801 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f54210ecb0e848679ff189ce5a74c00b : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4fb5b293f154459ba6a39c8259b85c2 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f97cc083da8d40cb8bba628020b74741 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25ffdc4c15514298a0d7128d07d6d5f8 : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/10078526f4e14c42a5c63bc71e742207 : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a03cb04633b416c98674f38c1ae1fa4 : SUCCESS in 7m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e7d86157d9cc4f3089c0ee0ed6070c57 : SUCCESS in 7m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/66cbd87f57564754a526ab34540b409d : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d69ca73bb7fc4ec1aa8b712eee310198 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d93ea5d573b472b81d169b5cff68ebd : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28e56c4e1dbc49cba7a88ac22f6acfea : SUCCESS in 21m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ada8f4ac7a8c4b86896d435f693f3ccc : SUCCESS in 21m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/90e1cb35598e4349bc3180b57895611c : SUCCESS in 11m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c978293b432847fea10335429f23e03a : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a92b0baf494b4848bb816a5899bc1c02 : SUCCESS in 19m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f19a18a932c740528fe093e10f909213 : SUCCESS in 4m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ed26e8eb79eb4c0fa04e373ba0b35519 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/681bdca88a074e1e9ad98503091039b1 : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8eb6bc364e634cb6b8869c32e4acca4c : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5eb52bb3ca9448228f5b225ee28231d0 : SUCCESS in 9m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04b4bbbc73c444bf949730e8568dbf19 : FAILURE in 1h 31m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c349df57ec64e68876a0ae5c6cbe22c : SUCCESS in 1h 42m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0899993c9944449eb5c3cd03bff8dca5 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4c66ee0c0e7c48da921e054f538fc1c622a8758f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 15:52:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97c568262ea6a801c973341e5ac1246b21effc94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0215ae0920a4e5c9d4f1eee0b091fcc : SUCCESS in 4m 19s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7e58689a872548e3b2d465d343660155 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59ef81ba88740808cce70c16910e1d4 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/859e398f146747e58b8bda8c8be9effc : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8158838a8624b97acc51f2fb24d13b2 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/457f8fd5e8744142aff963be8b0bdeb2 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4eec6c481f50486895827bcb9410cfad : SUCCESS in 6m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5095f48388bc4320bd411d4217d051a7 : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8a0e0aea52a146699c6f9c566539e12f : SUCCESS in 7m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b772c75d28744e4a882e33fdfad7d3e : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2cb69d205d2f45ee970d5e6db7adae87 : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5193fd7172f49ba89b4358c7be09484 : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/714802d580144c34971f80da983a9813 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/196eb2f6051341359ef95d5c3568f582 : SUCCESS in 6m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a2624168a6f84573be3caf55a29fb38a : SUCCESS in 18m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b3b5ffeaa8d4455e9889c694f604e897 : SUCCESS in 23m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/328557e64d7f416ba28f12e73b88946c : SUCCESS in 13m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e1484e24d0b40d99a19f4d6903a78b3 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/683e90ee50a24d3a8d7522adbdf9c631 : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bf4897abd2f24ed9b652abca4a672f5a : SUCCESS in 3m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a4bff11b4d924c5388f853c57020c917 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9060fdb83fb34588bb3d2ac14fe7ba79 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/de0812d6ebfe407f9aec2e3c8683f58c : SUCCESS in 8m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fef5f7c86c99469a8da5c1f2a58c5110 : SUCCESS in 8m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/991afd8d8eeb4770a0d096962d2b23cc : SUCCESS in 1h 10m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff5b2ed638bc4e7d8106c95497aaaecc : SUCCESS in 2h 02m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7f9a6ff705ca426fa40bce5d3e017b6c : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":1},{"id":"525ee4ba3eab382147261edecc2a7dbdfa15fa95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:57:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"518332ec6520cd30fec817ca0ac819bc95d8ea83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 22:13:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0105e7c9a34d8791ef127e860d0cd6 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e2fe4e73114c7da98fdc4f99c8e8e3 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d80c1f233be948fab7ee6ecb4bbc8de2 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e6a15495021462591a487ace3d70743 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86d4db97a916499f88b3daf896b712a1 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04119ddbc0d94daf91ebd67636002657 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fe62c00f8ec4dfea9f1055b3af48106 : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/88d4cfa4671f4debbea459722599a897 : POST_FAILURE in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/202449e9550c439189df0da9d63dd16c : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/212eb48c0a25469aa155238a8a26afca : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/17832975acab41d1809a4536a7fd4cb5 : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/51f55c5c493c4ed0a7c744c5a493a15d : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca9f70704fc74f4d93df0f0f3fa1adab : SUCCESS in 6m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e0bd6c449ab44db096bb94b3ca1cf82b : SUCCESS in 13m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0ea90f094bc244648d0f7b16b5025c31 : SUCCESS in 23m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fdf112adc7d84a86b382bd2d2e1bf2e0 : SUCCESS in 13m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ace21a19fa7940249323d713066e8da4 : SUCCESS in 21m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e9b1dccdc4c54c11a873b717923a6ad3 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0535d45d722a4fe8b480f8970bc2596d : SUCCESS in 4m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/93548830d0174a4a916e8f915e3dec6a : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83afc0ac57734169aedd972c3e2f9cea : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/74b8d48dca224de090802951e481b4ce : SUCCESS in 17m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/68f6b1483070424caaa2e010f57e5a73 : SUCCESS in 11m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8be652e53ee74d18aa8427706c12158b : SUCCESS in 1h 26m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fcbf2c73a444489b825ed00faf6104d5 : SUCCESS in 2h 10m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0b036adfb2f3432c8f5b9bb38d5b9d0f : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"826fd8110f57e74db886416bba28298ee632a1bf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-11 22:38:18.000000000","message":"Patch Set 1:\n\nrecheck infra weeeee","accounts_in_message":[],"_revision_number":1},{"id":"538f4c059bcefd0a8cadf5adeeb1458167744a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c86ef97c234dd3ad88b5256cfaa92b : SUCCESS in 5m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f2463ecb335e41cc9ab45be28a4d779c : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52db2cc0f5924908abd1f6ff700226ce : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/102d8578cfec4df7844e0ce6b6d8e889 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/694de2f976584db39868520371351d7a : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0ad56aaeb7d42dfb115a0ab071bf0a6 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26ea0f0cb561461298aa1176d03529c2 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a457ef94cd924ee090450e32cba6e1a3 : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ed0f53aca63b4d98ab1c4ad5b8a3a9c5 : SUCCESS in 7m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a1bb8dc0c4f4e4791d2a2503e149a24 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8e4e7116fb844b28a61329f6e3908f1f : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/605f54b2cd7e4c99944ace2ba968f27b : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7e5728da401e43d6b45cf9365131a952 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c0c59cb21434f9799b52b78952d1064 : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/77d8d840362045c28d14be21e531570b : SUCCESS in 16m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d8556802ea634a80b3263ee49a4479d5 : SUCCESS in 22m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b335d3ecbd4a40deba4082e484e47629 : SUCCESS in 12m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8f3f93628a764736b2364bf6fd1273b5 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a17f5ff0d3b4af7a5a8d77240458a67 : SUCCESS in 19m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e4235fbaa38a424886b4e5feeff3ae2e : SUCCESS in 4m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/aa2abaff7a184c82bf724192a10840d4 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a6064aa237f4d9dbac02901cc63b30a : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4734c8e7d82d40ec8dd45c1523263664 : SUCCESS in 10m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/921f7cd3523f48aa8cf19d43662873b7 : SUCCESS in 8m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17865059db974e2da7924c228dc02efa : SUCCESS in 1h 07m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a4ce5f2c431f4d30940213f2d4f2203d : SUCCESS in 1h 40m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf90cd8932d04ccfbcd507f3a62a9522 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b583d63e2a1a2663745dfac198cdfeca008850c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99a51372ce59b58ae3db64a00157fc55c412173d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 02:27:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb375b24a9d411c8e5886857ab46c8e : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55743c85aa2f4b3bac8aaa9e8b28e746 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3565901086c54f2eb116de206dbd9dc9 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf716eadc7843b4894617c5efecbe33 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f64287eb0f6f4b23abd7429aa4cddfc3 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19ccaf9a316a483ba06ec359a16cc176 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/309a684e8962425e933f98f5a54d1d8d : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/859b4916e5be427b96b3a0897bd2ad1a : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8ebba1ce73e8414c96c46ea0c93d959e : SUCCESS in 6m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3582d24993134ba29433d31c65d7f47e : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/48644df8e7ca4804bbef77848d6877a4 : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a6950620181740f3a695d2b08a4b9b0e : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1ca2b0aebce41bca4bb260ca93a87df : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0cbe79cdfd0b4ab5bd1cc14c3251b477 : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eef79ecebc614749a44eabd306603d4a : SUCCESS in 25m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b162e25e6b4748438c099b57d46e8ea7 : SUCCESS in 13m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/08dd846be1634e1d823007bf68c2b640 : SUCCESS in 13m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23ef08504aa14042b687e376782689ea : SUCCESS in 20m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/79a16a3d9c1f4759b76fae886ec91c07 : SUCCESS in 4m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3bf8768c8d7f448cb4dfda0c87baaeec : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3242f0781ebf4def864dbf4a5fb4125e : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8febc2ba6b04eb1aee53021fbce19a8 : SUCCESS in 9m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9131315f11dd4b1bae0c3502b1ab2585 : SUCCESS in 9m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3d8bffcf2d442cc84d4f620c40444e3 : SUCCESS in 1h 09m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f1f648b31434be7a92e31171e301d81 : SUCCESS in 1h 37m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/adce964fcf80475ea3d2fb3cfc773b5d : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":1},{"id":"31f242352de6d99a8ea9526c4e39bc7b46d06a9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 02:27:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e2bbf97086310e7d54ed51c5bddd91d83f9256d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 02:29:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135c6fda59294447b7aa2ef60f1a84f7 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"934efbe9881d9898b20c35eb34a4ada28c8df6e8","revisions":{"934efbe9881d9898b20c35eb34a4ada28c8df6e8":{"kind":"REWORK","_number":1,"created":"2019-12-09 10:11:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/697966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/697966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/697966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/697966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/697966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/697966/1"}}},"commit":{"parents":[{"commit":"be7531995ecf3a144d38904a20ea4b0f5da425f9","subject":"Merge \"update constraint for octavia-lib to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be7531995ecf3a144d38904a20ea4b0f5da425f9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-09 10:11:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-09 10:11:04.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.39.0","message":"update constraint for openstacksdk to new release 0.39.0\n\nChange-Id: Ifbebbc34d23737bda5b6c840d480d6629eeffaa1\nmeta:version: 0.39.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: I23f54adf07dabf798fce8947919f51439a20c766\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/934efbe9881d9898b20c35eb34a4ada28c8df6e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/934efbe9881d9898b20c35eb34a4ada28c8df6e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
