)]}'
{"id":"openstack%2Frequirements~698145","triplet_id":"openstack%2Frequirements~master~I50dbac0f230a0ce9d5608dba18d891b77002d785","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I50dbac0f230a0ce9d5608dba18d891b77002d785","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-12-10 06:15:28.000000000","updated":"2019-12-12 01:21:38.000000000","submitted":"2019-12-12 01:20:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698145-1576113628255-9875c5dd","meta_rev_id":"ee90cfd693c43a18a1160b87e3e09f50c92fe912","_number":698145,"virtual_id_number":698145,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 01:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-10 23:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-10 23:32:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 23:32:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 01:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72bfaf52417c2658be54cc3ffd25029465445086","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-10 06:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7577e341d9a2c78038f49380cd4a9b8f4631126","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 06:40:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"212132229e61d718b916593a5fb5b708aeec5a97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 06:40:30.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"e00f38d3db7d9182ad57315061538351dfdb2180","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 06:42:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"102be52d0685f5f670bb0bcea661409ea46bf3f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 09:10:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9ef7756f984cfeb6a72d088b38144d : SUCCESS in 4m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ba8ba169e28a4be78e0537067b733cef : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa7c2b5eb094246b8436132ace95b7b : SUCCESS in 10m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33519ce1e79f493788b01d6786ba4e46 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1281646708b84e2ea4c1640b53b831a6 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec0ae8335276407aaa5468cff3046383 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f72ed0d6ed074a6188d296426d532fa1 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b88ad2518264ebaad35094cdec5a392 : SUCCESS in 4m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8e0b5058e5a1434ca9f2f7ba329bc525 : FAILURE in 21m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f664020fa9354fa5ac13be5b95f56324 : SUCCESS in 24m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7cc58b16677541d8802ac195375199f5 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2a81b67d4f24dd2b7b6778418003a32 : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1fc8f3a1ff3840b5b54c524ca0ede799 : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8f6bec2c05344fcb6ec2874ae56d013 : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adf4261bd0cc4945b8dd0510950c3ae4 : SUCCESS in 21m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/194480b0821d4cedad63c89dd016dfc0 : SUCCESS in 21m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7e6bbc286e4c41b68b89f5390fa59848 : SUCCESS in 12m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e299696ba4a24708acb92608a99b5589 : SUCCESS in 13m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/571e66a42d41487fbf9185d3d31f5ccf : SUCCESS in 18m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e90c5b3f5dce4273ae2549bc46eccf1c : SUCCESS in 4m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/39f6b2d914984f5fa5204117f5d751e0 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7aecf4338da4c40838871ba398916a5 : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6f2b3d8b25f149e68e100c9a657ffb2f : SUCCESS in 10m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/48e8337d3daa4483a6343b123a2cd821 : SUCCESS in 9m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9cdc4111d2d465593b4d1d6fdb7b3cf : FAILURE in 11m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b73027301c0f469c9b4ae0d8b4e12012 : SUCCESS in 1h 51m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/affe4d3b302e45898e1e4fc6e11723bc : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a65641e20a942207834a75a8a38f068877cdc306","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 16:48:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f5a381f9389cd2189f8c545626831bbd3c7d2402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 23:03:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b2906b2c6e4403988a7558643bf743 : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d2ea8c68659c4e05b686b1de9f68a10f : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf53989827474d51b79aed77005b95cc : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99b6764607c545b8a8a0261f2e12829a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c6660c850bd41609e01943411816f15 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1663f2e4cb14b12b7d410e7c22f7173 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66f63ffac4a6458992ae247bf925dcac : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5840da6b4a0f4b40bbeccec6792414f2 : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/799110f7bbe045e9a1a44c3ef65e8a2d : FAILURE in 11m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36a0a62548c442c3b6da10c60750f40a : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb8757bc8c864a69a093da55e0f7e1a0 : SUCCESS in 10m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d3e1be2bda5487595ae21e5dd36d356 : SUCCESS in 8m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/524035d57ff64b4bbabe7b172f6c87ad : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39eaa3ec29884adb9a28c65d82bf0222 : SUCCESS in 10m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/25e86da998aa4c1ca309d1e2dad9b513 : SUCCESS in 18m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2f611c550bf455e8a67475ecc1a9474 : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/45de05864eab460e8f278447744174af : SUCCESS in 18m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/239d2cf5d907475d8cd12dcb1bd8d4b3 : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f93897755a334e91b3f535e2071eea96 : SUCCESS in 21m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c2f8f814052e47119707ed5d78306dbd : SUCCESS in 4m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9a64a49ee07f4334a688d774f110915b : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43a0ad7a21f44929b7648b7b2e26d2e4 : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b621721ce5d042cd9df30f6587726d93 : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d96ee9c63c854e33b747fec2de33fd6f : SUCCESS in 9m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e43f6c1ec014907af9919a3f251a8e8 : SUCCESS in 1h 06m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2252753fcab14127ada0a85229061653 : SUCCESS in 1h 26m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cd0e58aaa66f44fab3124351e82518e1 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":3},{"id":"b4582587a23e415a7cc6a1f632933a79ec2fb59f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 23:13:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"41433638f1378f3729aaf7a17d5d5d5f6fc81184","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-10 23:32:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c23938308d8583dd1d77239cd8b5f019557aed33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 00:46:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf0e603ff884b1ba9312e7b52bc2ff5 : SUCCESS in 4m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/21a215a238e54418a9a60c35c0964c9b : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc7086dcdd014c09beee61addcc9a1f9 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19edac739c5d49d283b9cbacd0b61ed2 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f6747bb1a164d8087311356830eedf4 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fdd9139365343c788836bee2f0da331 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a56f7ca9362941428fb94ed27e7bdc86 : SUCCESS in 7m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a455887e5634441abb6ad8868d87cebb : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c53b986c3cfb495ba19a3dfc2b792ccc : SUCCESS in 15m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04848bdfe0f94b8f98a3c23b0d62783e : SUCCESS in 6m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa96d0c78bd24b929076a8d1f8c46edf : SUCCESS in 11m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e9f3b3dc17245809657d1d641de9e24 : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6d64aa5d16b745c884f60327dbc34b9e : SUCCESS in 8m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9e9e89185f443e89b4ac66d7fbe5be5 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4dec984696a54ec38b45603227f8d2ff : SUCCESS in 15m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d608c40cd6f4dddbbab1d4557134558 : SUCCESS in 33m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0bfc85714cbd4f30a193293e8390d2c6 : SUCCESS in 12m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1ccd9a337fd14c1987c02fb841c4fbbb : SUCCESS in 16m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd027c0871e64cc890a03485ea94db85 : SUCCESS in 25m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/725c73ee3bc44115bcf2139b81a9430c : SUCCESS in 5m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7eb98f6a157f4095bdae5097d2d3c662 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a43e3e3553744df28e44d03fdb039207 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7e980dffd2ac4cfcb38c1e4ead31b9d9 : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2117e0a2014c438995bfe0f54eb215fe : SUCCESS in 8m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55e22c50e0214ba8a6faf22c2f95ccac : SUCCESS in 1h 21m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/01e694a12ef54d15972fa0b9800430ca : SUCCESS in 1h 31m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6c7a2c44f68f4c0c8f5eae8623bad7d9 : FAILURE in 9m 57s","accounts_in_message":[],"_revision_number":4},{"id":"969bfd1f5073605ebeb95f2102cadcc1bd23f700","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 02:13:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5407327dcfe4abf1253b939533a7cdb122a3837b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 03:46:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd84dc3228f4a16acd4115c204ef1a8 : SUCCESS in 3m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/33905a9d4b6b439483ebde2044a930b7 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09f28a1b0674f93b94b5841be4d2499 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21ae6f50f7e54056b588aaffa3e576df : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e6abcbcebfa40c38cbeb650fed9b2c4 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdbe5d2664864ef1817fdcc7fbda9ff3 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ffe005421e74d3bbbe08dc0ff0f30a7 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba0f806cb0494c048c200842c832adb7 : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9cda91c7f8cd455b8347b5a80ca81e1e : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9466a088dc244b469d7ccc43e2fef53e : SUCCESS in 6m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6e0735bcdc2645b9a7d44bbbe65aff91 : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3df7176ba4e45409fbabf0022b503b7 : SUCCESS in 4m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/392e93f3143d4801b7c9517acb7f3dfa : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cc9fa35d3a545449a9f1cfe5ef10d9a : SUCCESS in 6m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9349f9002a304297ba9fe5e78ef93cb5 : SUCCESS in 16m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5ecf194e6ea548c9bbf8bf8629e83cee : SUCCESS in 24m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5eba2d67f9a4c9f927910d174daa9b4 : SUCCESS in 13m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e43c200136bc45c4a91a14f566454534 : SUCCESS in 13m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e5919ffe57b94150a0667e6c361379c3 : SUCCESS in 30m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bfab02831cc04e078e5fa86f57a17f43 : SUCCESS in 4m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5a54740294e441c7a658a72977666e75 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6c58e5f740f475dae8b8de9cdb146f2 : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5ed10cd25dec4075abbc1e2818d585ff : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a31951e04bde4d9098997a50af131eec : SUCCESS in 9m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5750dbfe4744e22b19a4b21c83adb08 : SUCCESS in 1h 13m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1904ecd6e7ce436999ce1f5118fe2fa2 : SUCCESS in 1h 29m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/444bc708ce754d48a81478ab2bfa8006 : FAILURE in 54m 54s","accounts_in_message":[],"_revision_number":4},{"id":"24156ede9c3604df6881f71c4c06a38c9634d994","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 15:53:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2d84138b09f558872f38a927992e1a7cbba74277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:31:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249a93a413eb48f99bcf47ef77a596e2 : SUCCESS in 3m 53s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0c110af6a8fb4455bbd46449eb7c5c33 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed521f7c20f46b5b448aa9a18152e99 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e2d810fa5a54471bfcd61255275fbd9 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8db152930fe4cceba288d35ea3581be : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c79a5ba43b7d4e7780e0b5fea875b320 : SUCCESS in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/502c4e27ea674a5c84f88cd0bdb8f241 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/666bd70446344309883930a29aa10837 : SUCCESS in 2m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6bbffa3b2c3048ad9b5bc835bd952558 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6ef408d796e41dcaf390e42c2806e41 : SUCCESS in 6m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e520514560824236b0af3674f8e34369 : SUCCESS in 7m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c89e1f44709d4f5e9ea1e7eba4521f90 : SUCCESS in 4m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7f652a316a864a7a81982acff0b3bbfc : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09d8f95357aa4ff6932f8dcbb76290f9 : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d5a25537e926413fb334c154fb012022 : SUCCESS in 14m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/763750e3adc2408fa85d8f0f2ea9a4ca : SUCCESS in 22m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27b021fac8d14f9198242a80c4c19ee0 : SUCCESS in 12m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a616245c8a0d4c24a4ee5c15b56bffdd : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f26fc662ee7b4e7bb9cc70fe7a8f5eb5 : SUCCESS in 21m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/603e56c12ca542c8a8e846ab373b8c60 : SUCCESS in 4m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ca7ff5c707884980b0fc822082e44006 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e48180230c7147109d31b48e3c11920c : SUCCESS in 5m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/341b4f9d5c6a4e538a05126b1333a529 : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1de1e564213b43909bbcaa9f0ca041ad : SUCCESS in 8m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93cfea6433aa4badaa70e369e7deb3c0 : SUCCESS in 1h 16m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ecad5ee1f10446679c1abfed1b88d1ff : SUCCESS in 1h 30m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/81d11b2ae1174890b1380a67634876e4 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":4},{"id":"96c17acde0bafb9c5d2477611d267d7c141b01de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:32:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d9f3f15e3a072df3adccbca5d9f0658c68e48e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:33:18.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99be9cc9996b422e99b7fbade03e6e5b : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8105088fef4f40b1c27a3b8dc6708f : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/963adcc739e94aa89f5cc4d81e4462f1 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0467b2c65006425cb5f70073629e2d3c : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4881747f8bb248e0bf6e35ecc136f943 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8d4348cf3bc43ce938d1efbc1359d90 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/852ba5a721be435fa7cddd065c29a6b0 : SUCCESS in 4m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c04f3e7bc7994baf85d66e0e8660a632 : SUCCESS in 10m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ddd69627475d4562857ceb0ddc8931a7 : SUCCESS in 7m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f2315728c6b400cb27e0cdc97118efb : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eee5b863e44c4298bdb2a99a144d8432 : SUCCESS in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f15708abd16942c1913dcd6765489ccf : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f8a954314ce04bd2b48db700f06c0350 : SUCCESS in 7m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5b7b066a964c4dddb583c00815b49f8b : SUCCESS in 21m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7beddb7b1fc640f9b958f344b2835a4b : SUCCESS in 19m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27ff7c6576b84916a8609a524b32855d : SUCCESS in 15m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/66aecdba5e74447c8793d01766bd8117 : SUCCESS in 16m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db3e9897017b41f0999b15f71aba8433 : SUCCESS in 18m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a3c252e270ba443ab567d1b5400334ae : SUCCESS in 6m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b2354da0303243d1817d7b2ce4cc19d8 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d0640df1ac14148961a097c9d1a7cbd : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/46598e04f155428f9ab62ba6a1941e0f : SUCCESS in 11m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fef838f3e8de4f718aa46d7e6195dec1 : SUCCESS in 13m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b72a5ddecb184c9aae800cf50f56fe83 : FAILURE in 1h 20m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/34207cfc195049b6a2168fdadc63e7dc : SUCCESS in 1h 56m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/895cad5ed6a949be98867e378adb64d3 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d6d30bc5bc823272dcd0e9b5fc097e5819da7263","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 20:21:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2a3c6f1a68abd7867b6d4ad1bdb929723faba9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 22:06:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7a5832208042fc9232eef17038087f : SUCCESS in 3m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c3960297468d499a8af01934587d2ca6 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957b02a3a3a54b13adeafffb4fa64fbd : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6820b7981bb0466c84e9a49d5ca9e139 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01fd157f74a748f786aa2c7a69ad4c6f : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d83e1157b47346fd8b75f9ba4362fdcd : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ceb2e2ae959c43f4817c04c1c4348c0a : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98e623cfee554d1ea95d36a1b5138e49 : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e7c8d4685dc14c4c90a0c170d4ab1135 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9ae9ef859f74bafba5e3a5e7ce4853c : SUCCESS in 7m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/32575039a31c4fbd89739a8c686d6ec7 : SUCCESS in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41d1f89883c94c3690a112675e171301 : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/665a936d2761499b81c60f1839a74d61 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07e20776b75a4b15b869fc5348f08fc2 : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e317a9fd940a40c490f35e3ccbe22421 : SUCCESS in 16m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/29e99235a3ab4a0aade0f33d1df8e674 : SUCCESS in 28m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d51800b2daf64fe2a3d43cc2def4bb9a : SUCCESS in 15m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e08405a054e84686be72364964a6fef3 : SUCCESS in 12m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a4395432d004647b603f47afe1d3eff : SUCCESS in 17m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9007445e2664faea9353f069cb62cba : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4c6f34c67bc54cf38ffb27f2763f3460 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0bc57226b80349dda5d8e7849cd49a6e : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f0ba5ff6f074387b605d26c415f367b : SUCCESS in 11m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e022599fe2a245ba9cefcd15b0313d0c : SUCCESS in 11m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95835332600b457ea8c4de886d495faf : SUCCESS in 1h 06m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3af03ae0623343ce94609961505d8b0e : SUCCESS in 1h 42m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ba69cb88cd574dc4b6aa2cf6713b37ac : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":4},{"id":"6e7f825f974f5a07dd85a8cfc2d8ce7179e0fb2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 22:06:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2307fb4849cd85afb6cca8b6e9f4aba9714a4884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:20:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c07a6a0e5b5d3e543b972361fb57aecad8c35b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:20:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028e4c3d660141fb8c226061f2cf4f83 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b335245700d74c9ca2f40aed7cb78a25 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52f5c6f9b8bf41de96e4256416be2fed : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7304653259ee48c3ae170164ca907994 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9f3bd727a094601a7e8b82bf9f1ab47 : SUCCESS in 8m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83c31f42122d4f7cbfa7c18f7cdddd38 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/506a48dfd524493cb3d44b8dba946b76 : SUCCESS in 7m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bb6651da4f5f4a0abbcec687de773c4b : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/527bae6a22124657bebfe1040cc1aa89 : SUCCESS in 8m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7840b41bb221471dab3ebd4f712d85ac : SUCCESS in 11m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7bc5b00c249f4be192e821871e19421f : SUCCESS in 8m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4395cb98d40f48a2962e2fd3a66efdd2 : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35128336d15a40b4842dbb986d110fc2 : SUCCESS in 10m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6470e8e8105451a904c2af3e15899c8 : SUCCESS in 12m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b54a41b632f94ba59b67086365963bcc : SUCCESS in 22m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ebfd4c1f31f54a0baed57956bfd88582 : SUCCESS in 12m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/48d28cb1677d4409829d7f1efd5aeede : SUCCESS in 22m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc3a3e3583604f669f99d3aa9005cab7 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/953cbc913d5c45d9baea00e7fdc6c9b5 : SUCCESS in 4m 05s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/91d68c3b104f40719e428f71cae5b71e : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f08500c72e145d69de1949a1979b47d : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9b72197efc8f4210ad543f21d4675157 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/91c914dd69e74f39bf77dc100f23fe76 : SUCCESS in 8m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0b815cb97bf419fb28a00cda880ce42 : SUCCESS in 1h 13m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a3830739e2c0494bbe716415a4e82685 : SUCCESS in 1h 29m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fb08a216c6bb4def9da65c315db7c3e5 : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":4},{"id":"ee90cfd693c43a18a1160b87e3e09f50c92fe912","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:21:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291af5afd64b433a810817e82b34b449 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"12ebf8e57d6d16e076343d54f005784fb24e9cb5","revisions":{"74b9897d7422d72e8e18020c799dae81322ca974":{"kind":"REWORK","_number":1,"created":"2019-12-10 06:15:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/698145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/698145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/698145/1"}}},"commit":{"parents":[{"commit":"29928fd73545fc73da04903ed9d1c039e14c58b4","subject":"Merge \"Add python-cloudkittyclient to global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29928fd73545fc73da04903ed9d1c039e14c58b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 06:15:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 06:15:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I50dbac0f230a0ce9d5608dba18d891b77002d785\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74b9897d7422d72e8e18020c799dae81322ca974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74b9897d7422d72e8e18020c799dae81322ca974"}]},"branch":"refs/heads/master"},"20097726845414159dc23b9f44338f1884cffcc8":{"kind":"REWORK","_number":2,"created":"2019-12-10 06:40:24.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/698145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/698145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/698145/2"}}},"commit":{"parents":[{"commit":"29928fd73545fc73da04903ed9d1c039e14c58b4","subject":"Merge \"Add python-cloudkittyclient to global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29928fd73545fc73da04903ed9d1c039e14c58b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 06:15:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-10 06:40:16.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I50dbac0f230a0ce9d5608dba18d891b77002d785\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20097726845414159dc23b9f44338f1884cffcc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20097726845414159dc23b9f44338f1884cffcc8"}]},"branch":"refs/heads/master"},"50c337a5a13f5ea3345c6fe5a6c8d5c7dcd814d7":{"kind":"REWORK","_number":3,"created":"2019-12-10 16:48:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/698145/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/698145/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/698145/3"}}},"commit":{"parents":[{"commit":"29928fd73545fc73da04903ed9d1c039e14c58b4","subject":"Merge \"Add python-cloudkittyclient to global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29928fd73545fc73da04903ed9d1c039e14c58b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 06:15:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-10 16:47:57.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I50dbac0f230a0ce9d5608dba18d891b77002d785\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50c337a5a13f5ea3345c6fe5a6c8d5c7dcd814d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50c337a5a13f5ea3345c6fe5a6c8d5c7dcd814d7"}]},"branch":"refs/heads/master"},"12ebf8e57d6d16e076343d54f005784fb24e9cb5":{"kind":"REWORK","_number":4,"created":"2019-12-10 23:13:28.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/698145/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/698145/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/698145/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/698145/4"}}},"commit":{"parents":[{"commit":"29928fd73545fc73da04903ed9d1c039e14c58b4","subject":"Merge \"Add python-cloudkittyclient to global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29928fd73545fc73da04903ed9d1c039e14c58b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 06:15:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-10 23:13:18.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I50dbac0f230a0ce9d5608dba18d891b77002d785\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12ebf8e57d6d16e076343d54f005784fb24e9cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12ebf8e57d6d16e076343d54f005784fb24e9cb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
