)]}'
{"id":"openstack%2Frequirements~698333","triplet_id":"openstack%2Frequirements~master~Id329f5cc787a723d30558dd7576846eeda669337","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id329f5cc787a723d30558dd7576846eeda669337","subject":"update constraint for neutron-lib to new release 1.30.0","status":"MERGED","created":"2019-12-10 22:26:27.000000000","updated":"2019-12-11 08:06:24.000000000","submitted":"2019-12-11 08:05:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698333-1576051506996-a71cd6a1","meta_rev_id":"74fdeb90be929eaad129c19056a4ddc88bb2b31a","_number":698333,"virtual_id_number":698333,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 08:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-11 05:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-11 05:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 05:02:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-11 08:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ccf54fe57678ece10395c783cea3fc100e8c62c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-10 22:26:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c87dba5f0181bc8d6c800725a17dce32182ca59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 00:13:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce7ab5f2e9a40d09b39bbbcbf3c3ba1 : SUCCESS in 3m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a7178c86d54f4d22973327b4d1479a84 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c66fa372d7e4adc8660f9686f2329ba : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4405df1493d34cdbae53339c53d98383 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af4469f5077c488ca650a45cb5efc3e3 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bec52aaac33b4f43ba29cefe99bc653e : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/455c3f1fa3ec4b238ad03939d219ea8c : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f6cc2fbbb1ec49729d7b32e103255a06 : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e32cc9053c554295a217e335fd6c732d : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de919af45bb74f08bd0464149f572048 : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90568c44e04b4060968eec925da63ab1 : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/62eda34d26d649ef9f31ef8f0169657d : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6750c1336a884fa5883ac50ba0ea7b96 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a8e1d805fae42c497f712e3f8e6179a : SUCCESS in 8m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a4305932620546d9854f28203d502ee2 : SUCCESS in 16m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a1e30381e83f40a6a608154f580e733f : SUCCESS in 26m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a353a8d387f47e3be27bd8d9f7528d9 : SUCCESS in 14m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a1689585736347a6ad832757d4849763 : SUCCESS in 15m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5d349a27a2240139131f5d865d4eec2 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bda12a0d81e3423f8df9afaec6213d93 : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/294daff3e8034fe2a0ac666132b260dc : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb1056d80e214791b1d97688f4ff9552 : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/697494df99654d29946853d701a02068 : SUCCESS in 9m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/095590e87a4b4823b4d8ca628831e29b : SUCCESS in 11m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a2c40a63bad49cdb30c5e24b194406e : FAILURE in 1h 04m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/72a34ce55413462f8cf678ff65bd4a8c : SUCCESS in 1h 39m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/783669cb6a75475185a617df432beac4 : SUCCESS in 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a215acef2fab4d12c0018e511c2fcedea4acfcc1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 03:22:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae8b1b7b02960e269dfcb39d477003c5d48e8ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 04:56:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df995743f4c64b8e9e862f1bf6a908bf : SUCCESS in 3m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/af17674c9ea74690b7904778a60718d8 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c709ad3f4a4e2b907e758498e8afec : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfb940f3250544c3a84ea31a7627edce : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3a0f96317d34ea4a08a260ca215c5e6 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36f32e09b3a841439099c388ae21bcf2 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56d97e18db3c4eabb79aa8abce4c7dbd : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c549f654b47f492591499e4f9b3c2bba : SUCCESS in 2m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7410537cd2f14bfcb072edc3a8da3031 : SUCCESS in 7m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e919d318241e4eb6a8dda9a441e281c3 : SUCCESS in 7m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/813716c83cc343419b69a7358dafe1e8 : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9deb44cdbbde4014a6fb43508fb00cf7 : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/975c69330d56445d8313c9156227f465 : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ec880a2398548f08f41326e44ee8bee : SUCCESS in 6m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/384d4e0e3d3a4ff9ae048c41ccb3d82a : SUCCESS in 19m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/958a596c5b7845489f04606d79a0b4a2 : SUCCESS in 24m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c299612368df4c2c8e520d415b5b041f : SUCCESS in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b615aae9abb84b0188e94ba7b49b4526 : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47eb8c4a57cb4a3886295b947d1f17c1 : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/67a3b3b324db4db78edfcee43cd0af63 : SUCCESS in 4m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bbe8c1b21ba241878e755d8721ef7214 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4132f9e6f021414eb169f0073579761e : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/794e1099f77b473ab93aa5c2564941b2 : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abab6a5df4434159acff9e3ca65b22af : SUCCESS in 9m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2000d0f5fa1416b92d247fcc396501c : SUCCESS in 1h 07m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd07bf7580c84e60aa0458e8de84c347 : SUCCESS in 1h 32m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01b95d8627c44d8793f45ed7b576352f : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8fa0c182b1522fe11bf33ef54b3f761d942380f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 05:02:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"535d89ccfef4fdbe640fc37eae6c5a61835a8176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 05:03:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2da78bf825ace3965e661c555201d587de0243b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:05:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5334a459414e11a396aea20a9a38c4 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a88bf53f0847bd9e2d67fbce82b1b7 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61053b8ecac04c4bb6102c7e2b1fc62b : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/276f8906b47d43c9a9ba7d071c9a8fc6 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3681ebc79ee4ca5b09f1b33831b84fe : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d17ed2716c77443f99c4ac077d93e608 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f8a016b64fc49808231717fc02dc449 : SUCCESS in 5m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2ea79a61ff7f4e938dacadde2166793a : SUCCESS in 7m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3bd11f57907d4741b42b0ab39214845f : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/001dad049bc74f46a0f580d9be87e8d5 : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3dc99bdf4e1045cfa887fa7912dd962d : SUCCESS in 6m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a5302be0261d4628bfbc1b53e803ad2e : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6fb1a40e9fc463e89b768034b92bde2 : SUCCESS in 8m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9d932b66f2124b998976116e5fc6e446 : SUCCESS in 13m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c71159afd5a46dc85b2134fc56b67e6 : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13116f1396b14e768136f71777a47ecb : SUCCESS in 11m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7670f66ef7fe4adf87c9476c86975320 : SUCCESS in 12m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dabe333b18394222bc523ada1019a47b : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6b71022891d84bdb894b95c3935db0b4 : SUCCESS in 4m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1a1cdd0e6bc34bfc9b699662f5b1ad78 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab33fa7cd46e401b881f0b7d92fbfa48 : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8c498d79e68c4381b8eb905f22843e8c : SUCCESS in 9m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d181d103127c42639a4a54a5a871198c : SUCCESS in 8m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88f1d2df7d9844ba8c28aaf2b388e8df : SUCCESS in 1h 33m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff314afbca0f4053b6b1388791df07c3 : SUCCESS in 1h 39m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/11c3ab0f68bf4c99b4fe26ad2bd2477b : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0c85e0c6dda2b9d8c28c2c83d1c024abb086262b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:05:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"911e24fceb2355ddc700d344c88d14487a7f347d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:06:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f60ea61bf14c65b96e37b81178ecf9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d345dd8c77281495c7b0b6b97205e7b50d2fdd6","revisions":{"6d345dd8c77281495c7b0b6b97205e7b50d2fdd6":{"kind":"REWORK","_number":1,"created":"2019-12-10 22:26:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/698333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/698333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/698333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/698333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/698333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/698333/1"}}},"commit":{"parents":[{"commit":"29928fd73545fc73da04903ed9d1c039e14c58b4","subject":"Merge \"Add python-cloudkittyclient to global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29928fd73545fc73da04903ed9d1c039e14c58b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 22:26:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 22:26:25.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 1.30.0","message":"update constraint for neutron-lib to new release 1.30.0\n\nChange-Id: Id329f5cc787a723d30558dd7576846eeda669337\nmeta:version: 1.30.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: Ia2e394c8d6b126da10b55aca46b9187aa0bf6384\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta:release:Code-Review+1: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta:release:Code-Review+1: Bence Romsics \u003cbence.romsics@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d345dd8c77281495c7b0b6b97205e7b50d2fdd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d345dd8c77281495c7b0b6b97205e7b50d2fdd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
