)]}'
{"id":"openstack%2Frequirements~698405","triplet_id":"openstack%2Frequirements~master~I19beb1e90aa2d95d71f1b5ae1a31410462953942","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I19beb1e90aa2d95d71f1b5ae1a31410462953942","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-12-11 06:09:49.000000000","updated":"2019-12-11 16:42:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"85da6b32e381e30504299c0d3270e6c2504c4132","_number":698405,"virtual_id_number":698405,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-11 08:41:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 08:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e2a4403219b66bf2ca83fe018d2aed27592d901","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-11 06:09:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bffe049e5f5f3d75bb088daa3236679e34900cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:41:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604d8ed3f0044bf2867198b35236d123 : SUCCESS in 3m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d4c6227bc9094fce852203193f065dde : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3d800c7d9d449c9738cb4e4e069430 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae482d4671ef4833a3822f0ee7b376d9 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5ddfa4e1c1249baa62751242c1c8e4b : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a1372865c65448d9c453511c48b5af5 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98245a67a72b45c0bc3817ed40dd6728 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/649ab63e92be4f88bb54b26d949345f9 : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/71b22b7435ea4609a159ec53eb054db6 : FAILURE in 6m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d382dee152a04ad8ae0a6ce9eb46ba9c : FAILURE in 6m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ce808ff4799e41de861edb3f1da3446b : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/667fd6fa33ad4ec8856c9626640f1ace : SUCCESS in 4m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/befca5880b7046189766acc087a121e3 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af6cba20488c4596a3e698a81f3546cc : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce386544c8e9481a8e2e6e53ba02cc6a : SUCCESS in 15m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d6800333284549f59296331e8f0dfc0e : SUCCESS in 24m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e34b774dd0b477c9c32b971842c87b2 : SUCCESS in 13m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/313f0af3a86340b6b6cf643ac751a5ef : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf9ffff26d8e4b5881d356faca43b79f : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d6c7d7f1568d4cd4819f569886f0ab8e : SUCCESS in 3m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/80875e654a62453eabeb896879c5a1cf : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3a268b56c314245b7ff82f8f62b03b6 : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c4cb7db530ef4ffa864e5a2ab2d874d3 : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c41673fe2394c6da0dbd601a282a61a : SUCCESS in 11m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/631be99bcf384748a5a6fef8b94e641e : SUCCESS in 1h 11m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4202a5c1e76b440bb55fbb7b80d96107 : SUCCESS in 1h 45m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2dfd1e8d696047f3911976ce8a7685ea : FAILURE in 59m 36s","accounts_in_message":[],"_revision_number":1},{"id":"644608419feba7a9b2bbfda7e451eadc98db3123","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-11 16:42:41.000000000","message":"Abandoned\n\nuse the old one","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29a47e213f89176fa5ecac69108b2ed86beaa671","revisions":{"29a47e213f89176fa5ecac69108b2ed86beaa671":{"kind":"REWORK","_number":1,"created":"2019-12-11 06:09:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/698405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/698405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/698405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/698405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/698405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/698405/1"}}},"commit":{"parents":[{"commit":"29928fd73545fc73da04903ed9d1c039e14c58b4","subject":"Merge \"Add python-cloudkittyclient to global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29928fd73545fc73da04903ed9d1c039e14c58b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-11 06:09:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-11 06:09:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19beb1e90aa2d95d71f1b5ae1a31410462953942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29a47e213f89176fa5ecac69108b2ed86beaa671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29a47e213f89176fa5ecac69108b2ed86beaa671"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
