)]}'
{"id":"openstack%2Frequirements~698672","triplet_id":"openstack%2Frequirements~master~I679f4b1a32c9f7bda5c65eeed141fbabca3f922a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I679f4b1a32c9f7bda5c65eeed141fbabca3f922a","subject":"update constraint for os-traits to new release 2.1.0","status":"MERGED","created":"2019-12-12 10:10:10.000000000","updated":"2019-12-12 17:12:31.000000000","submitted":"2019-12-12 17:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698672-1576170575281-8734d19d","meta_rev_id":"725a431d59d0ec9398366a47b81705d252c5bcca","_number":698672,"virtual_id_number":698672,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 17:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-12 13:32:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-12-12 13:37:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 13:53:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:32:44.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-12-12 13:37:37.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-12 13:53:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 17:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f925dd66c92a663ec0a3342c88b97f78e16d8ef1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:10:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e144238940533acb89ffb7cfd0c958b16edfaf76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:00:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc32fe5203b462a9bef2e3896b2df47 : SUCCESS in 5m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d85f0ae5b4864076b0189bdc5e4f3373 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507f4df00ef14308bba6d86131377539 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50091ad85b2b494e995b1c0c3760a639 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbb703cfac6f4c3d801404cce1830599 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97449a9fed65425d8f6aeecc2edef6a6 : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14f189e5f7634562b87705c620b0012c : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f08ec169f8a40ba88301e1e1aa410fb : SUCCESS in 6m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/db05ca0662104ff98d63959547c92fbc : SUCCESS in 8m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/98a7797c7c1e4c708fe7afc644710cab : SUCCESS in 13m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/03484eea163f450f86db22147c073bb6 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b0b8277e997a41758cce382eb329b703 : SUCCESS in 7m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6947f346da5477cbb21875a55c2be76 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b74be44a16849ef84c7608ef5a3e6b2 : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e397dad951004b2d9ca1a4e4e4d2e67a : SUCCESS in 14m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a77ed13504b4bdda82222e4b9cea735 : SUCCESS in 24m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f4078f291476420498aae28fa0faa5b4 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/517c59ab84db4c06964e64ff115b578b : SUCCESS in 17m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccf77f82710443e6985b691b339a4aac : SUCCESS in 21m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/08d89f6a1c77474589453cf1a5973289 : SUCCESS in 7m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8a4c260161014abb8aa3eba478f86863 : SUCCESS in 8m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/234808ee5ad94a859a95373cada91dbf : SUCCESS in 9m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6ffc30c7dc0943068b2158b3b0b9b999 : SUCCESS in 10m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ffda3b7e73b144cea8e7fdb729ca7c18 : SUCCESS in 11m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8492d4f45bc84557bc61011ffea67241 : SUCCESS in 1h 18m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/030f6965f5584e478e70bfbfb0afcf02 : SUCCESS in 1h 44m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/47e9ef71e48b4b27b2e9a56d9de7c28a : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"73af21f81a855d389da4e7ee2fcb26d35817e47b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-12 13:32:44.000000000","message":"Patch Set 1: Code-Review+1\n\nwatching","accounts_in_message":[],"_revision_number":1},{"id":"8e4794cb786a6b1342ebfc05ff618fd35b7e43df","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-12 13:37:37.000000000","message":"Patch Set 1: Code-Review+1\n\nYes please, I need this for https://review.opendev.org/#/c/695220/.","accounts_in_message":[],"_revision_number":1},{"id":"d901afad144d45762f29779d81aa9180b8462265","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1c307750d48a5bb6ebb0636552452e5524d3b94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:53:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4553e7bca7e9a4295cc505747c6c475c4a28be2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:09:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"74036b1cbd3243b45f3fd78401fb10362c65e83b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:09:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11487357db0c48f6a3a1fe1e74a2ec3d : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bdc3805d7234c50baa7bfc97385a19c : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19b38521fe8b4a65af106e11a81489f3 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a60361e75fa14885aead67b17c288907 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b41f1b3e6914db6af2f1477fa98b8d5 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00806568d5d54a1099aaebc9123420a5 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc876b3d9dbd4f6daf9377d89ef58a09 : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2437d21cb32c45839f49caace8243125 : SUCCESS in 8m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0bf212063c3a45968c2fb1d3b48b619b : SUCCESS in 8m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/96fac7e098a54ab69a1b572f8e216f15 : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5299c491634c4a2db0f29afe0a867f0d : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/be9a64a42fc44422a36678e596370bbb : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1fe60a3270f40688e3e0f0564291da7 : SUCCESS in 9m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2dd61092a24547f88f61f9f33a9e526d : SUCCESS in 22m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ea4028d43274300b8046a8284d4d362 : SUCCESS in 21m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/43b4ba2cace14c92a1ec7642f1cc5847 : SUCCESS in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8469dff3b318454e9d40d342bb077dab : SUCCESS in 16m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/566720bdba0f4385ac85824b0396757c : SUCCESS in 16m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dfb6e95301204671921bfb01e8866c71 : SUCCESS in 6m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4d1a7b95cd6048f69e14b9af2347572a : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01188199a47446c993431e66ef64f359 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/09ccc9d9ccba476189aa73b4c42d78e6 : SUCCESS in 10m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5be3b8d22e2149eca3dbafc428351184 : SUCCESS in 13m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bca4c9a2c11548a29490fc0a17d7d0c6 : SUCCESS in 1h 17m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/793685380a834fba83fe09b3af16fb7d : SUCCESS in 1h 56m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7e2e663088bd438080fcff01fc40bbaf : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":1},{"id":"32edd87555b5d8ead0cc4e5e0251ed47374f623f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:12:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c79ac560e0440838fbbaaf75206cf91 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8ecb43eaf0eb75ed18488463a87691c5c4a1a75","revisions":{"b8ecb43eaf0eb75ed18488463a87691c5c4a1a75":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:10:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/698672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/698672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/698672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/698672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/698672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/698672/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:10:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:10:08.000000000","tz":0},"subject":"update constraint for os-traits to new release 2.1.0","message":"update constraint for os-traits to new release 2.1.0\n\nChange-Id: I679f4b1a32c9f7bda5c65eeed141fbabca3f922a\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I6b9ebb61c2c5483762cd01cd9a320b0ec32accae\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8ecb43eaf0eb75ed18488463a87691c5c4a1a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8ecb43eaf0eb75ed18488463a87691c5c4a1a75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
