)]}'
{"id":"openstack%2Frequirements~698676","triplet_id":"openstack%2Frequirements~master~Ic80732af74fadf1f96f79b958107937f7eba2644","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic80732af74fadf1f96f79b958107937f7eba2644","subject":"update constraint for monasca-common to new release 3.0.0","status":"MERGED","created":"2019-12-12 10:13:24.000000000","updated":"2019-12-12 22:53:20.000000000","submitted":"2019-12-12 22:51:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698676-1576191114291-600d80bd","meta_rev_id":"e003a14123269c65443d2e39e81717f26691c42a","_number":698676,"virtual_id_number":698676,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 22:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 15:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 15:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 15:45:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 22:51:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e8c7eb240b611d31ce5b6bae099882ffaa5b8e0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ace42757e78ad847971f48bfe89f7ce2c87c6ac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:42:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4eceac211c4f47b691fd2b55041a11 : SUCCESS in 5m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/63a9903376a9430cad43d9ccee6f1f71 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206960489a074c63b5267d351247df84 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77fa49addecf479084104ece1587041c : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d33abf87620482892489a54ac73401f : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c226f47dec0c4a27b467f6afa948e310 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50668ad3ca1a48db8479f9a43496cd95 : SUCCESS in 5m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d07d82e110640eba7cac7355850d260 : SUCCESS in 5m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cc738ce030b149f2bca41df7d980b539 : SUCCESS in 8m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c8f01a0c1cf9456e85468cf173b49c71 : SUCCESS in 12m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/758b2fcb712d42f6b0b4a4f86dfd4148 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f552604324454bf8bc06509d8be4242a : SUCCESS in 7m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1e1cc03757174907b88f8e830b88ef6c : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f456bf30d213496287018f09a073fddb : SUCCESS in 11m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e8f63a763c3f4ccbb9c394e2215ae3b8 : SUCCESS in 24m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5683c4aff6474319ab956c3f8c679020 : SUCCESS in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46cf5f82d11246e78034555dcdf17de0 : SUCCESS in 14m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/318bde9117a14cc98f2ace5f106b2460 : SUCCESS in 19m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ef0bd45c91a401499da3a07bc3adfda : SUCCESS in 27m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fdea3dcd8d694915884aaa7d326071c8 : SUCCESS in 7m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e9c4376abed14f59ab6fe87843bafea9 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b18355d4f3334b2fa44f386b81045532 : SUCCESS in 8m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0ad810ce372e4aaaa63e5262a31d6593 : SUCCESS in 12m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bbd8f033930a41ab96b98731590c04cc : SUCCESS in 13m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b0e27d16cad4fc7bfaf07056ef538b4 : SUCCESS in 1h 06m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d9356a52c93048bf9a631a1f7c0629cc : FAILURE in 1h 26m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/11f7deb03983440183e519349b4f1df2 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9ab24f53188e91adbb7b93ab5e097f339bae6c22","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:54:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"204db7af38ddd8cab864a1571c6078b1f9b829f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:40:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a296ce9a6d942d9a0ba03543bb812cc : SUCCESS in 6m 36s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/951dc030fbc64d989ee14cb2568e021d : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b661b497e71e4946b0ff164616d3cf18 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/607258bf62914bf2b45cfad92065c510 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a83ec99da84ee58620405e880d0c58 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa2e4b7a78684b80958a6b6632d16ce4 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cf5ca9508324740a365952e6f6f1faa : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e27b4c6e8e247b2a782ca5e716e6134 : SUCCESS in 5m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/49b5b38fbca44250975d5878f0fa3238 : SUCCESS in 7m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d0cb38f50ed4fd98468aced7d6670e6 : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/be1a5e6410d948f78b65646edcee50fe : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e4db3d27e194d6a95984d383c2d5c4a : SUCCESS in 7m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3434524c0c2d4d5e919a3248805536ef : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad42f2063a174df89399ad7dfa4e4dc4 : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/483638929a0b47a3b7dcd26fcd90603a : SUCCESS in 15m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15f2e4fae76a40aea4bf8357c8967e92 : SUCCESS in 24m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0df0711038ed491095b350135ffe764c : SUCCESS in 15m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bd5249d488134928bcfb82af1e436303 : SUCCESS in 14m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b02539a60ec4dea8003c76eddb09f85 : SUCCESS in 21m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ec78cda69a314797ab6f454703fb5026 : SUCCESS in 6m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/19a37b8a2e0a480eac57505656498322 : SUCCESS in 7m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ee4c1aadc064b638349a8dccec842fd : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/83d87eeb3ed64bb987624be3498c5def : SUCCESS in 12m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c00d21f5d58546329527efb0cbce738f : SUCCESS in 10m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9dc03f6f4574841bf0277f6618ce3ea : SUCCESS in 1h 21m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb9885e58e294269bb74d9ef95a7d793 : SUCCESS in 1h 40m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97a2525c2b244230bdab0ca123334ea1 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"26e32f3d5a6d10d1ef08c29d01d4c48e01f820f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 15:45:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c1a30f1782de8f7bfc2ed584a28d4b52b63ab61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:49:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d8b4629390850397cab65e4bdf76d82959caa5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:51:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f911d75934483e81f8b14a3f43dedc : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07fef4a1a6ee4178b09a0c7452136ef8 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4a0c3a02d614875bcde71066fc3bd41 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a7a40016701450180157f39d7bfd60b : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/111091a5f9364b5f9c81f7aa103c4cfd : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/834f89df62e745528eb55f52f7e11782 : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab26f2616c644cf7889c541d86f985d7 : SUCCESS in 6m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aafdb5693b304691bd7f7617a0366a0e : SUCCESS in 9m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/539df0b115c74434bf2e7c164906d72a : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16a80c1858f740d7a7ddc49fc0098eb6 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef85cf30532046d8840933861746d288 : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b0cf780746a24e20930abe028a2fa1ee : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0c65978293c4b73896cf7ff64f2cf0a : SUCCESS in 13m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ed95fe16203548a696eb7601a73358ed : SUCCESS in 19m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/54eeaece4802483c9e9bd51df4a259ff : SUCCESS in 26m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8732f7be1814af3b083f52660954530 : SUCCESS in 18m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d866a501e14d4cfc89dca1bf30159bab : SUCCESS in 17m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e37b5a73570d47c6a797bbbdd8a089a8 : SUCCESS in 24m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/570f46a4eee14f599dbe03d13dcc5029 : SUCCESS in 5m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e0226f768144e5c9af7cbb28d4dda2c : SUCCESS in 8m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a4edce7d0b247b39049c10ace1a5e26 : SUCCESS in 9m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/974c0db7c8244b0194a777d583b72bdf : SUCCESS in 14m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec6e85a7e268412c93ee47f8c966af65 : SUCCESS in 12m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/310335272da14be281d9282d20b1ee59 : SUCCESS in 1h 24m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5bc6b6e55fa743a9933fae38b19759d8 : SUCCESS in 1h 42m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/76397a71b5e847c3abfa07547f2e1a36 : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"08065022e30db274af34b27bb30aeb8bc13162e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:51:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"68011017d3bcbf936786308b627c96a94f08efe6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:53:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221d266bfa8c4793a249b694ee4cc44d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12c26bb28d6b84abbee38de46d7f0ecfdaa82ae0","revisions":{"12c26bb28d6b84abbee38de46d7f0ecfdaa82ae0":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:13:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/698676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/698676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/698676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/698676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/698676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/698676/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:13:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:13:23.000000000","tz":0},"subject":"update constraint for monasca-common to new release 3.0.0","message":"update constraint for monasca-common to new release 3.0.0\n\nChange-Id: Ic80732af74fadf1f96f79b958107937f7eba2644\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I8c00cd1fbfc1622647949991119652c119deee22\nmeta:release:Code-Review+1: Witold Bedyk \u003cwitold.bedyk@suse.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12c26bb28d6b84abbee38de46d7f0ecfdaa82ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12c26bb28d6b84abbee38de46d7f0ecfdaa82ae0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
