)]}'
{"id":"openstack%2Frequirements~698677","triplet_id":"openstack%2Frequirements~master~I2a50048679fbf6bc14a09315c0a97cc773f87b88","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2a50048679fbf6bc14a09315c0a97cc773f87b88","subject":"update constraint for oslo.reports to new release 1.31.0","status":"MERGED","created":"2019-12-12 10:14:03.000000000","updated":"2019-12-12 22:57:04.000000000","submitted":"2019-12-12 22:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698677-1576191309633-91269b4c","meta_rev_id":"6bf48bdcbf0610af6e03d6e54bc385db7c9e7338","_number":698677,"virtual_id_number":698677,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 22:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 16:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 16:08:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 16:08:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 22:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15a73f3c1d349c7adb96cc86656d8ebea8cc9ba6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ad926449173f535640939ec7953a6d10a260f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:11:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ff9d1252594a2e92831d138965f858 : SUCCESS in 5m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0a9eaed7c9ce45afb552a613e952cdf6 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911c59c6993445dd9e8a206f03fec878 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a981bfcd72a4d98bd96f022eaa15499 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7afe956d43634219833ce00192aff3a3 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50d6fd57f583423fba80bc4c3d588fb4 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0907ce83bac94521962d7660fd2e655c : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d7930cea6cb46168d8eb067df16e2cb : SUCCESS in 5m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/39a6c50858ae48cab12aa94a8946e4f2 : SUCCESS in 8m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b3a8e2d69f834a50a3315622255e4828 : SUCCESS in 8m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90ac15d3735c4a0d90a16ce2b13299c6 : SUCCESS in 11m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b9b40bdd4774bd9850e5ee8756295dc : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/36eba165f2bd4b13bad43385fd5ca8ff : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6906dbb756a04f368e3f2f46b2738f50 : SUCCESS in 9m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2f5de15e5b524f72bba64368c24fb09d : SUCCESS in 21m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fbb152d1b89e49c29457f2efdaaf24dd : SUCCESS in 25m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a70fa1937c804cd9875dbaf6a0f4d874 : SUCCESS in 16m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bf9a8d2cb43641d093ee806f129674d9 : SUCCESS in 17m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1db8fd6a11f84ec3ad4363d33468ece4 : SUCCESS in 17m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4144143b91bd40c08c1fbff44670f4bc : FAILURE in 5m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c48787774b55498eb800d0a0c998df0d : FAILURE in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3dbfc93079d540ee853681fc6007e6b6 : SUCCESS in 10m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/456d66f6069b455cad89e7338a5d477c : SUCCESS in 11m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d1c11cf0c8a4d749ca383b23e0bc406 : SUCCESS in 10m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d81fd5d7754472d9bdec876237c8a0e : SUCCESS in 1h 19m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/152dfcf0ce2f493089df99c142d94666 : SUCCESS in 1h 45m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8474d845f9d74f56ade2e719fa78ad30 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9e9e7481bfa9ce25061c2d596c1a6d69099e9111","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:54:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25aead1b4bb1dc28d14f2627fd336b7047c71f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae945d44ced443dacd60bf2f92225f6 : SUCCESS in 5m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0c894b1b62084fa9901024974d19a114 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078b9251b74746f0b7029ecae43b8590 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cfcab9d7b554ebb89874c662fdd36ee : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d22068ae4754605a705a64ee862e426 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20948c1a166240ecbeab697cfdbcc4eb : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d0a1500ca0d4f3b8d0f4af28bca00db : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ba90f9fadce4b77961dacb41966fb77 : SUCCESS in 4m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/506b4740e3524676ac52854d11c78c97 : SUCCESS in 8m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a50ea1a3a3e4915a4c818a004f19f08 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f45951321083464c983b43986d488352 : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/38f7c80ca4434a1e82a3b9b81d7b9e99 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fbaec03778354d139b8cef2faef7a86d : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e6850b1c9a54661b19c2c5c52b9857b : SUCCESS in 10m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e97c968b2b8f46a59491a2d5c7f42a9d : SUCCESS in 16m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d00855cabd4a49eba1602bb030a3ff20 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2bbcd16fcaf745f2b8d0e09e041cf897 : SUCCESS in 15m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0fc6f0d09aea4e61b57c5ee353bd8b15 : SUCCESS in 14m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/202ccda442d34c4fb89c769e56d7e349 : SUCCESS in 16m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2fb772eb55fb4537bee2d017b7663907 : SUCCESS in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/03a916e00fda487d94bfccc9d10ef0c8 : SUCCESS in 8m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ec72649028b495f98c2dc3da3e3992f : SUCCESS in 7m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/495b729371b14678bf3001466c636e15 : SUCCESS in 11m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1148fcccf5d0488b8db292ad83ab4cd7 : SUCCESS in 13m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7fbdbcdafe954e8c8ad15d11095caee6 : SUCCESS in 1h 05m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd4f7730ffe94c1986018dc6a4cb2a83 : SUCCESS in 2h 01m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/267e4e28c6bf45d88d790078fc2d7764 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e12a8a6a43c3c52d06f0463df668b816fd114090","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 16:08:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ab874cf7d39823799ab77295b1d662fc412b321","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:10:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7092054297663470ca926c48505d5c5fba1ebee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:55:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09db4104cf342269b2853ab9d667b89 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d8c083405348f0bfc2eb0608bfe461 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5324cc30244c4004915db9b5286c2e77 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df51138bd71d4028842fe554492c5d9f : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67104e0f8a304293b20c72cf13bb709a : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/667cedb4457b4889b41b0cb1703c123d : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41f4e7c9dfbf40948b13b359b0d071f3 : SUCCESS in 5m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d2ae4bfdf69647b79611aefcc2b46151 : SUCCESS in 9m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce9b9785f2b14f3cbd09e8e3599e74a6 : SUCCESS in 8m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64c7244ed8d74705afd9d9bba8d4fca0 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/47737a9fe99f42ecb49276c631e5864c : SUCCESS in 7m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fbd1650512ec425a9d193907deca527f : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff432173aa7e4db6ba488ce3b5de8b09 : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2d07feb4220d4e52a240c3eac3fef98e : SUCCESS in 25m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/004192eafd234d2e89ff826c9d302bf6 : SUCCESS in 27m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f3f9ddd80d94a85bc5a6739b25ff524 : SUCCESS in 15m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/663a5925556c446fbe55779dc463ce4d : SUCCESS in 15m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/223bb945db6148848a2f13aca023241d : SUCCESS in 22m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eedcd555c78a49b0a95578333f3d82ae : SUCCESS in 6m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/67cccd47d057437881d3b48f74cb82aa : SUCCESS in 8m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf8ebf8c931843899cbdcd3c8a2565f7 : SUCCESS in 8m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c032c5425941465b8a419bde23b1d7cd : SUCCESS in 11m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/553d9a2836ff499584d08846544cc84e : SUCCESS in 12m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/381965a8231942f7ad370f5399c7a336 : SUCCESS in 1h 23m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f2f103f4c0d240c58231a34ab758286f : SUCCESS in 1h 42m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1eb96d4f20454b709a87a7ad0899bf44 : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bd486d2c06a0e2a62b49c5edb758398a120af518","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:55:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fbe2b0acfaf8406f67376a89004c355a7b7f07c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:57:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ffc5a8b5344589b5341a6e058b6aa6 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85166a0dbf9dde6272af8ff380c13b5aa68eff29","revisions":{"85166a0dbf9dde6272af8ff380c13b5aa68eff29":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:14:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/698677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/698677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/698677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/698677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/698677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/698677/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:14:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:14:00.000000000","tz":0},"subject":"update constraint for oslo.reports to new release 1.31.0","message":"update constraint for oslo.reports to new release 1.31.0\n\nChange-Id: I2a50048679fbf6bc14a09315c0a97cc773f87b88\nmeta:version: 1.31.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I52903f2f12fafcfe0ff7c6a72bf52142ae7290d4\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85166a0dbf9dde6272af8ff380c13b5aa68eff29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85166a0dbf9dde6272af8ff380c13b5aa68eff29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
