)]}'
{"id":"openstack%2Frequirements~698678","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I5d8e3b30d908d8b987f9d286ab542243606f6537","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I5d8e3b30d908d8b987f9d286ab542243606f6537","subject":"update constraint for ovsdbapp to new release 0.17.1","status":"MERGED","created":"2019-12-12 10:14:56.000000000","updated":"2019-12-12 23:42:25.000000000","submitted":"2019-12-12 23:42:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698678-1576194145390-91269b4c","meta_rev_id":"6aae78e63f2e0dfb6dc065b875341bd556f0fcdf","_number":698678,"virtual_id_number":698678,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 23:42:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-12 18:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-12 18:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:54:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 18:11:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-12 23:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c867dd121131e8aeed20c900ce8aac7e6e052fae","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:14:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4167277e6192eeda229acb4ac1c53d070b1c3bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:57:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/198f81d48f9b422e8c7b98afaf5b3451 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bea1cc92ed4fb69e4717ccc30dc611 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e7a8e724639412590c87a107913bb43 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5e0758aaa36487091ed15d55f48730a : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce6fd81490c347afa63a44ca5aca2115 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbff941ea6aa45b6b268c3ceff1a5740 : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6a904e3a25942e79ae953a8d827befd : SUCCESS in 5m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d2278ff8772a410c85495a813c624e88 : SUCCESS in 9m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/362f2afd01fe4019991a210bb51b7d91 : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/41a2edade23740e7a16f1eb9860a7be3 : SUCCESS in 11m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4fe9f5d1239a4fc183bf5ebae1ad8faa : SUCCESS in 9m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a1ae7b1811064dcf843f91daaf521076 : SUCCESS in 7m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0edd619846c642d4b1608c657427288c : SUCCESS in 6m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0c90c05425624ca98c017be8f429b279 : SUCCESS in 13m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/970a10d2a7964598a8758a8fa1432be7 : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7be2f0e9a8c546229217db3d07cfda67 : SUCCESS in 11m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/17c4f94b1fef48d38adb10dc63cd71f5 : SUCCESS in 24m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bbb1872a0a884de1ba16676dd3a81ba4 : SUCCESS in 21m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/692e2295d17c4a1b80d04a1e5b60b731 : SUCCESS in 24m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dc06953e0f9d43fca751b314f386001f : SUCCESS in 23m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d7892c4e1b48491984f8b68c7054dd89 : SUCCESS in 16m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3477ee489d6b4c6e89c7899902c29a41 : SUCCESS in 15m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6f88925c8ae44bfa9880588745fa9012 : SUCCESS in 16m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20bdd01aaf6447a1bded44c0ec927784 : SUCCESS in 22m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26b05cc1d3634820afd27b1c19b355d2 : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b353b38ef433457abf9790d91108bb86 : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9f71988728db441d8dceba8aed46e971 : SUCCESS in 11m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d66f5038cfeb4cb2b88be3203089d790 : SUCCESS in 1h 10m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f97eba2619e4ac883a01d7293bd6ef2 : SUCCESS in 1h 34m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fca2985b18f641628dcad714757fc7cf : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f222eceec81ccb7999669b0143058b4d269b0dd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:54:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1bc1cb59309041800ff03c5e1299e20ce79c6c67","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-12 18:11:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good from a stable perspective.","accounts_in_message":[],"_revision_number":1},{"id":"fadee241de247ad04d6669595d62821d1ce52224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:11:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4071126de5aceaffc038c03640727d9617d9c05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ef450e52e0a7fac672d658e362b9185925d91c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf645b0ed264fb4bf0b6f6eaae65fb8 : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d72b40f88e248c8b35ae59da76b9fac : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecfe50e9cc2c462daaae35af240ce838 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdc74771f4ec4400b9dbbf24b3d87390 : SUCCESS in 10m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4459b31ec151441aab87f78a27cdbac6 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8657340227254457823a433542a77668 : SUCCESS in 5m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7bdeae2125be480bbac470d379f485a3 : SUCCESS in 9m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5c957b5c59754e98a251514cd1b9c440 : SUCCESS in 11m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2780483c146b40c3b9ad945a6bd3fb5c : SUCCESS in 13m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85014f9770f34a6d902f8faf7fc18b1d : SUCCESS in 12m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6646911f29b94c27941fbc0e456b29a3 : SUCCESS in 7m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/81830421b1864deaa34a0a111635233e : SUCCESS in 8m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8bd1dced80c049c9b9daaad2f71cb87f : SUCCESS in 12m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8712c9a452c841da8e8775ec2eb5d0e2 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bac120ce50c46d8916053399491fe04 : SUCCESS in 11m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/96a9c4e366784c8992bc7169fae1769b : SUCCESS in 23m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fbcf55fd5b1a4101a936efad60f58e01 : SUCCESS in 21m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5d60d857526e48b5ad1c0ea4b7ea456c : SUCCESS in 29m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d47d173ecbf14b2f92cc43ef0db9b79b : SUCCESS in 23m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a4ec29626e704108a00df9414d965a7a : SUCCESS in 20m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be234befa96548e3b6b6d6536c24c81f : SUCCESS in 15m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8df6fb5e46c74a7396a60e47a746d1db : SUCCESS in 18m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98832dbf96df44878062058d0b78f3f0 : SUCCESS in 23m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6c020b81c894d94a0102fd7e8613cd6 : SUCCESS in 9m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6c2f514d50ac4a1e8b8a6d23fcdc7ca0 : SUCCESS in 16m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e0084a9b8e75499a85990de6a94da163 : SUCCESS in 12m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cbe416b042548598f1457955b59abc6 : SUCCESS in 1h 36m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eeea46466bcd4aa4b8ce0929ba745484 : SUCCESS in 1h 51m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7421ce3dd96545ec85e1c899c304441e : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e68c31180ea1ddd5bf6a24e3a218c7d59ecfb869","revisions":{"e68c31180ea1ddd5bf6a24e3a218c7d59ecfb869":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:14:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/698678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/698678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/698678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/698678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/698678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/698678/1"}}},"commit":{"parents":[{"commit":"a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae","subject":"update constraint for python-zunclient to new release 3.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5a11d9bba2ea89f21c4c9d05a9e3cfd8a04edae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:14:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:14:54.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.17.1","message":"update constraint for ovsdbapp to new release 0.17.1\n\nChange-Id: I5d8e3b30d908d8b987f9d286ab542243606f6537\nmeta:version: 0.17.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nmeta:release:Commit: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nmeta:release:Change-Id: I095030a6e338a2c7d5862b9f24f8cf7f49423d6c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e68c31180ea1ddd5bf6a24e3a218c7d59ecfb869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e68c31180ea1ddd5bf6a24e3a218c7d59ecfb869"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
